)]}'
{"id":"openstack%2Foctavia~681219","triplet_id":"openstack%2Foctavia~master~I1cfc963cc40b73e4b83630832003f98baffdc0f2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I1cfc963cc40b73e4b83630832003f98baffdc0f2","subject":"Remove mock patching of jsonschema","status":"ABANDONED","created":"2019-09-10 10:52:15.000000000","updated":"2019-09-10 19:53:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"734f5ad2bee891f1eb096d6cf563de398a324f0b","_number":681219,"virtual_id_number":681219,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-10 15:50:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 15:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43f476853fce6ac0a68b0bbaffc6d22e08302c72","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-10 10:52:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6eb5892f15485a80183d96ed17dd60f65df14f0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6247d60706974a03b6f17a81bcc1d918 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e53a585f00b4d8d82836a88ab5cac7d : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759ca39f9b764c8bb5c032d3a838e8aa : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a4d337cd0e44bfa8308ad7d77943d46 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/643ff1964bba4949a654c9dc95d9a2ca : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bb62b60eb624069a9eeb10a2feeb9cd : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d771d3f35e674c9ca73a9f3a747fe78d : SUCCESS in 7m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f415c08015854f4da8bcc8f88ec29463 : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8aad6ca18a3f41b1be7652efb1de4eed : SUCCESS in 5m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c3845ee32998441bacab0e8e5438a04c : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7668e694ac3d45a39c260e618242c858 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d0451f1e360462baeaadc9231ef13f2 : SUCCESS in 46m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/92ef5863be6b43ea9a305731bebceac6 : SUCCESS in 46m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54a9989356e8467298aeac948815eac4 : SUCCESS in 1h 53m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9c6affafe1314343963260c1a9574ecb : SUCCESS in 2h 28m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c9eeb6e1accb404ba9b6d1512a303f47 : TIMED_OUT in 2h 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3850ea5ae33438187a985fc15fe1081 : SUCCESS in 1h 59m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/35929713557d4841a6f843f463029d44 : FAILURE in 48m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d5632c1ff1f4f73a6593e8d885379db : FAILURE in 2h 13m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1bf2fc2748e746389230cd9d30b4d899 : SUCCESS in 1h 13m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fd53a35a26a54fdcb069fbf059b38a3b : FAILURE in 1h 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"734f5ad2bee891f1eb096d6cf563de398a324f0b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-10 19:53:47.000000000","message":"Abandoned\n\nFixed in https://review.opendev.org/#/c/665029/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d668122e8671d14fb1c59b6f73db1893bf993b49","revisions":{"d668122e8671d14fb1c59b6f73db1893bf993b49":{"kind":"REWORK","_number":1,"created":"2019-09-10 10:52:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/681219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/681219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/681219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/681219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/681219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/681219/1"}}},"commit":{"parents":[{"commit":"fa70e9759d52137e4ba847d20e9a941491f6455d","subject":"Merge \"Fix template that generates vrrp check script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa70e9759d52137e4ba847d20e9a941491f6455d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-10 10:42:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-10 10:51:35.000000000","tz":120},"subject":"Remove mock patching of jsonschema","message":"Remove mock patching of jsonschema\n\nUpper constraint version of jsonschema got bumped to major version 3 [1].\nThis broke unit tests TestAmphoraDriver.test_validate_flavor in amphora\ndriver v1 and v2. This patch fixes unit failures by removing unecessary\nmocking of jsonschema.validators.requests. Tests will still pass on\nlower constraint jsonschema\u003d\u003d2.6.0.\n\n[1] https://review.opendev.org/#/c/649789/\n\nChange-Id: I1cfc963cc40b73e4b83630832003f98baffdc0f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668122e8671d14fb1c59b6f73db1893bf993b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668122e8671d14fb1c59b6f73db1893bf993b49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
