)]}'
{"id":"openstack%2Foctavia~681625","triplet_id":"openstack%2Foctavia~master~I05439ea1dd01c35bedcfc3eaa5d17ed8dd2ca348","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I05439ea1dd01c35bedcfc3eaa5d17ed8dd2ca348","subject":"Fix \u0027additive_only\u0027 parameter api-ref","status":"MERGED","created":"2019-09-11 22:54:51.000000000","updated":"2019-09-18 09:07:36.000000000","submitted":"2019-09-18 08:59:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681625-1568797179129-fa3c3f25","meta_rev_id":"de91c3745dd63987dac50ca212826a17d8b56202","_number":681625,"virtual_id_number":681625,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-18 08:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 05:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-18 05:50:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-18 05:50:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 07:38:49.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-18 05:50:02.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 08:59:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf1f522961cd4691312eb0e0fe32385260319f2d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-11 22:54:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ae6a539142e0d9ae9bf790c83b527d6f572660f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 23:09:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a150d3e025934dea8b66a9423392f967 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf7299887bf42b6b01afcf0f59e2a7c : FAILURE in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db08f30cf95646fc9e6d47e2c9288aab : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"89f04f42069a91ffd3e0830ef2426cfe8f1cbcd4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 07:38:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"964f87a52b5696ad03835b25f3c4b3b999dd1878","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 08:16:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fc2fb8c96bd81d0b40bec9d0b4d8042e15dc4b90","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 08:16:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e8c023ce26b90a3150df369a4e7edaba72a25098","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:31:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6a6c4f96bc74e1da41f35efa443740c : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a119644559d540f2b9c6257564d200f4 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306a459251da4a619ff967c53d7e10ff : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b0b63c95b833b07d749b121b996d5693edc244bd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:50:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3a34cf4eee83fd0a07fb388a16c717a19ec5dcba","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:50:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e49c58dda95955e0b228882e573331d3bc1371da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:59:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8927920220584adf948a8ac911cc7147 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08b139efd364a93977eb3eb3bd169db : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b99c0207354a19a1b1a97e1d0b3276 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d3114cfb9be6c4c3e8d48bb1c0d8c5c65b282b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 06:00:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b643b24d83ad06ef53844b02d59ab3543549d881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:59:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dbb24839f4914e27baaef61e0d1f720e : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd016ae245f4e8484bcc12c2f044547 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79fa3a3558b4bbb8694766b632c7162 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":3},{"id":"252f99f9c212a880e128d32526f769245672fd47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:59:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"de91c3745dd63987dac50ca212826a17d8b56202","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:07:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/eacc3b1f570f41219637242fd8525e81 : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b52156d69741a1abd941bb1a836413 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b9d357ac76b0624f728f186e4d5969872bd92e28","revisions":{"9c0469edd03f987afc07930134bf1222cc0dcea6":{"kind":"REWORK","_number":1,"created":"2019-09-11 22:54:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/681625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/681625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/681625/1"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-11 22:53:27.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-11 22:54:50.000000000","tz":-420},"subject":"Fix \u0027additive_only\u0027 parameter api-ref","message":"Fix \u0027additive_only\u0027 parameter api-ref\n\nThe \u0027additive_only\u0027 patch was missing the \"min_version\" parameter\nin the api-ref. This patch fixes that so users will know which API\nversion supports this parameter.\n\nChange-Id: I05439ea1dd01c35bedcfc3eaa5d17ed8dd2ca348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c0469edd03f987afc07930134bf1222cc0dcea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c0469edd03f987afc07930134bf1222cc0dcea6"}]},"branch":"refs/heads/master"},"b8f292c322b2b1ee95ddaf8eedbd159bf7fb6fd5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-12 08:16:34.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/25/681625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/681625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/681625/2"}}},"commit":{"parents":[{"commit":"e189acdfc48387c1cd84dfe6241ec22aade935f7","subject":"Fix a few nits from the main volume-based patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e189acdfc48387c1cd84dfe6241ec22aade935f7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-11 22:53:27.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-12 08:16:34.000000000","tz":0},"subject":"Fix \u0027additive_only\u0027 parameter api-ref","message":"Fix \u0027additive_only\u0027 parameter api-ref\n\nThe \u0027additive_only\u0027 patch was missing the \"min_version\" parameter\nin the api-ref. This patch fixes that so users will know which API\nversion supports this parameter.\n\nChange-Id: I05439ea1dd01c35bedcfc3eaa5d17ed8dd2ca348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8f292c322b2b1ee95ddaf8eedbd159bf7fb6fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8f292c322b2b1ee95ddaf8eedbd159bf7fb6fd5"}]},"branch":"refs/heads/master"},"b9d357ac76b0624f728f186e4d5969872bd92e28":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-18 05:50:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/25/681625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/681625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/681625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/681625/3"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-11 22:53:27.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-18 05:50:00.000000000","tz":0},"subject":"Fix \u0027additive_only\u0027 parameter api-ref","message":"Fix \u0027additive_only\u0027 parameter api-ref\n\nThe \u0027additive_only\u0027 patch was missing the \"min_version\" parameter\nin the api-ref. This patch fixes that so users will know which API\nversion supports this parameter.\n\nChange-Id: I05439ea1dd01c35bedcfc3eaa5d17ed8dd2ca348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9d357ac76b0624f728f186e4d5969872bd92e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9d357ac76b0624f728f186e4d5969872bd92e28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
