)]}'
{"id":"openstack%2Foctavia~681780","triplet_id":"openstack%2Foctavia~stable%2Frocky~I1a566353b7fb125184849b3a0f864d3ef6896d94","project":"openstack/octavia","branch":"stable/rocky","topic":"fix_exceptions-stable/stein-stable/rocky","hashtags":[],"change_id":"I1a566353b7fb125184849b3a0f864d3ef6896d94","subject":"Fix catching driver exceptions","status":"ABANDONED","created":"2019-09-12 15:29:59.000000000","updated":"2019-12-06 17:29:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32bfcdbf0deb0501b4094eacdeef24a7e6ed1913","_number":681780,"virtual_id_number":681780,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-06 17:08:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-12-06 15:58:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 15:29:59.000000000","updated_by":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"reviewer":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"state":"REVIEWER"},{"updated":"2019-12-06 17:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17fd2ed91a3d2b90278fa3624b121abb2f4e6e3a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 15:29:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"15aa6cf7f949a9b18f5c85c5bd0c4ee5cc822015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 17:29:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cdbbd34b8ed4c2587731f16c11bfea3 : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bedfe2e006aa442fabf2b05f58edba7d : FAILURE in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d83f0ff7db44b2e90a653837849ff92 : FAILURE in 5m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db0c77a172e1401ea881bcf5b7a46461 : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab3521b766a4a71b20ed545adfc99e4 : FAILURE in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/937c5236da9441f3af848b23c3877cd2 : SUCCESS in 4m 38s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9688038488a5483f8b93c0667ab1caf7 : FAILURE in 49m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bedf5786cf8a4c678299ae01fd30320e : FAILURE in 54m 35s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdc4334d44ad4b068c362c7aa322738d : FAILURE in 53m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f8e772b1962c4c968f00b46b009b4011 : FAILURE in 53m 56s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/00015eea25634c428fd9206d1387add2 : FAILURE in 7m 01s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/43015968dbef47d9a3a53e48640be32d : FAILURE in 12m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b4873ee504f46078f1a09fce95c467f : FAILURE in 4m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f350c40b83954047941f5c743d432a5c : FAILURE in 50m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0eb6f0b6203747579ab045bd1ef8aec2 : FAILURE in 50m 14s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0cd42bd482dc43019ebf4eb728387015 : FAILURE in 1h 01m 20s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4468b3246c044f43a420c8dfd63df7c2 : FAILURE in 1h 04m 40s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b8647282bfa04f8f865cb60d044e4b96 : FAILURE in 57m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6393d8c5c8474e37822033d18895f565 : FAILURE in 32m 43s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/152c387a356243f192cd86667745caa4 : FAILURE in 34m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56bf61c3417a4f4abac8bdfce5fe349e : FAILURE in 1h 06m 38s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ac3ae70c41ca46748e9b5f4d27e8c4c7 : FAILURE in 1h 02m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/719f33620e9742e7b42dec1966d2f999 : FAILURE in 1h 33m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773978ced1e74cccaee3cb6495b30706 : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f4d135e83f6bb9b9d36f293820587a18817dfe39","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 06:52:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0668bd1ee98e771251ef7f601168205084c8deef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 11:08:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b295d3a3d800463c97bbf1e1fdfd2c87 : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08d1b637e1c450591f71466ab171c43 : FAILURE in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/982358489c7944139b27f84885bd3f16 : FAILURE in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ede8ce4cf4044362970ce9ec7de0d322 : FAILURE in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50affb3c50684a36b5e2f7f90d500d0e : FAILURE in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4802ac8bd45b4d82864d6713506be38b : SUCCESS in 5m 57s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9b63dc470fd94e3bac7bb13c685eee97 : FAILURE in 48m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c4a6812081f345e6bbb4280a3a1458c8 : FAILURE in 55m 30s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77e73a147a0c41f89a6e86631eeff3ea : FAILURE in 51m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/091be5257e8b4779af2758c0edc91925 : FAILURE in 50m 59s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9b7c7172d4bd45b791ab3816dba6fd32 : FAILURE in 5m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/17a21b97d1344d999b0a634a3f6470ed : FAILURE in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90d0f2a4d0b14af68d1c99a576dcf66f : FAILURE in 3m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8d8a152ec3040c8b6d95bbc603bf8ca : FAILURE in 52m 07s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cea1731eadb94d288a99d7e8d22bcde0 : FAILURE in 50m 37s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/daa9ed58dcd34f93a64ed815f056e6d7 : FAILURE in 59m 16s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e9e1a87035b24da990af0c1b3b2aeb76 : FAILURE in 1h 00m 57s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ffd32a4305a64060a3a742565949c51f : FAILURE in 46m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29faa1e3381a4f7a86a651ad34950a9e : FAILURE in 36m 41s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5a2350093f7c496784df27bab92574aa : FAILURE in 32m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b27aa86597841fe94bbe2a20c161083 : FAILURE in 45m 21s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/0f1f3532c12c48a1b22e46e76b437f13 : FAILURE in 1h 11m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b3885bd29554368a98a7cddb840ce1c : FAILURE in 1h 07m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caeb8a7e6387428ca1beafd1cf742299 : FAILURE in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"430e598a74886c015cbefac181e1511d83dc38a1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 06:52:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6642b934ca3fd2995709b244ebcb879f047aa361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 08:38:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c1de191bdb34baca62b3198a3cbd774 : FAILURE in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718be5fc2bd0459b9a277160da74081f : FAILURE in 8m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8392b7911f9842a28f84c4a302385c4e : FAILURE in 6m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cc734747d3f849e5922da4e584c34b83 : FAILURE in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7443499b2d0b4182a996b5fda22d5f52 : FAILURE in 14m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d50cb4eda0b04008b7d09371d990f503 : SUCCESS in 24m 01s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a92eefa1376448799edb425463add623 : FAILURE in 1h 19m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2af58b7e6d4842a5beb94e9fc45df6c4 : FAILURE in 1h 19m 48s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/806c831e7ac84552b2c64f096e519126 : FAILURE in 1h 20m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9e0d3a3f395d4d2eacc4d34eac3260d7 : FAILURE in 55m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be0facce66e94d838d55f948eee40f31 : FAILURE in 11m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/114524059a6a49958518833c46e1e174 : FAILURE in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78e64512b1dc4b5dac587c3c9dd56f21 : FAILURE in 8m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/116f7b8363774024af45437b06628d49 : FAILURE in 1h 17m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e7a45e0bcb0d4bb68bb40cfea5a94a26 : FAILURE in 1h 07m 20s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e58dadefe4884376a9e806dcc04ce285 : FAILURE in 1h 05m 30s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5b63912c9e09433d8dde5d91c9a3b37a : FAILURE in 1h 38m 39s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e6d0f0f49fda4840b8d5991c33f653bd : FAILURE in 59m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/060d7a48a6734a8888c4298bac6e34cf : FAILURE in 44m 43s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/55631d53bc3141ba9bd1516f05a21b7d : FAILURE in 38m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a11c8da6b4a54ac19d21e80b6fa4a0a0 : FAILURE in 55m 19s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/edc6064226464fbaad56b42fdc5d6267 : FAILURE in 52m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4691e1b8507748489d0fc1aa5536743a : FAILURE in 1h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd57a785f8b24878a80cd4596d4a4ba6 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"00b5a97a4ecab479444d1ea4c70dea8a603fb56e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-06 15:58:06.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"739060c6d55b4b60c7e6c0ab4979a03008f6d435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:08:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b852433204246bba39ec720eb060352 : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14478e023b1d4a619bd32c4f5b2fa478 : FAILURE in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b00ddc068e548939dbd8558d251c8e0 : FAILURE in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a92a6a244f7745cea3021e5b0edf59f8 : FAILURE in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107108261e23461c965eac00f6b72390 : FAILURE in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2fc526ed61a4cca9f5c19a57fe0db60 : SUCCESS in 5m 41s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/36f335f7b5fe483ba163362ef499d1e3 : FAILURE in 48m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a132bf74967d42e889b2f31b11ca88b4 : FAILURE in 58m 52s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a1a43e49250461daaece42a28c1657b : FAILURE in 56m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d056d833112544ec89bf697fd72f8e39 : FAILURE in 1h 01m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/67172d9705504300938d8a570e7428fd : FAILURE in 3m 51s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2bdf4a91b4a745a1a8b2a3b97ba23de9 : FAILURE in 4m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b6db93559f542ff8f5c57675555739c : FAILURE in 4m 18s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e15a18ffd69404bbb3f79e817a37ef4 : FAILURE in 51m 23s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/195b3f14e453466f8445eaceedda661a : FAILURE in 50m 36s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f04c8016444b4011bb69cba09487af32 : FAILURE in 12m 00s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/aa66998eb113475f83dbb468de2eb600 : FAILURE in 57m 31s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/bb10ad1df13f4fb7bb05d5da054d1caa : FAILURE in 48m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93236a9c4128491592e6135d6bfb7a3a : FAILURE in 41m 54s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/81c581888ce9436a9550abfc48ed7ae2 : FAILURE in 40m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/290a6205774b4ad8be0ddcb0b9c85d5d : FAILURE in 39m 13s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/80bdbfe8ee4c41c5a5c04423fa7487f4 : FAILURE in 39m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/53b89b1d5361473f848262f05a962ed1 : FAILURE in 1h 00m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de948e50d98b422f859d0fff72aeb08a : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"864e8758ca17f6bf9f24e41d34f6159c5595029b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-06 17:29:39.000000000","message":"Abandoned\n\nOpps, no octavia-lib for rocky, so not needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"88cca17428f31ee826e62057d6fd958b0444b0df","revisions":{"b5f17d912e6e2e903d2de66e3f4205e6318ee1b2":{"kind":"REWORK","_number":1,"created":"2019-09-12 15:29:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/681780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/681780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/681780/1"}}},"commit":{"parents":[{"commit":"5e5d9dfdd71ddb56d7a10efce5e4241652f377b2","subject":"Merge \"Add failover logging to show the amphora details.\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e5d9dfdd71ddb56d7a10efce5e4241652f377b2"}]}],"author":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-03-31 06:27:16.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 15:29:59.000000000","tz":0},"subject":"Fix catching driver exceptions","message":"Fix catching driver exceptions\n\noctavia_lib driver exceptions were not cought by call_provider\nThis patch add those exceptions, and also handles the native NotImplementedError\nwhich does nto have all the expected fields.\n\nChange-Id: I1a566353b7fb125184849b3a0f864d3ef6896d94\n(cherry picked from commit e2defa6dd238d82ce4ee457e21ca047888d27106)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5f17d912e6e2e903d2de66e3f4205e6318ee1b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5f17d912e6e2e903d2de66e3f4205e6318ee1b2"}]},"branch":"refs/heads/stable/rocky"},"e30d8fc331a26beeb895ed1060e14fe65c2bd5ff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-20 06:52:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/681780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/681780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/681780/2"}}},"commit":{"parents":[{"commit":"5f474adf1b1e0d345e89b3974749338d1630369a","subject":"Merge \"Validate server_certs_key_passphrase is 32 chars\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f474adf1b1e0d345e89b3974749338d1630369a"}]}],"author":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-03-31 06:27:16.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 06:52:05.000000000","tz":0},"subject":"Fix catching driver exceptions","message":"Fix catching driver exceptions\n\noctavia_lib driver exceptions were not cought by call_provider\nThis patch add those exceptions, and also handles the native NotImplementedError\nwhich does nto have all the expected fields.\n\nChange-Id: I1a566353b7fb125184849b3a0f864d3ef6896d94\n(cherry picked from commit e2defa6dd238d82ce4ee457e21ca047888d27106)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e30d8fc331a26beeb895ed1060e14fe65c2bd5ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e30d8fc331a26beeb895ed1060e14fe65c2bd5ff"}]},"branch":"refs/heads/stable/rocky"},"eecb9211005c47435b82c51e6838de3f39479b29":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-08 06:52:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/681780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/681780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/681780/3"}}},"commit":{"parents":[{"commit":"d5aba906a1105018d222ccfb61716f15518cb072","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5aba906a1105018d222ccfb61716f15518cb072"}]}],"author":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-03-31 06:27:16.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 06:52:58.000000000","tz":0},"subject":"Fix catching driver exceptions","message":"Fix catching driver exceptions\n\noctavia_lib driver exceptions were not cought by call_provider\nThis patch add those exceptions, and also handles the native NotImplementedError\nwhich does nto have all the expected fields.\n\nChange-Id: I1a566353b7fb125184849b3a0f864d3ef6896d94\n(cherry picked from commit e2defa6dd238d82ce4ee457e21ca047888d27106)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eecb9211005c47435b82c51e6838de3f39479b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eecb9211005c47435b82c51e6838de3f39479b29"}]},"branch":"refs/heads/stable/rocky"},"88cca17428f31ee826e62057d6fd958b0444b0df":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-06 15:58:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/681780/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/681780/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/681780/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/681780/4"}}},"commit":{"parents":[{"commit":"1fffeaaf44c677c4e8990193762d0b920566fe11","subject":"Merge \"Fix controller worker graceful shutdown\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1fffeaaf44c677c4e8990193762d0b920566fe11"}]}],"author":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-03-31 06:27:16.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-06 15:58:06.000000000","tz":0},"subject":"Fix catching driver exceptions","message":"Fix catching driver exceptions\n\noctavia_lib driver exceptions were not cought by call_provider\nThis patch add those exceptions, and also handles the native NotImplementedError\nwhich does nto have all the expected fields.\n\nChange-Id: I1a566353b7fb125184849b3a0f864d3ef6896d94\n(cherry picked from commit e2defa6dd238d82ce4ee457e21ca047888d27106)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88cca17428f31ee826e62057d6fd958b0444b0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88cca17428f31ee826e62057d6fd958b0444b0df"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
