)]}'
{"id":"openstack%2Foctavia~682167","triplet_id":"openstack%2Foctavia~master~Iacee63120a0d60e312cc85c7fcb8e7351688af3d","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iacee63120a0d60e312cc85c7fcb8e7351688af3d","subject":"Set neutron client logging to INFO","status":"MERGED","created":"2019-09-13 21:35:44.000000000","updated":"2019-10-12 12:42:25.000000000","submitted":"2019-10-12 12:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682167-1570884039224-9bbf562d","meta_rev_id":"85efc02d0afcc5c5d5272e68c200802f7265a43c","_number":682167,"virtual_id_number":682167,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-12 12:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-11 23:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-12 09:46:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-12 07:43:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-10-12 09:46:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-12 09:46:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-10-12 01:35:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 23:12:58.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-12 01:35:01.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-10-12 07:43:31.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-12 09:46:52.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-12 12:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d272f987b0ca666bd59e73931b73fbd6f4db196","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-13 21:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd7c720372af06a2f0caefd0ea2d3e3f475ecf9b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-13 23:09:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"368eeaf4aec6db822040bf008d7b1c5bc4c3048a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:46:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e637bfa11aa94c04a9b7154b3aa28e95 : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caa6baa2ad84430a8f0775152e76a8ea : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abad82c35d0d419f927e344103a8f22f : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a95b0f19d1ef4f898d3dfbfd7dc244d1 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/438291e0cc4f47d5bb08de1c2507dd0c : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6beecd41f3f340b3a0a5d2a0af142d5c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ff616fa2f749899ba7963dc4361dac : SUCCESS in 6m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2ac6abe680764885955fa3dd233b0e91 : SUCCESS in 4m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f793f02336964633a5b903f8e0532172 : SUCCESS in 7m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/15a524220fc044539801d56f5ff2c4d1 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63bab73dd58b42cd9eb0150c78d4939d : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67972876b2af49b39404bd904b8f7d61 : SUCCESS in 48m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5131323b76e045d1ba3ce8d005833562 : SUCCESS in 44m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/872372dee92d44e58abadb86b765157f : SUCCESS in 1h 37m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7d1af508109f4c498fe43b53bdaeaa47 : SUCCESS in 2h 07m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c9923ac4404d413aace35276a4a7aeaa : SUCCESS in 2h 29m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16b491732d104919aee82f57a57dd2d3 : SUCCESS in 2h 02m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d13d122762845eda2494c89adc2f622 : SUCCESS in 1h 54m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/958f7e0a881f43cbb505219ab20bb4ec : SUCCESS in 1h 26m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ea09c72bd4d4fed8a68e87dd05d5650 : SUCCESS in 56m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b6c64124f374fdcaf930592c6b2ab72 : SUCCESS in 1h 38m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b82edc46de1447c5a864f88c3b9ae0af : SUCCESS in 1h 15m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/520128635f3f473a96f5144b56dd1e3f : FAILURE in 41m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6711305ff44c4693b04b40f751ac0ebd : FAILURE in 1h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d386b7f761bedd43b63ae8c60b1346099d2433f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-11 23:12:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c3ed8916e4b62b911b83665d52b9bf0f053f1c92","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-12 01:35:01.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"a08d11ba3f180833a42109fdb0fdacd67dd7e922","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-10-12 07:43:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9dbd8bdcb2dda793ec9380529690d7e0096fb3a7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 09:46:52.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1\n\nThinking this could be worth backporting. I\u0027ve seen users troubleshooting issues and spending time understanding these neutron \"error\" messages while that was not the problem at all.","accounts_in_message":[],"_revision_number":2},{"id":"3c4cdfffcbbef79de42e30913b68adc3c43f3f93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 09:47:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2cc3e5fc6f5044ce5aeab6b7c4172748b067a7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 12:40:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf1e62fd87324efb9ab5588f7eec67bd : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b92342bda574afe9bb5eade1cfb3b91 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f12bd38b4cea46c89ce19cf8b97533ee : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7527b1f76b2a48c29b279d0bf8a6e34e : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46414f8b519544079d43dab75952c551 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e96dacabd784c3d9ac31a5fdd290f73 : SUCCESS in 9m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b46edf4c0d5420fa4260cced88b6074 : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f756e8840f1479db5c5e115aca53be0 : SUCCESS in 5m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/574a0b9475bd48b3b8acce0aefd8c31b : SUCCESS in 42m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2c2a72bcb76d4fbb8a640e34207dce88 : SUCCESS in 41m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25ab939c123e4fa0bf30195203833cf5 : SUCCESS in 1h 35m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/348c7c8375eb47aa99e5a8a425fc4443 : SUCCESS in 1h 38m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/04e7cbaa714348d79163f6ea3da50206 : SUCCESS in 1h 25m 37s","accounts_in_message":[],"_revision_number":2},{"id":"3b476719ac072eb9b89aae124a98e1a052f95cb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 12:40:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"85efc02d0afcc5c5d5272e68c200802f7265a43c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 12:42:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394d57140aed4ba3b3043528d353a8a7 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e795406343dcd3d33f5c18de9b0d6313370afa7a","revisions":{"a7341a84e8b669d1dc9b11f1535838b7be32b71b":{"kind":"REWORK","_number":1,"created":"2019-09-13 21:35:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/67/682167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/682167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/682167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/682167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/682167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/682167/1"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-13 21:33:09.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-13 21:35:43.000000000","tz":-420},"subject":"Set neutron client logging to INFO","message":"Set neutron client logging to INFO\n\nThe neutron client will post debug messages with the word \"Error\"\nwhich is misleading. In reality it is a simple 404 that an extension\nis not enabled.\nThis patch raises the default logging level for the neutron client to\nINFO to suppress these messages.\n\nChange-Id: Iacee63120a0d60e312cc85c7fcb8e7351688af3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7341a84e8b669d1dc9b11f1535838b7be32b71b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7341a84e8b669d1dc9b11f1535838b7be32b71b"}]},"branch":"refs/heads/master"},"e795406343dcd3d33f5c18de9b0d6313370afa7a":{"kind":"REWORK","_number":2,"created":"2019-09-13 23:09:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/67/682167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/682167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/682167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/682167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/682167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/682167/2"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-13 21:33:09.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-13 23:09:29.000000000","tz":-420},"subject":"Set neutron client logging to INFO","message":"Set neutron client logging to INFO\n\nThe neutron client will post debug messages with the word \"Error\"\nwhich is misleading. In reality it is a simple 404 that an extension\nis not enabled.\nThis patch raises the default logging level for the neutron client to\nINFO to suppress these messages.\n\nChange-Id: Iacee63120a0d60e312cc85c7fcb8e7351688af3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e795406343dcd3d33f5c18de9b0d6313370afa7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e795406343dcd3d33f5c18de9b0d6313370afa7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
