)]}'
{"id":"openstack%2Foctavia~682282","triplet_id":"openstack%2Foctavia~stable%2Frocky~Icddd2873abeb56a389a35356995df6dde70872b2","project":"openstack/octavia","branch":"stable/rocky","topic":"patch-jinja2-stable/stein-stable/rocky","hashtags":[],"change_id":"Icddd2873abeb56a389a35356995df6dde70872b2","subject":"Fix template that generates vrrp check script","status":"MERGED","created":"2019-09-16 07:14:29.000000000","updated":"2019-09-18 07:22:21.000000000","submitted":"2019-09-18 07:21:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682282-1568791260418-d6d561dc","meta_rev_id":"8370d506e7518ee7e0ca166e176ddd30d9487a2e","_number":682282,"virtual_id_number":682282,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-09-17 11:01:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 07:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":2,"date":"2019-09-18 05:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-17 16:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":1,"date":"2019-09-18 05:34:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-09-16 07:14:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 07:14:29.000000000","updated_by":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"reviewer":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"state":"REVIEWER"},{"updated":"2019-09-17 16:04:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 05:34:16.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 07:21:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9666443d2259c1be661b5cc58bbd7297605617f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 07:14:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"2cb0bf3eb97c6ced1a609b93a8d4af2e16b4eaad","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 07:14:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 70d97efb4e283264e25c1f5d63fd0707d1217a18","accounts_in_message":[],"_revision_number":1},{"id":"13483a74eb3045eb67e866298024b02c02b49963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 09:24:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408edf2409824448b23285e04457b084 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bedd1e25ceea41b1bf795631c63add10 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6189321386d4eeaaa612e0f3f6ec174 : SUCCESS in 5m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c7cd4b0e2ef45e0b3e2c8ff9b030c40 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21125779e26d47f98100a704ec694b59 : SUCCESS in 5m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bc94ab88e56d44d8918f65234650274e : SUCCESS in 1h 12m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4d82b3238dd04df7ac3567f0b9f36be8 : SUCCESS in 1h 14m 06s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c9f03a835814327b62c866e89933e06 : SUCCESS in 1h 35m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9c5b355f5a2d459dadce4f1619512c29 : SUCCESS in 1h 18m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/298a0e6e6eab4ff88b690c43ec017e08 : SUCCESS in 4m 31s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3d15c466723644a687df39bd079d761e : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/665877a55a99411d830be69040164efc : SUCCESS in 5m 34s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bef50c358e44412bbba8fca430e8a0f : SUCCESS in 1h 03m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/75bafc7087ce4dd0abf079045927826d : SUCCESS in 1h 19m 47s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/aab987ec53a245dbb9100a99c15e49ef : FAILURE in 1h 38m 16s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f6eb275972a949ee9635a0d3158b08a9 : FAILURE in 1h 33m 18s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/6203653ff0b44859973c5102e0d147b0 : SUCCESS in 57m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89cb87bb4da040c091fd1463a5a3e727 : SUCCESS in 43m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/79a301010b68468ba4858b0b3191dcef : SUCCESS in 1h 02m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1b7c96a61ff4cd9bf084b1bfcb8355d : SUCCESS in 1h 50m 14s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3b4c8ba4aec44b57b73ed39fc1c21048 : SUCCESS in 1h 54m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae579081cb3349018b8babc728883fb7 : FAILURE in 1h 07m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/281776db9d7e404a880c9137b7d0fe2d : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b2e197b4f69dee498f314bd751cbd34c5ee79944","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 08:26:26.000000000","message":"Patch Set 1:\n\nrecheck\n\nThe usual suspect: Error plugging network\u0027, u\u0027details\u0027: u\u0027RTNETLINK answers: File exists\\nFailed to bring up eth2.","accounts_in_message":[],"_revision_number":1},{"id":"f2844458e8b564bdd19e85bf1f4d952a3df5e02d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:54:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98223055dcdb409bb99d259f6049d419 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29f7bcbc8ec408181d6cbc4e0936e47 : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/122ee027c2c64f8ea65b6ad88ebfc7de : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/89711125083e48a7818fe51ffd845ed1 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18edf660ff484072a7f3a8e464755d7a : SUCCESS in 6m 30s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c9b27d4726ce492081069518b6f27462 : SUCCESS in 1h 14m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/903988c6eecb401c97ca7ebbe3880761 : SUCCESS in 1h 19m 26s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0ef44f36e234ab6ad238064b4278957 : FAILURE in 1h 08m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d55506c438b2432092bd7b9e47b63473 : SUCCESS in 1h 32m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e14b57a722bd4464bf63e6222d26d2c8 : SUCCESS in 5m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e73118edf56143c3a3b2e4b2a21432f9 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24f1b3a7692743d2ab404f3da0713a48 : RETRY_LIMIT in 1m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f963d3cea5854433bd81f0e12170e3d6 : SUCCESS in 1h 01m 14s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c9ebc85320e343738bf475aac037db4c : SUCCESS in 1h 06m 23s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/da6c4bcf14c246c7801845c98d2dd32d : FAILURE in 1h 28m 22s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/fd774f9f0e6e43fe8eeaeeaf477ae821 : FAILURE in 1h 34m 06s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/6b10b8c7f3604051b3f46510f665a48b : SUCCESS in 1h 08m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cf20853d8dd4050bca74909d3540324 : SUCCESS in 51m 38s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ae70a348dfea4d7880e98223c2624695 : SUCCESS in 54m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9a51beccdce40159971240e2e7f4307 : SUCCESS in 1h 41m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7cdcfbb23afb4aa5b1aa4a3d5e931e65 : SUCCESS in 1h 34m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5235a582b2a4dc88df1e13c9680feda : SUCCESS in 1h 27m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c746981b202a4a94836832c48b63816f : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c35236271939999c83a1cca2bbfed5c441b5ab6d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 11:01:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb3baee834db6d64862aa8dec5d507262fb5dc44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:35:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2e08609700c404e9102113a4b31729a : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0567e6e9648347ffbafb31493b7a5e63 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/736948f34da94e098939f6f9baf3b173 : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8efedc56a1a44a25ad93f8e2faf5792e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c762cccf809453a9547899435410325 : SUCCESS in 7m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/744688f0298c4f6fad3da721c8b2edd3 : SUCCESS in 1h 15m 56s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6da2f5f3f23348ecaac30dd6bca622dd : SUCCESS in 1h 18m 55s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/373783d53f084e7dac4a599800d007e7 : SUCCESS in 1h 08m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0e2b2ce92f514b609afba2b37e70d8bf : SUCCESS in 1h 33m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5d23c60071ba4393992183b4cffcbb62 : SUCCESS in 5m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3418f32abbec4287aa0a58c387c4369a : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e60a15a1aec94fabac07c4582a05cfaf : SUCCESS in 5m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e62cf3873f946a6a57076d9a86f25ad : SUCCESS in 1h 06m 32s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f6783f71dd6848bf89fc968818881ba2 : SUCCESS in 1h 05m 32s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5214dc7878aa48fabec8bbaa4df39e59 : FAILURE in 1h 22m 28s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b404119632d9471ea9a2cd62dda4d7de : FAILURE in 1h 20m 27s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/9b4b6ad6e7f04b2dad7fa381ec1c0cd2 : SUCCESS in 1h 12m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/836267a7f52d478a9da1e6b9d42118b7 : SUCCESS in 52m 28s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/df6b5d11b0ca413b8755bc752a62e494 : SUCCESS in 57m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47960842cf3c403d8e2e7455665f435a : SUCCESS in 1h 50m 00s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1ebadef8d2ab42b0a89c3c07523def28 : SUCCESS in 1h 53m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e4d914589c9d4449aba7de4f48ed628b : SUCCESS in 1h 06m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9957d1ddc15455c8f5299695ef9028c : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5796fa826d51c989bbb7da87ef3310cb079a7053","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:04:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b3db113f2087d64de51dd09edf59b09b44380371","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:04:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0541177b7fa031750144d47515d1e9b2dfba19dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:04:52.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"61204675b221c27a0012100733eecb1663b0577e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:16:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d86262553b04ed897814d756362fc70 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3aa03b946dd441ab72226177ff71b18 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/431588b519544ec1babdbeff614f3335 : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a67b3842531741c092775197130739a6 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7334c0bf379140a69677cc3d19558180 : SUCCESS in 8m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/07a41d6f391b45ce9ad118b59dc96818 : SUCCESS in 6m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7e824e693b3f40caa9290b1c227cd68c : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c7ab35da1d344778931b3b224849dd6 : SUCCESS in 6m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa2f379bee7b40e18cb741e9d60b82d3 : SUCCESS in 1h 01m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f3e9c438e0bb4f749279a8084a77432b : SUCCESS in 1h 01m 09s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cc1b45cdaccc4f0f9aaae9bf85c14b63 : FAILURE in 1h 28m 47s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/befcd59ef66549af94ed7572c15a21e1 : FAILURE in 1h 32m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bf2c16b2a875454795897e953fc6e6c7 : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06bd2fce47314b9fbbf1432088108a9e : SUCCESS in 57m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/90aece8c51ca4d2191e9a796a18861af : SUCCESS in 1h 14m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/317ae5005784472db22de7f7b70e080b : SUCCESS in 1h 44m 44s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/73d78a02a8e6404aa7804cc634d87baa : SUCCESS in 1h 28m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03df56f3a57d470abd357e98092c3827 : SUCCESS in 1h 18m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/893eecea9b5543d38d0d6bf8a0c3c89d : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5a8e4440722043943067ce1901976ea01609f3b2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:34:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b061ec2cae4ed943d0264ff9217a87b48ce602d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"143adb0dc13b116d9ce2a84256d6481c60930628","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:21:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b85e2d3f431e4175a7c13414376e3254 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4361773ceaa84a169f32089a4f52e257 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/445fc83b3e3a4f35aae21e3cd5536b0a : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db8074ba5eb64e3dbec3406f4a20334b : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f72242a3a2458f9b3351e0b7c26202 : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e2ec7a4eb0ac4ae69633bbbb5fa6fe4b : SUCCESS in 5m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d937eb992063463eb0ae8ed5bd035b5a : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5c93177907443b5b2597a26b4b64152 : SUCCESS in 6m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66276e060e5b47f19de88713879bd8f2 : SUCCESS in 54m 36s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8f5abeb6f2224e38acff087b5077c2d1 : SUCCESS in 54m 15s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f1dcbbe66e484be4b4c53d83cfc6c4db : FAILURE in 1h 13m 19s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c2238f70e22c4209896a86264a3d6146 : FAILURE in 1h 17m 43s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2a280d85978347a494c69169fe660292 : SUCCESS in 1h 13m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42330052a9f749fbb8b7a7da3de97325 : SUCCESS in 47m 51s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5e3d624ddd0e427890bfaa8f9bff5716 : SUCCESS in 1h 07m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab5b755c49a4424d98144479f2b745d4 : SUCCESS in 1h 45m 00s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/97efb8346666468fa40ed9244da98108 : SUCCESS in 1h 23m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59bd4d4727784b109fd2f53832e561f2 : SUCCESS in 1h 13m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c4432ec892743bbacef4d5510e27fdd : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4cc0f57d6942ba39fcf7830f8587febbfd35cef7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:21:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4427b27e1d7a7d3648a48e4a5dcee4402fb7de10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85250640abfd4fb185b97d05a6e74914 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6906f9f88df0d3de1680bfe57563594f728cc718","revisions":{"6906f9f88df0d3de1680bfe57563594f728cc718":{"kind":"REWORK","_number":1,"created":"2019-09-16 07:14:29.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/82/682282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/682282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/682282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/682282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/682282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/682282/1"}}},"commit":{"parents":[{"commit":"39e7d0f9569495f09c7fd19cfc0ddad90247afe2","subject":"Merge \"Fix L7 repository create methods\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39e7d0f9569495f09c7fd19cfc0ddad90247afe2"}]}],"author":{"name":"Rene Luria","email":"rene.luria@infomaniak.com","date":"2019-08-26 11:50:42.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-16 07:14:29.000000000","tz":0},"subject":"Fix template that generates vrrp check script","message":"Fix template that generates vrrp check script\n\nCorrect the inline comment to not include an empty new line at the start\nof generated /var/lib/octavia/vrrp/check_script.sh that leads to this\nkind of error:\n\n\u003e  Aug 26 11:49:32 amphora-12184e15-1ec3-4d80-98a7-c7d1ddb6716f\n\u003e Keepalived_vrrp[15265]: Error exec-ing command\n\u003e \u0027/var/lib/octavia/vrrp/check_script.sh\u0027, error 8: Exec format error\n\nChange-Id: Icddd2873abeb56a389a35356995df6dde70872b2\n(cherry picked from commit 905499162b33fbac1fd604688d21b73f559ab70d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6906f9f88df0d3de1680bfe57563594f728cc718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6906f9f88df0d3de1680bfe57563594f728cc718"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
