)]}'
{"id":"openstack%2Foctavia~682283","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Icddd2873abeb56a389a35356995df6dde70872b2","project":"openstack/octavia","branch":"stable/queens","topic":"patch-jinja2-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Icddd2873abeb56a389a35356995df6dde70872b2","subject":"Fix template that generates vrrp check script","status":"MERGED","created":"2019-09-16 07:14:36.000000000","updated":"2019-09-18 13:38:31.000000000","submitted":"2019-09-18 13:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682283-1568813827624-ab427080","meta_rev_id":"781191029552e9d0b4359d2f697270a90426d2d1","_number":682283,"virtual_id_number":682283,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 13:37:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":2,"date":"2019-09-18 05:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-17 16:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":1,"date":"2019-09-18 05:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-09-16 07:14:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 07:14:36.000000000","updated_by":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"reviewer":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"state":"REVIEWER"},{"updated":"2019-09-17 16:03:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 05:34:38.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 13:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76489661903fe831d5a06ae8adc5bb773d4bc9bd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 07:14:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"7de1c6586eceafc2bae282f38c9e2407e83eb970","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 09:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71fe037c05644ed0818084da2abf58ea : SUCCESS in 10m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68584baeb16246118d3145b77899d792 : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/03055c4138da4973a4f50db4af8e533d : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1a8b29879d40f9814d32a47a3b6b77 : SUCCESS in 6m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3d22561de432421abef561ff400998f1 : SUCCESS in 1h 29m 05s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/523973c0a9d04de8aef8405f75274786 : SUCCESS in 1h 27m 02s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57df706b424e4d9882f500aee0f97ea9 : SUCCESS in 1h 38m 54s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/24c427a8a3fc4524894ed967ddb1909e : SUCCESS in 1h 43m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd05fb0a63564a909548beb8b2250329 : SUCCESS in 8m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d5c8fcf809a458f8eab37ea58fedcca : SUCCESS in 6m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1a86eed8b2764a23afd0e1903651ae9e : SUCCESS in 7m 10s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f436ec1f5e4b4a9581648555bc9c5969 : SUCCESS in 1h 15m 38s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f4e291b6e194498fa23b2734c1e4575a : SUCCESS in 1h 02m 25s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/15de8c3e4fea406ebb811a27e3a8337f : SUCCESS in 1h 05m 42s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/ff4411d67d574ce8acb18ff0ce50271f : SUCCESS in 1h 12m 42s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/fe1b42a683ef4795a9bbcfb3e166050b : SUCCESS in 1h 04m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e2f1341bf81417a8dadc64c0598df2d : SUCCESS in 44m 02s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/9ce8bd1903004f0a967220a5b2478278 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a94b637793a74b3fa45f2dd5a4653344 : SUCCESS in 1h 47m 39s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f21e477202a4454abe3638e95fd6d0d0 : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0e60fa615b2386e85b6f5389bcc4b36af11f3033","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:03:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"7c25c0fcc20cbe19d71e8cba477f6dfd217da6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:04:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"209269010b0d96f8a7ea85d7db8f85c916f75c25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:08:27.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"4708d28d4ec246f2e2558aa49c2aaa5c15a10207","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:22:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a0f1f4151594b009900b7506efbb9a3 : SUCCESS in 9m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9858b4a14a334bfb8fbb8fd92237331e : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ae66b86f2c248ad83991a61d76b6f41 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4fa00597b84de6b17c87d2c1824d3d : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98703f8e32074911904701c9b7a17160 : SUCCESS in 9m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4c5e096cd44c489aa383bf8cd2dbe71e : SUCCESS in 8m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d47c194c71fa4190bc17277698ee577c : SUCCESS in 8m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/35d8540b6d1642f38d4ceba5d40be9e8 : SUCCESS in 1h 23m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/67efcf800b394c718ccc1b22aab1f160 : SUCCESS in 1h 19m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c83ed9ebc1f453782ec702afe79b7d5 : SUCCESS in 55m 13s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/137dbc28eb27425aacb4ee35421ec5b8 : SUCCESS in 52m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a51d32e068f44b4689754e94f93c4197 : FAILURE in 36m 24s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/266dfd6ec3d04d7d8bcf95dbc1661d31 : SUCCESS in 1h 26m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d73f0ce7fd074748b7339090dd49e5ab : SUCCESS in 1h 00m 31s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/686543050347494c870224a17b993054 : SUCCESS in 48m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89d60a4057cf4a69b6dcdaaa7edbba92 : SUCCESS in 1h 39m 53s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/6b946155777c4ee5bdf1e852a7ab8de4 : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3708a84c334eaa06bd9f1797b35d3342c438c77f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:34:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0e78aa520adc81e7036bb3976dc498f87ffc4ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:20:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13afc91cdf1445468548da4de480cd72 : SUCCESS in 7m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d39e711e3af4507a9c8cfe97bd0d6bf : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/369dddf7527b499b918fdb65a93dd379 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4fd5561555b4a5abc91479e9a4106f9 : SUCCESS in 7m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/cba08e9edf5542e1b13303298b235ca5 : SUCCESS in 1h 41m 50s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/04dd5241d3534b68be12de9ae2c87b17 : SUCCESS in 1h 43m 02s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba7b84444b9144e7803e447023b5a5c8 : FAILURE in 1h 26m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ce68b3159c694c9cb3a924476bdeee8e : FAILURE in 1h 00m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8c8560cd1344ddb2eb79f3db484cd6 : SUCCESS in 7m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/49394f47b32044c6b907aa47337b74e1 : SUCCESS in 5m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b236f206ca844b33b4cb90e132f18e3d : SUCCESS in 5m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9671f0436c04713892258678db97030 : SUCCESS in 50m 59s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1c3a8d37a5f24017968d975ba0386d62 : SUCCESS in 1h 02m 23s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ef0642d00b8b416197e27463f97e8a60 : SUCCESS in 1h 00m 50s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/da7c457c5fee40b68fa5c94ec4e16198 : SUCCESS in 1h 24m 44s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/423e72bdc0744431aebeca54c9f1f946 : SUCCESS in 53m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7087cd932434685ac60fb73dffd8a28 : SUCCESS in 48m 45s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/04035929354b46e28d6f033f707f0214 : SUCCESS in 46m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0c893f5f8c44773886330e6954e2b64 : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/c9cce77af55c4be8b272c519dee4eaac : FAILURE in 1h 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"656318da67873f6fd443973c5cff0bc17d014883","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 07:23:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64b2ea7fda6f485d1987d0024e0186b0584501c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:32:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc60f807b89c49379b7a45d1a80ffe75 : SUCCESS in 8m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/558e88f9dc1c42afa869fbfaba51cabe : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b72c913587e94866a24238a4e8e86c68 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d49a391dfc4f03b8a8a77d8d16bb7a : SUCCESS in 7m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a43e5c0b488c4b999bf5b4987f797481 : SUCCESS in 1h 25m 16s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d6c86fd873a947b98c4d39b7f163b8e2 : SUCCESS in 1h 25m 52s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/138f562ebc594f05a7279de38fea80dc : SUCCESS in 1h 32m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b379c2ac90bd46ae8deecb537cbb4b18 : SUCCESS in 1h 30m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1008bdd1b01d4117be1781bf85fb8438 : SUCCESS in 8m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f4e0796b98054e35b8d8fd9df9d29981 : SUCCESS in 6m 31s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8856bf65fd9c4159af8d365583899b81 : SUCCESS in 6m 58s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee8e5aa986b24eab8993c39a65082ac2 : SUCCESS in 1h 02m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7f88c823fb1042c7bb075a8aa32e27f4 : SUCCESS in 1h 00m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b594dcfd04b04eb2a2616fb324dbded4 : SUCCESS in 1h 14m 53s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/29f17bb4a6ee4ff49b6ff5cdd11c1e7d : SUCCESS in 1h 18m 58s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/2bff133830a04ac5839f8736b662a75d : SUCCESS in 1h 10m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01274a2c89db4107bcdca4e4bea4b2ec : SUCCESS in 48m 42s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bd4a71af3cd646ca932ee551decb6546 : SUCCESS in 47m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a36d236f160844bbbd54fe94aa69fc5a : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5e25bc5c3c8b4138b0f31476598ba534 : SUCCESS in 1h 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"904c3de1c6d75bd34a9d40b8192a238f54eea360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:32:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97ab442011a94ab87019b3041dd3291cafcdf8ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:37:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"69e791ab14b970341717768e1d3b27b8d12c34c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:37:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f78077ce7aa1456e923740d36b44f706 : SUCCESS in 8m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7948a9a64e9548a39dd5ad8acbf50dd3 : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/95e0ac05c4fc4b5694d6ed4829d4ae14 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f3adc3ad1f4ecfa4120207f59211d5 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87bef6e755d6405e9875e14406618a67 : SUCCESS in 8m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5344a3ac0b014c7b820887ba16626e77 : SUCCESS in 8m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9f71363cc93945d68725a7d3c7529d3f : SUCCESS in 6m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b0154f95ba7b49849541fc36ab6f87d5 : SUCCESS in 1h 18m 03s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6edd90dc71354a478608fe08ff947c77 : SUCCESS in 1h 17m 38s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f01aa50b6d64b6583cf03da3860a4f8 : SUCCESS in 46m 56s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/92c81ba797e14691bdb416b0fad17c38 : SUCCESS in 1h 00m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e488d2d2203f49e0aa7c4ebe00cb5386 : SUCCESS in 1h 08m 40s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7b7c076f1b13471f928ba7663f73e449 : SUCCESS in 1h 16m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/facd8dc095d04e6194ba81f8237efbca : SUCCESS in 45m 45s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/cef9a1330de147bda6e413a59c008835 : SUCCESS in 47m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac37828a091d4d22b861acec669039b6 : SUCCESS in 1h 34m 34s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ca972c19cd1a4ebfa3a4415e08a80952 : SUCCESS in 2h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"37ad3bc4d1ed6cc11e14294a2fcd78c527c416c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:38:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d451318cd4eb40dfb920b147dbaa42ab : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70d97efb4e283264e25c1f5d63fd0707d1217a18","revisions":{"70d97efb4e283264e25c1f5d63fd0707d1217a18":{"kind":"REWORK","_number":1,"created":"2019-09-16 07:14:36.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/83/682283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/682283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/682283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/682283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/682283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/682283/1"}}},"commit":{"parents":[{"commit":"f71539ab0eb9e02e24837325af6a93c5e396692d","subject":"Merge \"Add failover logging to show the amphora details.\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71539ab0eb9e02e24837325af6a93c5e396692d"}]}],"author":{"name":"Rene Luria","email":"rene.luria@infomaniak.com","date":"2019-08-26 11:50:42.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-16 07:14:36.000000000","tz":0},"subject":"Fix template that generates vrrp check script","message":"Fix template that generates vrrp check script\n\nCorrect the inline comment to not include an empty new line at the start\nof generated /var/lib/octavia/vrrp/check_script.sh that leads to this\nkind of error:\n\n\u003e  Aug 26 11:49:32 amphora-12184e15-1ec3-4d80-98a7-c7d1ddb6716f\n\u003e Keepalived_vrrp[15265]: Error exec-ing command\n\u003e \u0027/var/lib/octavia/vrrp/check_script.sh\u0027, error 8: Exec format error\n\nChange-Id: Icddd2873abeb56a389a35356995df6dde70872b2\n(cherry picked from commit 905499162b33fbac1fd604688d21b73f559ab70d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70d97efb4e283264e25c1f5d63fd0707d1217a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70d97efb4e283264e25c1f5d63fd0707d1217a18"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
