)]}'
{"id":"openstack%2Foctavia~682327","triplet_id":"openstack%2Foctavia~master~I6c5ebdea8149166f8d0ebb69cfe63692892f5ab9","project":"openstack/octavia","branch":"master","topic":"osc-fix","hashtags":[],"change_id":"I6c5ebdea8149166f8d0ebb69cfe63692892f5ab9","subject":"Fix openstack port show content handling","status":"MERGED","created":"2019-09-16 09:48:41.000000000","updated":"2019-09-17 05:17:01.000000000","submitted":"2019-09-17 05:15:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682327-1568697322655-7b169225","meta_rev_id":"755bc3e4e6747719e216d93b82650ee7b2619963","_number":682327,"virtual_id_number":682327,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-17 05:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-17 03:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-09-17 03:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":1,"date":"2019-09-17 03:16:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"date":"2019-09-17 02:12:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 11:26:47.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-16 11:28:12.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-09-17 02:12:33.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-09-17 03:00:05.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-17 03:16:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-17 05:15:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72b7c01829e376b309b3131bca43f73aafb42142","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-16 09:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32e27cf8480d38c6293f4fb8058280b0d3ab523d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 11:26:47.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4d1e65967e61f3023b30d72c9339a1d8b3aa28c3","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 11:28:12.000000000","message":"Patch Set 1: Code-Review+1\n\nLets go with it!","accounts_in_message":[],"_revision_number":1},{"id":"fb8f2011ee9f37bc6433e33401716a013769bc3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 12:03:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c53a6fe1a444601b8f1606e346c53dd : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06e0d8b5d7a449c88a8ef1f276012be3 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e60fd7d7fe440fe955a6f8f1858f9a9 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49d148a7426342e9a9c6295151d56b4d : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf4efc8593d466d85c54b2a3954c7e0 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d75a6682f6a486099b19039dad3fff8 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b358acd8a2c468db96e579558666553 : SUCCESS in 7m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/48ea10766fe34f0990dfcfff361df96b : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f815775f0cc4a7697f532bd5771136a : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2ddc9fc2ee4d4b928bfca66abd4adf0f : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a379bad21be4d9982d6bbeafb58b696 : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/824eef4c42664c7a89c6d6dbf504c5fa : SUCCESS in 50m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6c16db00bc0e45a98c60e9e661057bd0 : SUCCESS in 45m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59c19cdf8513419db235b4d652995710 : SUCCESS in 1h 44m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/36cf6f2041d54968a62630e553c21339 : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5bb642db670747019e3f5c6683e60584 : SUCCESS in 2h 10m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25546741561b4594b74a4fb4feb07192 : FAILURE in 1h 46m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9183f087aa64ba1aa745b3e8f3c80a2 : SUCCESS in 2h 05m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/979afdc85f5740748096bff6a751aecb : SUCCESS in 1h 10m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb16e0f596d34866ae733ea0b3ec70e8 : SUCCESS in 56m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/804bf051b9e849ccbdf0a7d987cd91cf : SUCCESS in 1h 51m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/effd237caa514c1ca43cf3014e35fd71 : SUCCESS in 1h 12m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/74afa1ee0b1449458bdc05a085ae3fa0 : FAILURE in 1h 51m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/389ad216e8dd4df09f99088c42a9b3d0 : FAILURE in 50m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d4b7962216ba18a200ae5d2a8559b1aca78244a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-16 21:09:27.000000000","message":"Patch Set 1: Code-Review-1\n\nI think we should make this backward compatible with the old client format.","accounts_in_message":[],"_revision_number":1},{"id":"d6bb89be9aa939704430848a428d0976859bb6e4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 00:08:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"524ffa079b02aac48c4b18a9d44c8017a27fd956","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 02:17:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44d2023a6f124c7799f91bcd27acfa28 : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8f3e4b4286e4ced95436b45160fa1a8 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafda522b71b47e98f3db6e8fbe592de : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff4309a64a0a4b338d0f084f1a4eef88 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34f7d083eadc49479944654dbacf6878 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc8b9adc544c43d38c1b207c3c78bcf2 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b963252b884a49b9af29e39baca61c : SUCCESS in 6m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dd7dc3432f6540fb8da475d8e48d2603 : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/10fa31ff0f3347e0bfd7b56cacc6730b : SUCCESS in 6m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/95560da30ec745fcacf1eb8f13343d83 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c49ed24fc872456d809a37001561ec6b : SUCCESS in 6m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6f65b56f15744d799d396d6ed382752 : SUCCESS in 43m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ec4ba5d67d914dc18f5b01b600d57807 : SUCCESS in 48m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bd3ac05bedc4b33847ff2aed9d06082 : SUCCESS in 1h 51m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b8701c42ab61496b914b13b316b24dd5 : SUCCESS in 1h 43m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/517a87c14a44431088e6a83f4a4e43bf : SUCCESS in 2h 02m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ed6096d06bf414cb6f6e89e9e25bf9c : SUCCESS in 1h 51m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5245b87f223f479cb585f5bb13a118fd : SUCCESS in 1h 58m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1dac2dc0bc78442288786f29fa73593d : SUCCESS in 1h 26m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d5c88fea29c48108c68297deefdc93f : SUCCESS in 58m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d18e0724c5c4a438c54fa33d84fb795 : SUCCESS in 1h 25m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9faf4a7ca8654bb1ae61099b00c0dfd7 : SUCCESS in 1h 05m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b08a0530cd174631a9f906afaa62daf5 : SUCCESS in 1h 05m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/969cd6002a3c413699da1484ce64862d : SUCCESS in 1h 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17ac8b688065b97a7a9f0bcb4910d589d988803a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 03:00:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c76af554f64b301644ff7b433a26adfc45c684c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 03:16:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"b45335bbd2bffdccd0d4bed852dcab6a3ebacde9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 03:16:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9cdecea2e2bb9495e05426dafe6516d225d9346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 05:15:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87f47ed05d12478a912333cce58d823d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01fff994c2914f3094c8a6087ab8f2ab : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc2cd75037014d20950d41572e1354d1 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbb979f31583407b8f7025ef6b26ff20 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3cf037d2d044aab85de0df44b135516 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44774706dc3843d898679498d4ec9870 : SUCCESS in 7m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f4587ad41434ce281b0e9bbe6290b73 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b45f90f6cc134a679e00a4fa4d3ac95e : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ffa34d182dd846bf8f3855c909dcbf97 : SUCCESS in 39m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f743849ef1c0441d97d3817e20c278a3 : SUCCESS in 45m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d62002e16f14d7bbeb65d06ecb56af8 : SUCCESS in 1h 26m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/53d1876b876348979ef78b2314378ee8 : SUCCESS in 1h 57m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5de79de4ec7f4aa9ab387bd94437ace9 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9be52a23f7c19a8403c87f6690b652eea39c4fc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 05:15:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f43605af86e006bb9a8bb24fcba28ded2c29a3a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 05:17:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f0c64b3bbc432885a6f3723d5f49c9 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","revisions":{"774f202068c0f8c57f3c9906d07c650f37937a3b":{"kind":"REWORK","_number":1,"created":"2019-09-16 09:48:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/27/682327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/682327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/682327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/682327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/682327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/682327/1"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-16 09:29:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-16 09:29:36.000000000","tz":120},"subject":"Fix openstack port show content handling","message":"Fix openstack port show content handling\n\nopenstackclient 4.0.0 introduced in\nI9878f327e39f56852cc0fb6e4eee9105b7141da9 a new format for displaying\ncolumns with complex python types.\n\nIt breaks our devstack plugin because we rely on \u0027openstack port show -c\nfixed_ips -f value\u0027 to find the ip address of our management port.\nThis commit fixes the parsing of openstack port show command.\n\nChange-Id: I6c5ebdea8149166f8d0ebb69cfe63692892f5ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f202068c0f8c57f3c9906d07c650f37937a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f202068c0f8c57f3c9906d07c650f37937a3b"}]},"branch":"refs/heads/master"},"926179c97dd2f86ad3476a7f63b10ed121d9bed1":{"kind":"REWORK","_number":2,"created":"2019-09-17 00:08:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/682327/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/682327/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/682327/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/682327/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/682327/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/682327/2"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-16 09:29:16.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 00:08:35.000000000","tz":-420},"subject":"Fix openstack port show content handling","message":"Fix openstack port show content handling\n\nopenstackclient 4.0.0 introduced in\nI9878f327e39f56852cc0fb6e4eee9105b7141da9 a new format for displaying\ncolumns with complex python types.\n\nIt breaks our devstack plugin because we rely on \u0027openstack port show -c\nfixed_ips -f value\u0027 to find the ip address of our management port.\nThis commit fixes the parsing of openstack port show command.\n\nChange-Id: I6c5ebdea8149166f8d0ebb69cfe63692892f5ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
