)]}'
{"id":"openstack%2Foctavia~682420","triplet_id":"openstack%2Foctavia~master~Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2","subject":"Fix the amphora no-op driver","status":"MERGED","created":"2019-09-16 17:07:23.000000000","updated":"2019-09-19 00:19:13.000000000","submitted":"2019-09-19 00:17:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682420-1568852244644-fa3c3f25","meta_rev_id":"47de34722026cb519268dd464d9543ca2b547e37","_number":682420,"virtual_id_number":682420,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-19 00:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-17 03:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-17 16:27:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-09-17 16:27:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-17 16:27:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-17 00:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 03:40:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-17 16:27:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-19 00:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"029726f466b0f6e21298215b25a0188ba2b86b3c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-16 17:07:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3bac7ee1eae159526c99f886f873feab81800f5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-16 17:07:37.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"5ee255c27c2a307b4f62330e19e8743520b03abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 18:37:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdede4c55af54551bcb80e4fe5e2fa8a : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79aed0415af248ccaa546f4ca6789846 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd5a9f041be4176a253c0869b6d4982 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/181e76cf16b84d6fac682c8088c51fb6 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e640c1b138a41c18f042abd9b3d2c8a : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/939c480eefff4484a18c916260013c89 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82399198ca434420b544370b412fa672 : SUCCESS in 8m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb4ffccb7e9044c7b1bfb96ae65d1428 : SUCCESS in 11m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/81522513a2d24894bbc0f5c2eb3070ad : SUCCESS in 9m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3c935f7b674242afa87f83deba277bc2 : SUCCESS in 9m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eac9cb8c99894ce1bd874e68284ff659 : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4916cafbb8694ea5b152bf0e866d0dd3 : FAILURE in 29m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0440d4cad84c49028588c1daf2003d78 : FAILURE in 30m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/232fe011f3b84e7685849b8cefd9e067 : FAILURE in 52m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f4173ba9c44a4c23ba9e4602deccef44 : FAILURE in 57m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5ae40c6b96e4485abef3f4dc9a7e04df : FAILURE in 39m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d03e1fd612034f1a8b9c499adf7a1181 : FAILURE in 44m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce574ee6110c48afb159b142df065312 : FAILURE in 41m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/884b6c2840404066b1f0bf33016db42f : SUCCESS in 1h 20m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1269c4d5269241888a744850c176224b : FAILURE in 42m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c79c65078c584c1fb41df4d670a5193b : FAILURE in 43m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a8c8b72eeff415ba82378b840557c2f : FAILURE in 40m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ab37bf7a56ca4415b3b42ee2bbe521ea : FAILURE in 34m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6156e3d8993745d6bc761d9550638f7d : FAILURE in 51m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31e9e57808b51569f090130931c72f2100c01aa6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 00:13:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c8b89e831cbc845d35eb3f8fd8f483064480354e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 02:46:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08dd55571dc54baab85c836217ae7031 : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b75a3854edf4725a260acbe2566ee8d : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7efbd3fbd3764226b6fbe6de9df9f194 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11c28a55d68d42ab8a375f40f2169ad3 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/349dda212d4f4fb3b3a5a54fbd3d1a73 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e5930e4a82a4d2b9375a23b265d128b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef45e2114704e50949d6376aaf636b6 : SUCCESS in 6m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea58a6dde7d545fca0713e2be330fea4 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3907507c789b4e6f84527a0a1d2c2851 : SUCCESS in 6m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3d2269b271024ee393f550eab1e572cf : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8654183479654765bae24a89e5150e49 : SUCCESS in 5m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1fed4282ac548468a892fd93e36d304 : SUCCESS in 49m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6f4df3ed73d3404da830078e2f16140a : SUCCESS in 46m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3001d9b719b843e581f0b099fc4d91f5 : FAILURE in 2h 30m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2e45e55df9b5487b879951f36205cd8a : FAILURE in 1h 57m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c863b015c973448c992aa39a8c0563c3 : SUCCESS in 2h 15m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17794db888f743808059f363c0d9f0a0 : SUCCESS in 1h 13m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/623f86fa31404123baac99d0525c3c5b : FAILURE in 1h 37m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37726af20dab4ecbbd12690cbbd82a25 : SUCCESS in 1h 22m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/088558a5acdb497f9b550cacfa02e235 : SUCCESS in 55m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8048cd236aef42858ea21c6bc6879e14 : SUCCESS in 1h 37m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52bcc36afc9b4baface6241d9441440a : SUCCESS in 1h 14m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d71b3f1d2da54260828fe01130fe8c50 : SUCCESS in 1h 13m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/48020acbd3474e609d07313f9964c8e9 : SUCCESS in 1h 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a70dd5f1c97bb16ade04aac54029c4982798861","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 03:09:17.000000000","message":"Patch Set 2: Code-Review-1\n\nUmm, are you sure about that? I guess you may be right, but then looking at it, there\u0027s a couple other places you missed, with at least two in octavia/amphorae/drivers/haproxy/rest_api_driver.py","accounts_in_message":[],"_revision_number":2},{"id":"7b428ca834bfb28595b454b1429c172c560dad1c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 03:40:40.000000000","message":"Patch Set 2: Code-Review+2\n\nIt\u0027s confusing that we have the other name, but technically it\u0027s not directly related to the bug. So, I guess this is fine.","accounts_in_message":[],"_revision_number":2},{"id":"25ebc21ffa7e344c5f4519edd81b9564f3c3dc63","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 05:34:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b17766c763617cfcaa8932aa6098de6d120ca662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe966e3569fa426da7b0aa21a177fc76 : SUCCESS in 7m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc0f44330a1346e8b7029aebe10aa6c2 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee489c80a73a423ab3290665c85aacd0 : SUCCESS in 6m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85487f665390431983d39118e7bd9844 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed96e6781ad484a895801924b6537fa : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf52bdeb7d3a4e58ae1930247bc4f4db : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c8e1069ea94a2b8835dda1f5d5dd01 : SUCCESS in 7m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/57d5538cbe654100bd7e4869f59d71b6 : SUCCESS in 4m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c00aa81524a421e8b7f4b01de4ba227 : SUCCESS in 7m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/151356a7f64a45afaeb40c0193584af5 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5e017499b5d403abc5e3d5fa02a5db7 : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18a9af700f2b41da9ae2c4e21bf25521 : SUCCESS in 48m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5a8c7f039a5b492b93d3ece36e258255 : SUCCESS in 45m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15c0212cd2054550b9129842517cc033 : SUCCESS in 1h 43m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c6af053282eb4fe2b19e9af2e2f6fe59 : SUCCESS in 2h 02m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/47ff9621bbc040e781cb2446d6a3769f : FAILURE in 39m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac36cfc0858848fb809ca95dbe570b52 : TIMED_OUT in 2h 33m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/19ecfdccfb2a44f0b5a1bfa4cbff7f2b : TIMED_OUT in 2h 36m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52d81ff1cc7c4770b2ed6609ef3a8818 : SUCCESS in 1h 34m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c39c469fd66d4bc6ad9a4a82c6423260 : SUCCESS in 1h 16m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5b8e29e9baad4cf4976e18b513b994b6 : SUCCESS in 1h 49m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b74a800a537e43f0ab26c35d1c15ca15 : FAILURE in 1h 19m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b45dca5ddfb04dceb05adf370fa2ca5b : FAILURE in 59m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dce7f1b78b624ea593dfa6127e86fcda : FAILURE in 1h 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fce0b7064178ad5f5fbd8739391611d1ed39abb3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 16:27:27.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"80ee94eafc657dcdd1ba338a272f1e4ee902fffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:27:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ffea62dff8d14ed6f32ffda6a42e969decdb51b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:26:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/311ee11dbaa34966af85e823bbef98b3 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e42489b81a44d8b72fef5465f86af8 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/441fcfea222740c48424aa96d1034f09 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d570a7c0f59402cab83e38c4372d96f : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2009a3c5590d4919afbc702ddda5a53f : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cddd62e37aa3473b91922d62d920c3f9 : SUCCESS in 7m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/68a8b2f14b154d09bfdb5be12ff15d9d : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2126573ec9774fdd8493f389c613ad22 : SUCCESS in 6m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a37212854a044f69d5986a1a89c12f9 : SUCCESS in 49m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e52219b8ca874b9da6753a4ef580c9df : SUCCESS in 55m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/324af7b8d3f44c098847b068ff543c95 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a7b0a251443e479b86797bcfaf1c1664 : CANCELED in 1h 06m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2014f07263e1445ebd50dfc66a2ae066 : FAILURE in 1h 06m 50s","accounts_in_message":[],"_revision_number":2},{"id":"c5aa7586bb149d72bc5fe67660e248b75bdc408d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:34:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"69f8a32b3f3b64421f2bc84a9392cd5d88635f5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:13:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0902bf351243beab094486394731b8 : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7a192cc26e447d0b847d280372d73fa : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16aec30894384f44a384389d60bc0e3a : SUCCESS in 7m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/938698e3bc1d4bf2a17670b3bd77daf0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b0a882f83a64fab81d03430331b23a8 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/158dd6ffe9b74f92a21ddb7117a8b94f : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6098ccdcb1c34bf4bf91cb2459586857 : SUCCESS in 8m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cabbe4e7e8bb436485dee824d40b6d31 : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a9d2f1420e46490bba2ce981c7484a19 : SUCCESS in 8m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1445d211f4a04acf9b9b2981f3adc6ff : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a903adabe99e4889a94394b6c0f5a319 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d7a85b8c6be45869744b5a825fc8fed : SUCCESS in 44m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/762e0bdea135487696f0a985e3d3c068 : SUCCESS in 58m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b00587199df34eb1944199f4822fc8e0 : SUCCESS in 1h 40m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3ca541ff0f8a497884befad09337a1cb : FAILURE in 38m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bad36b8092804f7599cb6ebaa0d8dd9f : SUCCESS in 2h 15m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a5971480661476c88a8e711364014ca : FAILURE in 2h 25m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2057cd804ea247a096980425229f3b35 : FAILURE in 1h 58m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a6f6dbcdaf564192ab196dc6b1e17073 : SUCCESS in 1h 21m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8d6fab9897044a482140fbe6431b517 : SUCCESS in 56m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/507d7f3b86324d2b93dfb526c71db84c : SUCCESS in 1h 26m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67e533a6040348a0b07084428118a9e4 : SUCCESS in 1h 14m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a257fff1b99d472084088c637929d826 : SUCCESS in 1h 16m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0f82982e847949fd902427c761512ea5 : FAILURE in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffaa132620252ef12d7605a3deea437e5bdc0f76","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 18:32:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb765fa8cbb8482b8a4ba89dd5e30228d6cf9876","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:15:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f94ab06991ce4b5e9d63aecdc5e929ab : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88875f0beadd4d888129be8a04c4f920 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b5e3bd2739a4b3cb625707081c6a31b : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/525a2870a21145d99355a7f35fbb4221 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d7a72588fb4bf9b917b03c43a0fd41 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2535dd4020c944309a8db0e2193e2022 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6798ab4996174d85bb0ce313be0c7e5a : SUCCESS in 6m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0702eaa190f44c1e9edef70b45f6abb3 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62a5535a7191400cb245e30ed35e9809 : SUCCESS in 6m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b9fd52a1e590447aba3c1743239512c0 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/64e0058b720f4de1b64c2526ab8389e0 : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa884a0343f34bee9cdae82fc8240ea5 : SUCCESS in 45m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9d4d33f794ac4a7fb028833ef0e08fc2 : SUCCESS in 48m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/196be39dc561427d96a6af007edfe71e : SUCCESS in 1h 35m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1a1daca33a8f43ebb2ae5b799c20eb46 : SUCCESS in 1h 45m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f70b80ef4dec4b21915d4be22e2d8a92 : POST_FAILURE in 2h 29m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edcefcf678884c799ca55d6a407c5591 : SUCCESS in 1h 23m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4e1e7017d134891bdcc29404f8d944e : SUCCESS in 2h 01m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ce932a4143a4ee49333e161d1b1fe82 : SUCCESS in 1h 24m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab68dcda68844b83bf16110ae44d6750 : SUCCESS in 53m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b0cbddb58a0c4428804024991fc2418a : SUCCESS in 1h 30m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40ba437122c54fcb8c908ba6005951ec : SUCCESS in 1h 18m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/75eb6056750d4712924eab3f21294918 : SUCCESS in 1h 09m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/26468fd8cbec4866bfe1bde5b6b8048d : FAILURE in 1h 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"104387631bad35b21b21c41e0e199918f6d7fcd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:16:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e43aa886234c4e240d6cbe69c95470b6a64c881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:50:30.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"573869f111a4c2b0fd4d49917c60b37645471921","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:17:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a06641b46f8a5d904256a998c1f93b7d13830733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:17:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a292cdc112b4a6ab19fac21966fecaf : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f84315a8ab3b44dc9eb0c5ce8e1e5ef1 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16bc9b9a41a74377a7ad7dcd2c843bfb : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeae447a8f0741169a5daf1b4dac7f4e : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e4b4351fa774d86932b18342086b0c6 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0998ac14d96f4e32bb4248e0e9f532c0 : SUCCESS in 10m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f98dd1f0ba8a4472ab98cb05a8faaf66 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a9d9f8f78694dae9ee2e75c93350da4 : SUCCESS in 9m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d12d4a363536434098e015cd7a8395e0 : SUCCESS in 51m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/35c74e1995ee4b808e46f79679214a00 : SUCCESS in 54m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3c8b9db764e4b29907052bae7766ddb : SUCCESS in 1h 45m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/84200c9d907d4ef3bee851d103438f70 : SUCCESS in 1h 35m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5afb1ba7bef4c5db16ea4ae244e7c1c : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":2},{"id":"47de34722026cb519268dd464d9543ca2b547e37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:19:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9bfd6ca7e64f3e946d7f808d7267c9 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5defc1e8a4adc6c8a8420b2005691845d0121939","revisions":{"55138cd1edd1b1c4f53dcd37607681096eab8960":{"kind":"REWORK","_number":1,"created":"2019-09-16 17:07:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/20/682420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/682420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/682420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/682420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/682420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/682420/1"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-16 17:05:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-16 17:07:22.000000000","tz":-420},"subject":"Fix the amphora no-op driver","message":"Fix the amphora no-op driver\n\nThe amphora no-op driver had the wrong method signature for the\nupdate_amphora_agent_config method.\n\nThis patch corrects that issue.\n\nChange-Id: Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55138cd1edd1b1c4f53dcd37607681096eab8960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55138cd1edd1b1c4f53dcd37607681096eab8960"}]},"branch":"refs/heads/master"},"5defc1e8a4adc6c8a8420b2005691845d0121939":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-17 00:13:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/20/682420/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/682420/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/682420/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/682420/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/682420/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/682420/2"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-16 17:05:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 00:13:09.000000000","tz":0},"subject":"Fix the amphora no-op driver","message":"Fix the amphora no-op driver\n\nThe amphora no-op driver had the wrong method signature for the\nupdate_amphora_agent_config method.\n\nThis patch corrects that issue.\n\nChange-Id: Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5defc1e8a4adc6c8a8420b2005691845d0121939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5defc1e8a4adc6c8a8420b2005691845d0121939"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
