)]}'
{"id":"openstack%2Foctavia~682812","triplet_id":"openstack%2Foctavia~stable%2Frocky~Iab9fcca6e4df88b26e4d88003e2c2cfb48d90b5b","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Iab9fcca6e4df88b26e4d88003e2c2cfb48d90b5b","subject":"Do not run non-voting jobs in gate","status":"MERGED","created":"2019-09-18 07:32:30.000000000","updated":"2019-09-18 13:49:36.000000000","submitted":"2019-09-18 13:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682812-1568814298046-d2cb535d","meta_rev_id":"185d7fbd629fa3f762faf61756db129bd068c273","_number":682812,"virtual_id_number":682812,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-18 13:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 07:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-18 08:03:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-09-18 08:03:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 07:39:08.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 08:03:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-18 13:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"422fd9787adfadb6e08b6399822d5eca8772ed29","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 07:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a58397a112b8dbdb2d1a2600ffc6ac7d52762ef2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 07:39:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff540121bc145879feae81ce475663782c8f27dc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 08:03:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThe multinode jobs were temporarily switched to non-voting in https://review.opendev.org/#/c/556549/ until issues were figured out. The job failure rate has not improved since then, and last runs have all failed 100%.\n\nhttp://zuul.openstack.org/builds?job_name\u003doctavia-v1-dsvm-scenario-multinode\u0026branch\u003dstable%2Frocky","accounts_in_message":[],"_revision_number":1},{"id":"3ca857d4c5eb1f19640fd349fcef97edc7230316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:02:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d4453f1bae948dab8fb6c71db4c7132 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9115e7a2b994cf7afef1a3e7e916bdb : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65c86101fb2b4fe5b414eaf40fa1a221 : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83a425ca973a4df4a8a75b30a33a9b49 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac93b45ea7804a70a77b00bc080b70ec : SUCCESS in 7m 51s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/116d2faaa9ab412ab80d09253f1387ac : SUCCESS in 1h 22m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f29f26ef44bb45e39676d4aa3373ac0c : SUCCESS in 1h 22m 17s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8428005e6fa34b94912f194bcd771021 : SUCCESS in 1h 30m 48s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e03db1b1e67e4a289ee8a0bfbfb111e5 : SUCCESS in 1h 21m 40s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/388346da310e41878e662ee63d20e4a6 : SUCCESS in 4m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/693c2964b06f40159479b82f11e43e07 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f297a5ec98194158adb94ddb1cf5aae2 : SUCCESS in 7m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a01de290a7d44159be17645af34291bc : SUCCESS in 1h 17m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8d9763f3d7b544e8a7e44a5b42635c46 : SUCCESS in 1h 05m 05s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/dd1c89bf83044cc8a8f48d8f441fa9c2 : FAILURE in 1h 12m 59s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/38e54ffa92d647afae0c5f12b07a7dcf : FAILURE in 1h 30m 17s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/5b9ae9ef82834333959a4cd9ab3a8397 : SUCCESS in 1h 13m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c2671682456a4d92a201eedf7a986d64 : SUCCESS in 54m 24s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/3e9eec8dade04cfbb2d50a2bd4cf5a2b : SUCCESS in 50m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ac2c096c2844e4ba781d7fd19d31ce7 : SUCCESS in 1h 31m 08s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f975851b328749d38f2e8b6d33a6d1ee : SUCCESS in 1h 43m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ade4f4c962954f5bb953ff475d25b1f1 : SUCCESS in 1h 19m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67228562103f4efca9564ec17bec0b33 : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":1},{"id":"59853ec6680f8154c8ed210370ca55b45e8806c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:02:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"acb764f28c0dc9560f2f7521d2d06b31adb1b2cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:44:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a9ebb1e52064db19a9cfe691c33183a : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c031ce0c1c8440d297ad0ae5ae6b87d1 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c42ab6eaf52d4be98c90a1f4108e54e7 : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57be9d7344324340bcf1093785945857 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d7aeda123e4dcfa1f76bdea0689ece : SUCCESS in 8m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/87ff47428b294b49a188780f0fd72f70 : SUCCESS in 4m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/52a87ef2a7364beba74fb162539a3af4 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/effaf8b6b71047858da3089bc74ca53c : SUCCESS in 5m 12s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73477aa3791e4371b35d62ab4cb556c8 : SUCCESS in 1h 01m 26s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1bfd5bd8e8e144c59be1c543ca78c498 : SUCCESS in 1h 05m 13s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8e6c38cb38ee419e88ca40d0423fbf2a : SUCCESS in 1h 21m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c11b4efae5043eaa8f55b009621dfe1 : SUCCESS in 49m 37s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a9e02c1c4a0a498fa5efa1f18aa70315 : SUCCESS in 51m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caaf8cfe8b9141a98118e7ba2cf2b3f1 : SUCCESS in 1h 31m 17s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b922a4308039443a95fef8b6512f8809 : SUCCESS in 1h 39m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/169fe3694fc34005a03dc2be9f98d5c3 : SUCCESS in 1h 12m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33af2864a180408faf6fc9ba33dfff81 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"64c9f45022aaaeb59246d65d16c678d31ba20d64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:44:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"185d7fbd629fa3f762faf61756db129bd068c273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:49:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c608bfe4c04414a48346ed67af56ba : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf0fa1d7dd5adac4d80c44a088d4ccb1d5161648","revisions":{"cf0fa1d7dd5adac4d80c44a088d4ccb1d5161648":{"kind":"REWORK","_number":1,"created":"2019-09-18 07:32:30.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/682812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/682812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/682812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/682812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/682812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/682812/1"}}},"commit":{"parents":[{"commit":"39e7d0f9569495f09c7fd19cfc0ddad90247afe2","subject":"Merge \"Fix L7 repository create methods\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39e7d0f9569495f09c7fd19cfc0ddad90247afe2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 07:31:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 07:32:27.000000000","tz":120},"subject":"Do not run non-voting jobs in gate","message":"Do not run non-voting jobs in gate\n\nnon-voting jobs in gate pipeline are just wasting resources and get\nignored, do not run them.\n\nNote that master has no non-voting jobs, so this is not a backport.\n\nChange-Id: Iab9fcca6e4df88b26e4d88003e2c2cfb48d90b5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf0fa1d7dd5adac4d80c44a088d4ccb1d5161648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf0fa1d7dd5adac4d80c44a088d4ccb1d5161648"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
