)]}'
{"id":"openstack%2Foctavia~682833","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ieab13c1152fed64da7390891b315b5e67513ce3e","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ieab13c1152fed64da7390891b315b5e67513ce3e","subject":"Fix base (VRRP) port abandoned on revert","status":"MERGED","created":"2019-09-18 08:23:52.000000000","updated":"2019-09-18 21:12:43.000000000","submitted":"2019-09-18 21:11:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682833-1568841071082-1830412a","meta_rev_id":"91f2cdd1842d860da6459e89e7b64e86acdce768","_number":682833,"virtual_id_number":682833,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 21:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 18:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-18 13:39:46.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-18 18:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-09-18 08:23:52.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 08:23:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 13:39:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-18 18:30:30.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 21:11:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80b5a4bf2d2bece1e8ec739847c17785c12e38c5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 08:23:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"837285b4b99fae42d3d7985e277839ded3521b58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:04:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/412ac1776d80492abe607cabd5b1ccd8 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6aa5e8de46443e88b82f5f00412560c : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dfdc247adfe44ecbd8e7904da475e5e : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42e4eb4575ad4436a84623c1dcf18840 : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e6c0e224e8b84758ad477ed47422bc58 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12e548c02a324726a0b84fca5a7c6944 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ee328608b841bd951f260cf50a0320 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/649e0f98809c46e09af6477f667d7d2f : SUCCESS in 5m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d26e66bf50f946959657670956cfa0af : SUCCESS in 1h 23m 36s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6ec201b7c9444da980f9b3a3eada69a9 : SUCCESS in 1h 33m 40s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b2778aecf79443b9dc920cd7b49f57d : SUCCESS in 1h 27m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a1b685c95be241d1812a740ee185d65b : SUCCESS in 1h 34m 55s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f93ff323169c442a96bf09c57579cd95 : SUCCESS in 5m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0daa2af8e4b748148fda675d7e3d9547 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9710e81aa4a34d23bb61eb4bb2a8cad7 : SUCCESS in 6m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e23eb2ec65f240999a4abf562880d2b6 : SUCCESS in 56m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dd435d59f6f64260b3e1e172933b8ba8 : SUCCESS in 1h 09m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d223e2a10664081af827ee4d191634d : SUCCESS in 40m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3045e1d3cf074b4c81207d79aecfbbfe : SUCCESS in 48m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab037caa14d243618fdebd936334fe36 : SUCCESS in 1h 18m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0bb4c0d88e7846faacd8bd384ce913d8 : SUCCESS in 1h 43m 45s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a76ee2591b024dff928181d8147507b8 : SUCCESS in 1h 38m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d8683f109ea945f98bc1449fa33b256c : SUCCESS in 2h 17m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0d8209e16c64c17934930c7dc5b072e : FAILURE in 41m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9d463d21ba243e6a00b1177e8b630d8 : SUCCESS in 1h 57m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba89239f7e334273b0edea2bd63252f7 : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7af9853f9e24142b840fedd8d730a0e : SUCCESS in 1h 18m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9dd32cc340e949718712c23b43b3d7c0 : SUCCESS in 57m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ba67d71bd3f3b14b6ccf8811efc331739ea175d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 13:39:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d0807ade31274c2fda67dc5893ff6ee1d329afbb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 18:30:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86358d58ea0f2f215a844c5108a431cb8780ec3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:30:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16ec541a83df07e7b73d8745dda345d4006969af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:11:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ad3d1b19f554481b8d91c0439e9ea24 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788bf75b8d8e4ab4b627d6d3f9728fd5 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bff67eaf7ab54c959591e30500754440 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/007b240183224d4e9895bdeb6442a943 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b11e01e24e4668a329147301263852 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2901b5af8be24136801a6352ddcbd21b : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3612ca652fa94f5b961af8d10bd1f20f : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c3253813667640fd888aae63175d57e1 : SUCCESS in 7m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6043d3d566b14268a723ea41779a0440 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dd4fb65faf142e7a145b7f29244e7f1 : SUCCESS in 6m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c95f72edb3e4079a79ae79777a7322f : SUCCESS in 1h 08m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cfdd843497a84bee8c209a147982b342 : SUCCESS in 1h 04m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/24d5bf57dff9454991e654421280115d : SUCCESS in 1h 26m 04s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a912dce672f14299bda666d3ec3134ae : SUCCESS in 1h 24m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27c2b26bf0f34f21ae1124830cf2cdf9 : SUCCESS in 45m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/eac7cb97c23149c98062dbcaa431928d : SUCCESS in 46m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae4000511e02421ca66e5d0f95d775a7 : SUCCESS in 1h 33m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3306685e80fd4cdc8aeb5122eee401f1 : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a5038981e7fe4248b167a7e981eac58b : SUCCESS in 2h 10m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67cfe28616974c2fbbd5ccb52bea2163 : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d92d85c7501b771126f93bb86edea54081540757","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:11:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"91f2cdd1842d860da6459e89e7b64e86acdce768","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:12:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ca19322fb24acaa94e3b2c05017cd3 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb15da1d8f0843738fcbab6256e98aaf : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"701a900164689aa88549232961f91c7d9c38c03a","revisions":{"701a900164689aa88549232961f91c7d9c38c03a":{"kind":"REWORK","_number":1,"created":"2019-09-18 08:23:52.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/33/682833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/682833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/682833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/682833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/682833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/682833/1"}}},"commit":{"parents":[{"commit":"cf8d68efaa0ca52f5f906f0d174ab84c7029b71a","subject":"Merge \"elements: add arch property for ``open-vm-tools``\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf8d68efaa0ca52f5f906f0d174ab84c7029b71a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:31:07.000000000","tz":-420},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-09-18 08:23:52.000000000","tz":0},"subject":"Fix base (VRRP) port abandoned on revert","message":"Fix base (VRRP) port abandoned on revert\n\nIf the base (VRRP) port fails to attach to the amphora instance, the\nport would not be deleted as part of the revert cleanup.\nThis patch splits the two plug VIP phases and attempts to clean up the\nbase (VRRP) port, that was created in the first phase, should the port\nattach fail.\n\nChange-Id: Ieab13c1152fed64da7390891b315b5e67513ce3e\nStory: 2006468\nTask: 36399\n(cherry picked from commit f20dc158441faad8422f7836dc7d0af50dcbcb0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/701a900164689aa88549232961f91c7d9c38c03a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/701a900164689aa88549232961f91c7d9c38c03a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
