)]}'
{"id":"openstack%2Foctavia~682836","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ieab13c1152fed64da7390891b315b5e67513ce3e","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Ieab13c1152fed64da7390891b315b5e67513ce3e","subject":"Fix base (VRRP) port abandoned on revert","status":"MERGED","created":"2019-09-18 08:45:33.000000000","updated":"2019-09-19 19:04:19.000000000","submitted":"2019-09-19 19:02:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682836-1568919757055-844a795a","meta_rev_id":"bddfc46c5758e93ed1cb1b0142283e775c1e23cb","_number":682836,"virtual_id_number":682836,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-09-19 10:00:06.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-19 19:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-19 05:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2019-09-19 07:21:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2019-09-19 07:21:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 08:45:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-19 05:43:25.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-19 07:21:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-19 19:02:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05c768eb6a0511fac6608aaa102cfa03a599d9f7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 08:45:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65f5c14114c61b976797d123e6aa5282665d9306","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:16:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7ce9e773070472b909b26467b989ec9 : FAILURE in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b27b7f744b8f4ac39da5e8923d457a3d : SUCCESS in 7m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c19bc7ace50c4b4db0fb6657a60ef886 : FAILURE in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa24be41775b4e03bfda6c0967a36126 : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5746dc549744528026e5250fba029a : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e836ad9ff2234d38b4955ea0e98cf0a2 : SUCCESS in 6m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ef206276378342cabf7309e886e30b43 : SUCCESS in 1h 16m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e05dbe8f644f4ab4bbedfae75ae4f751 : SUCCESS in 1h 15m 38s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07b1e108bc2b40189fa1d903a0eb467e : SUCCESS in 1h 32m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/490ecb60d8384ab89ab50cd34d10b68f : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9439c5c8b91a4cbda3c1012ffa77abf0 : SUCCESS in 5m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/93883bc365134a688d7bfc5eb9b23476 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79350860bd024ed0b7c0fe7ad0ded19e : SUCCESS in 5m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5600783273874639b3a3ec0c5d442fcb : SUCCESS in 1h 02m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5715db35477841629b81ddf251c9a99e : SUCCESS in 1h 04m 28s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/25e70d6f301b40678aa6f5999d32678e : FAILURE in 1h 30m 25s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4f72131162e8427182a6348c458e28b9 : FAILURE in 1h 28m 43s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/bcf645a2908144d39a1818f26d0cdd76 : SUCCESS in 1h 09m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c5be5aba2514005a47abd6f5b40ae79 : SUCCESS in 54m 09s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/874479feb5ed43feb0e1bd03dcb8cb3b : SUCCESS in 50m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/115255899549447795e45847db128b64 : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/fa4d785673ac47e6a487f7415f6cf1af : FAILURE in 1h 42m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f335b9ddf8e4a6088f40b8b07813879 : SUCCESS in 1h 14m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf070102dd4c43f79a8656cef05d3bd4 : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"56c358f7b8d5cb90f7adcc07872f657c0306e935","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 12:32:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e5977768d5bb3f835ed6271b28695fff53a3ac14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 15:28:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc753b0a7db7462bb720898fe537391c : FAILURE in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e933e1b52e9c43c4954c621d14c169d3 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d44393445f16406582b8719f7e673624 : FAILURE in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5c4a7b131100488ab9f0962d1b4c9729 : FAILURE in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ebd9875f6e40469b3b3aafe19bdc33 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/127ea402acfd4430b16aad5c0edcdfe3 : SUCCESS in 4m 51s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/82d8741304f546569483d8b37d4554c5 : SUCCESS in 1h 08m 48s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a11b6dbe25d54c41886e4071623d3d9d : SUCCESS in 1h 44m 44s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72b5dd172b894ac497048544b02c7c35 : SUCCESS in 1h 46m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7e59aa4d91a347e9879c72ea2eb06773 : SUCCESS in 1h 39m 10s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/786684b7aac94e84b70b5be441bf2173 : SUCCESS in 6m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a3fce9d91593481ab79aff4ad0234bdf : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40bf0c2da77e47568b5950f9e2e24282 : SUCCESS in 4m 42s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03e4420fa15f4823934034e00cd876ee : SUCCESS in 48m 26s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/714b4b5ae07e4d85b0e0792c8b33a894 : SUCCESS in 56m 21s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e414e54de2f84ae8888cbd0061b2a4e8 : FAILURE in 1h 33m 06s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9e2abcc7ed5c4ac39607b978b2358230 : FAILURE in 1h 42m 26s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/13925f7450c54bf6a83c81c95b1ecbf3 : SUCCESS in 1h 12m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c7faac26dfe464daa724ef354d8a95f : SUCCESS in 48m 47s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/b4d649e3b4194a179dd04fd1288ba58c : SUCCESS in 48m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47b30892f89b489f8b53a58cdbc4e587 : SUCCESS in 1h 32m 47s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/e0e74ded6d2348a7a52fd418f8acf65c : SUCCESS in 1h 37m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0b5922d731e4d66bbe14227825bd8e0 : SUCCESS in 1h 12m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad3f272ea1114328b9b9c63db868e451 : FAILURE in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5ebd7fc266ea4283a651c23e317f1b4636043026","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 16:17:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"943de4179cd4ac15983c4624195a1657836f58f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:03:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6796a44bc72b412f845c7f00e237382a : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a174311dad90450a9586a0972933a245 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c50413e7cb8421c83354a5f7e1a4a85 : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef6bd0dd2b634edebe5a43d38e07f6a3 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72f3f9a80be4eea8f70fc888862bdb1 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bc9981368ab41a689db15cdfcf05959 : SUCCESS in 5m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0de6de1696ea4d279c46c58b4c6fc1b1 : SUCCESS in 1h 15m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/592fbe83bd7644d29d861c1249855801 : SUCCESS in 1h 21m 38s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4891d1c4893947758217c74b7220cd07 : SUCCESS in 58m 50s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/75c732420a0745949900a01fcea0c2f7 : SUCCESS in 58m 33s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6d6b96bc18db49e982fd4bc61f7c687a : SUCCESS in 6m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/74047af1915a46e4ab5ce002c728062e : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d77aa12e5efa4acb8a3a32c897a57a8d : SUCCESS in 5m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/968d2b2c2d10434a9cc4fb554b5fce5d : SUCCESS in 1h 03m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e9be1b3577ff4e1e9ff09ad814b577a3 : SUCCESS in 1h 03m 58s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0e5b1292dda94ae3ac0bc1dc62517a0d : FAILURE in 1h 25m 43s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a23759493d5b46b9ac71f31ec2911347 : FAILURE in 1h 41m 58s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/a474adebfcc644bf8f732cdfa04a965c : SUCCESS in 1h 14m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44aecc6bfc634f06afb0181a0fe19105 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/fa78633613ea42e5adb51a59b325653d : SUCCESS in 44m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85717cabba0c4e9d848406804343360e : SUCCESS in 1h 38m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/aa0fefe054574c8e967926a55413ec07 : SUCCESS in 1h 36m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a2e91aa3d5d44768b63f7ba067209ef1 : SUCCESS in 1h 19m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dab835d461f94234ad26a5bf74fdf6b5 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":3},{"id":"ce56a237369bb25a445f19174d2549e33d2bb9b1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-19 05:43:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c630b6599cb4b7a1881c049139c37a1cc7f8301c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 07:21:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"91d71241e8a8567323d920a59bcc1b8a6e809aaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:21:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"71394e8a0c56def761d2ec1ec224029edd715cc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 09:59:24.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46be3c20e8124701b32b550c813d2004 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8440d6328e4abf995ae7eff78c1bfd : SUCCESS in 9m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf91d7aea81646978085c5c990d393f0 : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/847fbdf5a4a74ef6b5f98ccc0845df64 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15a311f6b4941528b59eb7cf8272604 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fbc09ab718a4bab9e6bb85cd79ae6c7 : SUCCESS in 5m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1242bad78d6c4a94a0d6613134d32d19 : SUCCESS in 14m 53s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/09ba4aa5a8d543ef9a6de4ad84a4a279 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dea4e2bcf35c425da9b6ddcea60ed0bb : SUCCESS in 6m 43s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77e19ba4e4bd4a3e9de624ee7308b348 : SUCCESS in 57m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8c127c4d75ee4697ad50f441472a3a9c : SUCCESS in 50m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ae48460300b2445ab0ca7b7231b9437a : SUCCESS in 1h 18m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d7426a2218f64386982a80b1a7cb8955 : SUCCESS in 53m 34s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/9cdd075926a34afeadf04ead9522b02e : SUCCESS in 1h 02m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca392a4db69341c3b922ee8316acdde3 : SUCCESS in 2h 06m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9414d04b3c144261b6721316f1a5e6fb : TIMED_OUT in 2h 35m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ec002da9e744e2db38873314b0c5a91 : SUCCESS in 1h 05m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4abf8da925c4ac1a3bdd62a7934b6b0 : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":3},{"id":"a910309351792117b710edc416d53ab6270db076","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-19 10:00:06.000000000","message":"Patch Set 3:\n\nreverify","accounts_in_message":[],"_revision_number":3},{"id":"057defbee16f13f4a4f8c75c9b7d572267c84847","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 10:04:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fb9dae497a79a4899f2eb09a2003b3c87bf07420","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:07:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dce363d0139b4cc69c0b50ccdcb6b09a : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad5db92c44a4b6daecf73ce2b2babc6 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da8ce4eb925441ab849f63b611bd84f7 : SUCCESS in 6m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/683ab1c8b8a84927bb0b9de6e5610280 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026590a6d1c648c380e4131757649dd8 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c71f274040974a4daef7307c79dd2079 : SUCCESS in 6m 29s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ac26cddb255848f5a14c67add5604976 : SUCCESS in 1h 16m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c946975fcebe4701a8f938b845b56f6f : SUCCESS in 1h 19m 09s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16b8b4536051422ebac191ac869cbab2 : SUCCESS in 1h 32m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4844918f43124e7da363e147ff28ede3 : FAILURE in 23m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2e7b374b906a4f68b5073d301dc8394d : SUCCESS in 5m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/db58de39aa494a54802bc8adf5ba332b : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e0a7862b8a1446fa1ffc59495952099 : SUCCESS in 8m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be537804ce664cd5a212f034906610de : SUCCESS in 1h 08m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c1813e600e2045afb1b8aba2fc0701c0 : SUCCESS in 1h 17m 28s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/964e200d85944d4aa9b0bced37d5e3aa : FAILURE in 1h 31m 15s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/2d7cd47a2fe34578ac4851a055c1b5c5 : FAILURE in 1h 33m 39s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3bab2a9dc5d046d0a5fe036b610b25ec : SUCCESS in 1h 12m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08f6c20541324b04859e380b130974dc : SUCCESS in 49m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/0e94d126de234f85a48e4284d13e0158 : SUCCESS in 47m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0a1a927cd934cbfa9db75e39227c385 : SUCCESS in 1h 30m 44s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ef9727c247774de687a79fa654a26a6d : SUCCESS in 1h 46m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ce8daac427c460ebf15bd22eaf7dca7 : SUCCESS in 1h 19m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bdf08c22bd14807a16c16a7e882cf4b : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":3},{"id":"39ae3b107b94fdb3bb34305b7630341330f37019","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:08:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6d7c613cc73fc5175ec68add4a070129858a8c6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:02:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/510dddcaf8134a63adb797393b805975 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe641cdc2a804afc8a04a971dbaaa59b : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aac67130139b488bb5dd6fc0b52a0af7 : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53b51c331d0149198e833d02dc9f6eb7 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03561373b6a4e3680398f97a6470b9e : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e459437d73a4fc68f01d32589591a0e : SUCCESS in 4m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2e31685281ee45a7b86aa9db04fda45f : SUCCESS in 5m 55s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ead3dd99ad18484a8880df39f9db05e4 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8bc7f9afb70d46cf9ea92568712b9f8a : SUCCESS in 5m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/896cb64800c04e92b545063afd24d0aa : SUCCESS in 59m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c263cd6491c34238a35e7f0a9d50eee9 : SUCCESS in 1h 14m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/55b6c792b96b41b88bca4497f3ed6cd8 : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf5ddb13cb1a4bdc9da39950d9c3b303 : SUCCESS in 53m 56s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/2a0933b4a5d6491a8f2617ea1f74a2db : SUCCESS in 46m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e0bb277826a48e79f90c0b256152ec6 : SUCCESS in 1h 45m 59s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/e4537cbbe52a460e90744d84233f8349 : SUCCESS in 1h 51m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/30c5fc66e71f420289b5a4c98b8fd8dd : SUCCESS in 1h 15m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2be18112b28e40a7b5f1ad97391f9db8 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":3},{"id":"12739d2b97df094bfc3b29bb0f81b8f6afd0578d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:02:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bddfc46c5758e93ed1cb1b0142283e775c1e23cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:04:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2c622702524ea088a6f6bf33e1163d : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5101ba5fe384ab4b7595314d769f52e : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0aace5715c9f1fd98dc02776fa1920163c998e33","revisions":{"bd47d3554012fab6dae4dd34c2008458c93fce26":{"kind":"REWORK","_number":1,"created":"2019-09-18 08:45:33.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/36/682836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/682836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/682836/1"}}},"commit":{"parents":[{"commit":"b0e3a002505f152f2e998e8af0704c6fbd2a945f","subject":"Merge \"elements: add arch property for ``open-vm-tools``\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0e3a002505f152f2e998e8af0704c6fbd2a945f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:31:07.000000000","tz":-420},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-09-18 08:45:33.000000000","tz":0},"subject":"Fix base (VRRP) port abandoned on revert","message":"Fix base (VRRP) port abandoned on revert\n\nIf the base (VRRP) port fails to attach to the amphora instance, the\nport would not be deleted as part of the revert cleanup.\nThis patch splits the two plug VIP phases and attempts to clean up the\nbase (VRRP) port, that was created in the first phase, should the port\nattach fail.\n\nChange-Id: Ieab13c1152fed64da7390891b315b5e67513ce3e\nStory: 2006468\nTask: 36399\n(cherry picked from commit f20dc158441faad8422f7836dc7d0af50dcbcb0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd47d3554012fab6dae4dd34c2008458c93fce26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd47d3554012fab6dae4dd34c2008458c93fce26"}]},"branch":"refs/heads/stable/rocky"},"28bd4a5cefed31c2c57e66961b4ed10098105b50":{"kind":"REWORK","_number":2,"created":"2019-09-18 12:32:24.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/36/682836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/682836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/682836/2"}}},"commit":{"parents":[{"commit":"b0e3a002505f152f2e998e8af0704c6fbd2a945f","subject":"Merge \"elements: add arch property for ``open-vm-tools``\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0e3a002505f152f2e998e8af0704c6fbd2a945f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:31:07.000000000","tz":-420},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-09-18 12:32:04.000000000","tz":0},"subject":"Fix base (VRRP) port abandoned on revert","message":"Fix base (VRRP) port abandoned on revert\n\nIf the base (VRRP) port fails to attach to the amphora instance, the\nport would not be deleted as part of the revert cleanup.\nThis patch splits the two plug VIP phases and attempts to clean up the\nbase (VRRP) port, that was created in the first phase, should the port\nattach fail.\n\nChange-Id: Ieab13c1152fed64da7390891b315b5e67513ce3e\nStory: 2006468\nTask: 36399\n(cherry picked from commit f20dc158441faad8422f7836dc7d0af50dcbcb0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28bd4a5cefed31c2c57e66961b4ed10098105b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28bd4a5cefed31c2c57e66961b4ed10098105b50"}]},"branch":"refs/heads/stable/rocky"},"0aace5715c9f1fd98dc02776fa1920163c998e33":{"kind":"REWORK","_number":3,"created":"2019-09-18 16:17:18.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/36/682836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/682836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/682836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/682836/3"}}},"commit":{"parents":[{"commit":"b0e3a002505f152f2e998e8af0704c6fbd2a945f","subject":"Merge \"elements: add arch property for ``open-vm-tools``\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0e3a002505f152f2e998e8af0704c6fbd2a945f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:31:07.000000000","tz":-420},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-09-18 16:17:01.000000000","tz":0},"subject":"Fix base (VRRP) port abandoned on revert","message":"Fix base (VRRP) port abandoned on revert\n\nIf the base (VRRP) port fails to attach to the amphora instance, the\nport would not be deleted as part of the revert cleanup.\nThis patch splits the two plug VIP phases and attempts to clean up the\nbase (VRRP) port, that was created in the first phase, should the port\nattach fail.\n\nChange-Id: Ieab13c1152fed64da7390891b315b5e67513ce3e\nStory: 2006468\nTask: 36399\n(cherry picked from commit f20dc158441faad8422f7836dc7d0af50dcbcb0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0aace5715c9f1fd98dc02776fa1920163c998e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0aace5715c9f1fd98dc02776fa1920163c998e33"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
