)]}'
{"id":"openstack%2Foctavia~682930","triplet_id":"openstack%2Foctavia~stable%2Fstein~I0ce2eabfd4dd41210312ea3b7f6274c9a6d50e44","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I0ce2eabfd4dd41210312ea3b7f6274c9a6d50e44","subject":"Fix cleanup of expired load balancer entries","status":"MERGED","created":"2019-09-18 15:13:51.000000000","updated":"2019-09-20 07:07:13.000000000","submitted":"2019-09-20 07:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682930-1568963120757-f233d7c7","meta_rev_id":"e3cbf0a459d2a1ab220ed0d59ffc5864c2219f9a","_number":682930,"virtual_id_number":682930,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-20 07:05:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-19 05:41:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-09-18 15:42:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-09-19 13:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2019-09-19 13:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 15:42:53.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-09-19 05:41:51.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-19 13:58:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-20 07:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d69675142e8775fcf28b17a85fa64d422f9cce2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 15:13:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7a24cd3be364030103ee72238e40f8e7d29c7779","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-18 15:42:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"28fcc005396ed86c7059ca0d88f9fe78d4c7182b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:10:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ed2793729f440118d96d8933db13e86 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2069ecde6bfb4d04ad5499c39ecc1dc3 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78325d24e2f44c5b9a471c505a88dc1b : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1be41e8023054e0aabba0a54d511ed0f : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f903a3f760584a3a9c5c845cd4393375 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7120648bbc0476d83a347fbfda06d00 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c6694e28a64fe7b3c1cb3478e80873 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd22f7ed7f61475ba191021dc502cf45 : SUCCESS in 5m 02s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/20d3418c36fc47efb9391f6869dfbbd8 : SUCCESS in 1h 25m 59s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5301012fe1514860925be4a0fe714b25 : SUCCESS in 1h 26m 32s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/479651566f6c42b4922ea76dae9b5b95 : SUCCESS in 1h 38m 47s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fc93fa5b27ec4075b1912fe14c10be17 : SUCCESS in 1h 54m 12s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e03bbdbf6c45496bbe49c23f7dfb2cb5 : SUCCESS in 6m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2754d96848a94a3c8d2160671cab9af3 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fff733e23c54f8db3fbf1b3419583b5 : SUCCESS in 6m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9cd9fb1744b40f38c067eb094182a03 : SUCCESS in 1h 15m 25s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/eeab1ba6687a48c1ae134adcc70775f6 : SUCCESS in 1h 26m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4367c1f8f65a426e8dfa30410b54de3a : SUCCESS in 46m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/96ac8a256e8e4bfc9772cff72b121517 : SUCCESS in 44m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/186f5e705ac445fca4b4c169c23a9b7e : SUCCESS in 1h 39m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3788a2033ac240ecb9281fb54413a51d : SUCCESS in 1h 35m 32s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3c0b7ce2274648de9f8c2a59d836472a : SUCCESS in 1h 44m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c924510a289744c3a113de72adeab2e4 : SUCCESS in 2h 21m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0951b62e91894bddb9791f0242c19cbd : FAILURE in 37m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/164d2adf41e0436a8be2e5a232954bcd : SUCCESS in 1h 50m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c76004c72c0448a4be0a1dd2cc2e7ee4 : SUCCESS in 1h 44m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38dae598347a466e9efd58ffaa256409 : SUCCESS in 1h 21m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49cd7c19c5a04297af3a6bee12755716 : SUCCESS in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43810b8a663891f2bacd55c35c1c7ebca0243d10","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-19 05:41:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f905e4f7d579439a1b901fe6e96166b75c62a84c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 13:58:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"93da3599319e3298b05a1156ed61b101504f129e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:58:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"790519e93e7b83c7b826860b373ef92cba8c80d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 16:38:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48875c7eb7e34dd49002d8a5d6244aa1 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f637b98f0c448a918c42b8d2d03379 : SUCCESS in 19m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0125565eb4a84a929e186e9db6f0bfc2 : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/54798958fe2e4736a301cbf225428c91 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64f22461b2f44ba7b4a4a8a7429922d3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58855b6ce834d0396596ea54e6271d5 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c054084c9a74823bd1a2090474da15b : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/efa009cc346e4735b29c6e2c344d875a : SUCCESS in 7m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a99170720cfb4cdf8759906e986f69a5 : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/591393fd4cec4dbd90071b56ad5ad1be : SUCCESS in 7m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/545472bd6b744de59423785102cec89e : SUCCESS in 1h 15m 38s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a24a23488cdf467fbaf9df463c409f48 : SUCCESS in 1h 14m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9050aec2baec4df386945a3eacc6229f : FAILURE in 18m 57s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/cfc3e753d2ee4e7fbc717fbbd578161c : SUCCESS in 1h 18m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9422ccc7d6f4e91a8bda6cf307373a8 : SUCCESS in 43m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/24263cb08cfa4aa187f9b680f3b4f25f : SUCCESS in 44m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/253cdf4c9dd5435cb1ba3ad452540670 : FAILURE in 1h 01m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fd55d3f646f3461a8cbfb331b26fd003 : FAILURE in 44m 56s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/af8b0fb9bffa4284be52eeb949d6e12e : POST_FAILURE in 1h 36m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0222d9da1e98401293370b15c5f6a21b : FAILURE in 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3e6281e84b251ffe5904955d7c9d7b88fc407f43","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-19 20:02:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6aa671ec9e2cae994e8b0eec1551f844a895295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 23:26:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a8e76a7f07c4e82b95c585169416717 : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a41222f5fec40eeabb818faa3833840 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074efbf9ab484bac9739f9b7a57ff728 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1033ea886bd451fbdd3759a0988fc72 : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08c340e0182d4e828b9a4dc7d3f4776d : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f86ca6a0370458bb3805f4a14e44cc5 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e64f6027d43420fb447061975a248d2 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdfe3a46568a46d9afd1e12d33abdc27 : SUCCESS in 5m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/91ead20956d44a64b584991f6994f0f3 : SUCCESS in 1h 21m 37s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/87216dbe471d48de838298eb23baab5f : SUCCESS in 1h 35m 45s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0181552932d4e5ab93886901131529f : SUCCESS in 1h 26m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/50fd55809f4e494ba6c0b4bf713c2258 : SUCCESS in 1h 27m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ca4837261a3e4ad2836d05fcde7243ee : SUCCESS in 5m 39s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d15c96793430457aa9023d7f3d224bd0 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e455c80bcce248b9987d91d26f82d88e : SUCCESS in 5m 17s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b04382b3eae49bb9dc212378344b1dd : SUCCESS in 1h 00m 07s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/02ed7e875c6547eca38b4a58a8745d67 : SUCCESS in 54m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb497a9e97164d9e84f72922ef6eda0d : SUCCESS in 46m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/00f242f7c31c4330af64a652bfb760fc : SUCCESS in 42m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a9a22f22e60405aa08d59a86b86b46c : SUCCESS in 1h 39m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4931084805034657a20a06483c17ba6a : SUCCESS in 1h 48m 22s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/89aa412b90f1451a811f44939adeaba3 : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/020cbd4795f94b46b0217c7ba5ab9bb0 : TIMED_OUT in 2h 36m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dd6fc603d244a28879bfd53b385f364 : FAILURE in 29m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8bec0ff29c0a40378788f9d0e926d84f : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0da5eff0dc5c4f7ca6b4e52ed57e316a : SUCCESS in 1h 59m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65c6a3b1e0d04eac9689796463f17966 : SUCCESS in 1h 16m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e58a85b8317d463d96dc7bfe89289a83 : SUCCESS in 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bebc24ed85d95062d0bf2efbc3bd9c9e43a7212f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 23:26:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1ce8dce51b7d9b756b2cf13ff49d39485e5fd14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:32:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23fca3598ad349f697a88c87dd3d90a5 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768212a7a84a46ce942dd46e5a7b78f8 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d61250be120498a953827f149575bae : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2ba3bee3e50f4497a71ddc621cc0b3fd : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5e872c9faed4e209ff43340ed55828e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c003b01bec4549a9513cd67960345c : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dcbad8ab20645628c6691572ac108d3 : SUCCESS in 5m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d337a91cc004e0ca90e46ddb647e112 : SUCCESS in 5m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b10f6d871b8649ffbd1516eb25b03183 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9708cc732a574618bc45ab06ff4e1194 : SUCCESS in 6m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02bb4537759745879370676696f1718c : SUCCESS in 1h 09m 05s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b49e464af8fb4dbeb18c999f6f4bb2be : SUCCESS in 1h 04m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e907822af1364372b3c1520799c546bc : SUCCESS in 1h 22m 04s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f2e9abbcbe6249e9a6e4a33e4c293aec : SUCCESS in 1h 21m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03985c6129b64190aedc1ac990c0ecc6 : SUCCESS in 48m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0924fbb228ae49d79f8d8901fecd11e9 : SUCCESS in 45m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e29033b3951f444c87dd4cbd115f2aad : SUCCESS in 1h 25m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6a91ff8e2f8448c7a8c0ebb1d5e41d70 : SUCCESS in 1h 37m 02s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/60b80ee5e7c34800a22a225ab55deac1 : FAILURE in 1h 04m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/624b6bf524b7434facb63f415e3f76c4 : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"800b4331877ebfc074c7082114bfa9b6ec859397","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-20 02:34:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8fe3bf749d06c0b30368ef7112d11e069f4f7ae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:06:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ab9fa91d9014726b8dfbe27f9bdf706 : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d22b447287548379689e9fcb35e2f8f : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91fa81e638e4f3a80759e84d96b4783 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79adbd7c9b4443b990d67e76e94f8e3a : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7204f2945b5740639b79036f0cacf1d9 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abf47bcb8b5143d483329c81bdb3d7ab : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f553dc0845245ec92892387aab3d3cf : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c5205836e9b4d53a30de6d544dbded6 : SUCCESS in 4m 40s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/07346e44841444e8b4732617f0cdae81 : SUCCESS in 1h 19m 55s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/dc4e77188000415eb3fefe85592dc4a0 : SUCCESS in 1h 19m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1da1c5c9ad74a078128f56039acb282 : SUCCESS in 1h 40m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8141a196290342c19ea6c5625dc5cd41 : SUCCESS in 54m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5b09888139c7441c90525befb7e79fd7 : SUCCESS in 6m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cfa698c5136347559c52f82aefa3fd06 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/953c7fd8b3e348088bc0665930f97468 : SUCCESS in 5m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26beef6026e74a9587566e02771039a3 : SUCCESS in 1h 07m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8dc91daf9a2446f38df40d1efa0f0b85 : SUCCESS in 1h 07m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1aee533c887400a8cc59299f27c08ba : SUCCESS in 46m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/57a8690d2e46484ea0ac6933c1836326 : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b006f71a0f4490f992269f1d7c3b699 : SUCCESS in 1h 41m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5056c546d9244e67b50e20c73c01305b : SUCCESS in 1h 57m 08s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/797f4639700d4832a33ee249eefd9a39 : SUCCESS in 1h 35m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3d8121714bfc4cd18cc7b219147f9ae4 : SUCCESS in 2h 18m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a498b5bffbc4481be0c9b7202dca223 : FAILURE in 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d89124f2d1574992bc0980a2cb167ef8 : SUCCESS in 2h 02m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ad353362fcd4e2983a091d4cf954e16 : SUCCESS in 1h 53m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c79a288acf04e4db897950da2ea7287 : SUCCESS in 1h 18m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44c6151af76e4193924111f560738d6e : SUCCESS in 1h 04m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4da4dccec3dbd1540753cad5ba89885275c4c18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:07:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9c19367115a9697b78aca7c66884c68ce8e0e84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b823b631fd825e5d119743a803ff04655f9aad77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:05:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da07e9d0fefc49ba90c1cda5b5ac2e60 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd65d7d6a6ad4f11825ab195b48ed358 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b683e009da344f4d8cb73eeecb2db679 : SUCCESS in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eb08c486146244d2b4d8de695fc49bd3 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f26b324d4de34d55bc4a455e1286d2c1 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11ea4ded2c541e38a899a2ea00ebae3 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d797606bda445c8ff753ccf502b604 : SUCCESS in 6m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5886e16e6984baeb6fa76fd8dea34b8 : SUCCESS in 5m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/619b2a2cb5f94bbcba7b33283b056e65 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69adef17e6074dc99486843bf462eca5 : SUCCESS in 7m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb1cc23727a0415ca6e7c280eb80ec2a : SUCCESS in 1h 07m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a8f162e922a54903a8bc238a6d6c6e1d : SUCCESS in 1h 07m 14s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3dec5bb472b049408ddfe200c57d3519 : SUCCESS in 1h 21m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/cbc0fb86cd3f4308b35ea7e04f66a82f : SUCCESS in 1h 20m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43cc280c99084d03933e64cc3c138fee : SUCCESS in 42m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/136176f6213d4f91af236926327acd96 : SUCCESS in 45m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36d0edc449884dc083dee1805d636c3d : SUCCESS in 1h 46m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/067d9af529ca430f925550a1873d8c47 : SUCCESS in 1h 40m 54s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e2c54898ee614689b50a429c4000bfd9 : SUCCESS in 1h 39m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de90b9d4715240b387f9698d4fbbcadb : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e3cbf0a459d2a1ab220ed0d59ffc5864c2219f9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:07:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653f92120c1b4d64ac4d903797c8c759 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f462d6aa6d924303883f858c011bb97c : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cb64148d599e03e1b236a089070f66e1ad1a0f3","revisions":{"8cb64148d599e03e1b236a089070f66e1ad1a0f3":{"kind":"REWORK","_number":1,"created":"2019-09-18 15:13:51.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/30/682930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/682930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/682930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/682930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/682930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/682930/1"}}},"commit":{"parents":[{"commit":"f556b84cecf04a36d4a7369c629deb8ed3e774d9","subject":"Merge \"Fix catching driver exceptions\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f556b84cecf04a36d4a7369c629deb8ed3e774d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-05 12:52:02.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-18 15:13:51.000000000","tz":0},"subject":"Fix cleanup of expired load balancer entries","message":"Fix cleanup of expired load balancer entries\n\nFix a bad query filter, deleted load balancers have a DELETED\nprovisioning_status (instead of operating_status).\n\nAdd a functional test that checks that deleted and expired load\nbalancers are correctly selected by get_all_deleted_expiring function.\n\nFix a similar test for amphorae that was buggy and not enabled (because\nof missing test_ prefix).\n\nChange-Id: I0ce2eabfd4dd41210312ea3b7f6274c9a6d50e44\nStory: 2006496\nTask: 36458\n(cherry picked from commit a4da3ef22001e1229541a3bdfe232f6967987cb9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cb64148d599e03e1b236a089070f66e1ad1a0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cb64148d599e03e1b236a089070f66e1ad1a0f3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
