)]}'
{"id":"openstack%2Foctavia~682950","triplet_id":"openstack%2Foctavia~stable%2Frocky~I932ad625595333e97b0ead074ce64a7341af338d","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I932ad625595333e97b0ead074ce64a7341af338d","subject":"Work around strptime threading issue","status":"MERGED","created":"2019-09-18 17:16:45.000000000","updated":"2019-09-19 16:24:35.000000000","submitted":"2019-09-19 16:22:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682950-1568910154575-7126b0df","meta_rev_id":"cbea781b16a90e5525bf891d3898108c4c0e7e56","_number":682950,"virtual_id_number":682950,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-19 16:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-19 05:43:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-19 13:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-09-19 13:55:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 05:43:03.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-19 13:55:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-19 16:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b824ad1908ed204ffc1c7a7c026b3bf02918d79e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 17:16:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"b54cab04adc05f0c9a7f1dc7bdeabfdfdec6197e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 17:16:51.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit d6c1f8ec846c6065573d36f5db261eee0483dc4b","accounts_in_message":[],"_revision_number":1},{"id":"86413bc008b5f4d6d75d258a427d8a01a19b78ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:13:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/499b3a530a154eca8d45f1a969a24773 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818839c7f28a4bcaa7f377db7e87af89 : SUCCESS in 9m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91b1fc402e21426d9e7161e897c7f221 : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f8f503d249b54104bbd2d5367afa1f16 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fbc3b9abc054372ba2b7ba1eea0f969 : SUCCESS in 7m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bc79bcdc511142928ce6a8995a43b0a5 : SUCCESS in 1h 16m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/61bd0c7d74954458b8b95f52ae22adda : SUCCESS in 1h 17m 00s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1511e6bed514e1797680897d3ac2c9b : FAILURE in 1h 19m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5661681e0bd84cd08b204a1fb372e03b : SUCCESS in 1h 35m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f696df02ba74e878f820a0baff4171e : SUCCESS in 4m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/913576908b6645d4a3178acdc8df0fd2 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f231ed086e6c43bf95d1c540834f6939 : SUCCESS in 5m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1f074cfb25c45c3a0e0adea509c70f8 : SUCCESS in 1h 13m 13s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5252591139c84a3c9eaf4dd45155eca7 : SUCCESS in 1h 09m 18s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e404c1e0149f412793e342e2c0566a41 : FAILURE in 1h 28m 38s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cc64f4678cf247aea163c1abece4d7b6 : FAILURE in 1h 34m 40s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/5dcf4ba68bd74962bd230fa9786a0313 : SUCCESS in 1h 05m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a088dbe419042ac84f3dce57f4fbae1 : SUCCESS in 45m 13s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/996c808b9f544410a36f130c5e7f2167 : SUCCESS in 45m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f78928ab4bdb4c59abe7252ec67ee4ab : SUCCESS in 1h 35m 21s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2837b49bbb544bffbb6841568ec5de9d : SUCCESS in 1h 43m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e5d9bd9bbc8423f8ba4df1585f87093 : SUCCESS in 1h 16m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b49e4803e09e4b8f8e22532870ba40ff : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"789c7b187123d19932aa3cfdc4120b0ff778d8fa","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-19 05:43:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28fea91b437f41e0b2e2dc71d1d4da0de5fc55b9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 13:55:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"aae2e397eeee5f6e2ee39650a276045be3673c8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:55:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17524aca9db28c86b9410c7eaa823e2a4f493e34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 16:22:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73e3a4eb2969482aaed2896b101e7e83 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf1f5e8d0e747878bc48993c01af95f : SUCCESS in 9m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b819ab0ddac5427586d20d1d1aa49b4c : SUCCESS in 9m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/feda18cd6abf4f1aa4e4bb219dabe862 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6452a920ee5a43cabdb0bf8c53086d8e : SUCCESS in 9m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a0b9b54c56934545a7e2cf16c8434f77 : SUCCESS in 8m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5d1a363f3ac549238ba626366c6329c8 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/306144f0bdca46789aff33cccce0c2d5 : SUCCESS in 5m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc5ec9bb4a054a0e84da8d96abbaae3c : SUCCESS in 1h 09m 57s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/196c61eaeb3845f49e04186d37db67c8 : SUCCESS in 1h 05m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/042fb9a6ffb244e0b3c30e8a4e1b0d96 : SUCCESS in 1h 14m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ce557fa25b84c2ebe9d4c193cda5eb1 : SUCCESS in 45m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/c32f7851480d4db8b55935e8d479412d : SUCCESS in 50m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fb2c1ff3781490cb8eb344d3268b310 : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/4253b0f3c50b4514908e9f890a40c963 : SUCCESS in 1h 48m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ae179bfcdb943079b4b2cd8c99bf293 : SUCCESS in 1h 11m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01f815bc448f47ebbdb352cb1beb82c8 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f4691df1ec032f0d9b5a803cce06a7515ae7602b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 16:22:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cbea781b16a90e5525bf891d3898108c4c0e7e56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 16:24:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad972c3768264db7ba8aabf5e23e56dc : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3cc8f8fcc210d390399fab004b589bf3981942d","revisions":{"e3cc8f8fcc210d390399fab004b589bf3981942d":{"kind":"REWORK","_number":1,"created":"2019-09-18 17:16:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/682950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/682950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/682950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/682950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/682950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/682950/1"}}},"commit":{"parents":[{"commit":"af41c75033e95ec9fbd3b008921f2c7648a7a471","subject":"Merge \"Do not run non-voting jobs in gate\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af41c75033e95ec9fbd3b008921f2c7648a7a471"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-10 02:41:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-18 17:16:45.000000000","tz":0},"subject":"Work around strptime threading issue","message":"Work around strptime threading issue\n\nThere is an open bug[1] in python strptime when used in multi-threaded\napplications. We have seen this occur in the Octavia test jobs[2].\nThis patch works around the bug by loading strptime early.\n\n[1] https://bugs.python.org/issue7980\n[2] https://logs.opendev.org/37/673337/12/check/ \\\n    octavia-v2-act-stdby-iptables-dsvm-py2-scenario/440c965/controller/logs \\\n    /screen-o-cw.txt.gz?level\u003dERROR#_Aug_09_23_54_29_426364\n\nChange-Id: I932ad625595333e97b0ead074ce64a7341af338d\n(cherry picked from commit 0720397d5c47e2af33dbe9fdf8f25b95ce620106)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3cc8f8fcc210d390399fab004b589bf3981942d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3cc8f8fcc210d390399fab004b589bf3981942d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
