)]}'
{"id":"openstack%2Foctavia~683537","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I78d75afd872fe27e9e4227ea8220f74d83595aac","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"I78d75afd872fe27e9e4227ea8220f74d83595aac","subject":"Flush the eth1 default route inside the netns","status":"ABANDONED","created":"2019-09-20 16:36:21.000000000","updated":"2020-03-19 15:49:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"162df15892d73b54f255651d501a01c3d6b82612","_number":683537,"virtual_id_number":683537,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-02 02:35:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-02 00:54:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 13:19:55.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-11-27 18:42:49.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-01 10:05:23.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-02 02:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"797bccbac74b72349ffac2b64e97aa82d7e96805","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 16:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c165e654c2a449f72254efcb19d7f1031178f81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 16:36:40.000000000","message":"Patch Set 1: Workflow-1\n\nThis is an experiment at this point.","accounts_in_message":[],"_revision_number":1},{"id":"40dc1726473322b751490e2653c9fb3ee41fb7e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:54:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1180913759481099948ecfa2eab1aa : SUCCESS in 12m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28b9070905c94420bdaa0738365b098f : SUCCESS in 7m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8efb41f14c9b4af79540a0c1f9d0823c : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852af83b87b24a119147d9fe73c15598 : SUCCESS in 6m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3d82e0f2694a423a8243154b48db3519 : SUCCESS in 1h 27m 39s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/de858b80adab4ebb8b62bd874e45e11d : SUCCESS in 1h 27m 32s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e247e95887304f79a8468c642cd74f6d : SUCCESS in 1h 07m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a9402d411933454c86fa10c123b0f685 : SUCCESS in 1h 23m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8acbe7a10084090a5cbbf78659b4c44 : FAILURE in 9m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d9d7dc4ac34d42f3abac3e59706603f7 : FAILURE in 7m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/53fce0e4ddf64921aed242a4c59ff99d : FAILURE in 6m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a5adf95f062480c88395ff1a3dd2e61 : SUCCESS in 1h 03m 18s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5361f51998b44f76879c9b476b07f83a : SUCCESS in 1h 06m 50s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/9fb107e534d44ec6b7f1371f762d912d : SUCCESS in 1h 36m 58s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/3663913b434c4283a15650300cb1b2cc : SUCCESS in 1h 17m 31s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/7ff68b7536f14b1aba1fbc954c4a2a65 : SUCCESS in 1h 15m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/831a6ea314cf43d8a404275e63c94fad : SUCCESS in 46m 46s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/28728accb7634b23ae8600c5b982ac84 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a93f62e50ef4a68b3bc2de5eb447f33 : SUCCESS in 1h 40m 46s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/51d25d9c22524c4abdf4f73478b263f4 : SUCCESS in 1h 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"885c3d903855e51a8d8f1069156d30b4f173c829","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-28 21:32:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"efec424ed67b4b6be70905104234601a3293568e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 23:09:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43187c120b2049b093e2797b0b70fe0b : SUCCESS in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/813d92fec8a346d39ec0be7394552a09 : SUCCESS in 6m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f75a496f388645c1b14c9fe2555aca0f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba05fbce0144e5e8c3dad33af4b2b0e : SUCCESS in 6m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/db467977f81041f7a2f5c8a7f917bd52 : SUCCESS in 1h 13m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/31323a3d8a6a4037a0e75a673834db62 : SUCCESS in 1h 16m 02s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0f81d4bd5ea4653a7447c4cffacbb55 : SUCCESS in 1h 20m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4e5d8ebe19654639ae60cbde174ae024 : SUCCESS in 1h 31m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/061080d755884089a077f3979c755ca2 : FAILURE in 8m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b8519e25dbc3434daa5ccef343fb7ce6 : FAILURE in 5m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b25b8b5ff6cf42c08b1eae350002676e : FAILURE in 7m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90977e18c44a498685cf57bf8cc7249d : FAILURE in 41m 11s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5585a590f8304b5a95513451902c87a5 : SUCCESS in 59m 23s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5a68dd4e23dc4646a389b083bc6869ed : SUCCESS in 1h 06m 37s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/45b50551a62e43038ad75c9f2a1ad60d : FAILURE in 1h 03m 24s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/613c8f0b3a714d49b8f54e28989c62b7 : SUCCESS in 1h 03m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99a3f463678043cdb1a591732eb02043 : SUCCESS in 38m 18s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/4432f773c2cc4a1bafa6377ec85e8b7a : SUCCESS in 43m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56b30c594afd452f86d2b6609ee559a1 : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/fa631fd1ef4c4f0ab403263663db6615 : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"72c368f259bb71c26ccbc975c06afaa26b22d0c3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-29 16:49:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbbaf54536403483331cc0bf963fd0d0207cd13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 19:56:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69908062e2c4b0eadc2e9f58e40cd5f : FAILURE in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3199bba07b1a4e5db9aef0617bfb5563 : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a25189969e44428e91543d06513a551a : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d3e5be0af840b8b4046f0ccf07ce67 : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/821e23b6181a4599bf9a6403bf980c84 : SUCCESS in 1h 14m 18s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/df83730fe8704047b934ec4c5320cfc1 : SUCCESS in 1h 38m 02s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/321ac534c95c42c985744b9bc2bae474 : SUCCESS in 1h 33m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9bf2daebe66743bf81a711800626dd7b : SUCCESS in 58m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/612cfae6c2d040ddb8d83d31f038acbc : SUCCESS in 9m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e4e4e6ac07fc4219a38f422e88851099 : SUCCESS in 6m 15s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0356ae0cdcb84c9f977bb58d00736b50 : SUCCESS in 6m 26s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/614f52e40fb048d2aa223fcf77886a8d : SUCCESS in 48m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/28e7100681d047b49c85c8a9a376e6b7 : SUCCESS in 53m 52s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f7f1c1905dfb4c98b1d3e8aa85b99468 : SUCCESS in 1h 11m 52s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d1a67111be204701af1c67cab3d68a21 : SUCCESS in 1h 02m 23s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ff5f955f899f41ae89232a5309fedebf : SUCCESS in 1h 07m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0061dfb96f846558e395d0c957defcf : SUCCESS in 44m 23s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/efecf3f582b64ccb88cec66886bfcb69 : SUCCESS in 42m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/593e4080467f471ab9c87b53a63e606c : SUCCESS in 1h 18m 43s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/58b4b4f327ca4edd8a6e5a5904799f2e : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":2},{"id":"30eea60610ce83ecc04df9c9abb79738044876a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-06 21:39:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a03fd8deb7c57057f3f2e9355d65d0862ffdc287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 23:14:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d57b8459dbf462abdd3acccb14e02ab : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/423f1df811ed4967bd5f03c0205e5938 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e660f5b64ad47948d4a924fb5ff6f5a : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c6772150084b2994ba55c2643890df : SUCCESS in 6m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/56a13fdc057f44079db23386e8671089 : SUCCESS in 1h 21m 54s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/69acaa9036824d25b4aae23a8dedb5cb : SUCCESS in 1h 13m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6660eb7ebcbf4ec89d4597f974703d13 : SUCCESS in 1h 13m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6274c2652b8e42feaaa900eb7e3b7564 : SUCCESS in 53m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/878e333ee1dd48f7910f2a10561d9dfa : SUCCESS in 8m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/35ccdbcc39204f609b023bfd16284f1f : SUCCESS in 5m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ac38860367ba4831937c58ccb46d9b15 : SUCCESS in 5m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6dd11334b5284e2f80a54a2f326ef831 : SUCCESS in 52m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9c8e34f995f841a2bee6c0f0425362cf : SUCCESS in 43m 42s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/87fd1adbc3d04dcf9fca1bae2ef0312c : FAILURE in 38m 59s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/6fa2f51b161c4691a1238a93175eebe5 : SUCCESS in 1h 11m 10s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/6810b36590044f77bb16d4a8dabf91ba : SUCCESS in 1h 05m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d670d5eb8fc3437ca023661adcd917fd : SUCCESS in 52m 06s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/92b968c5ff554b4faf769506a2ae9166 : SUCCESS in 43m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69fbb06785194063804c574f1e95c6d0 : SUCCESS in 1h 22m 47s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/95a748da895b4616801ed2eda33ef650 : SUCCESS in 1h 28m 49s","accounts_in_message":[],"_revision_number":3},{"id":"406e512f69540ca51922a9143c3a9bc9b63b9b09","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-06 23:40:49.000000000","message":"Patch Set 3:\n\nrecheck\nDIB issue","accounts_in_message":[],"_revision_number":3},{"id":"e8571137d5453ff88afd1d40b2d7ae25197dfcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 01:51:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab8260136a5d4b8a8a42d02ea6d0ce87 : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c757c349595c4edd9f5f8bb1342707a7 : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4930d268f0cd4ddfae69438cab626d39 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c7dbb51ddd4e6993de2b43ecdd73fa : SUCCESS in 8m 02s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2726fc51df254193a7b60bbf86d53384 : SUCCESS in 1h 45m 06s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/97f4ae4f8d8e4f43a65bd1e9c15a31a1 : SUCCESS in 1h 19m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b6941cecb0c41d698c1152ee5af268d : SUCCESS in 1h 23m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/73bf80bae1ba406094e0e7ffca7cc4cf : FAILURE in 2h 08m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57398fa01484f15b4a12c171266e433 : SUCCESS in 9m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a38b9959695041b984b5a29677b4f771 : SUCCESS in 9m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/47044a9e441949e3bfa86dbb96774a3a : SUCCESS in 5m 44s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71b31e1429194592bc4699d06f0b941d : SUCCESS in 54m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3e40f351f1ed45bbb5cf2565d5875ed7 : SUCCESS in 1h 49m 45s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/1c9d7dc7140f435ab7bf9dd4fa32f0c8 : SUCCESS in 1h 23m 10s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4cca0028435f4f6f9fb083bd99bc66b9 : SUCCESS in 1h 06m 41s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b2e819f77a924222a1457e31be04324d : FAILURE in 1h 11m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34cb5357f9e74298ada3250f978988df : SUCCESS in 40m 39s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/29157960ba6b4b6893e91d5f2e851ecb : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25b2bccb47df48c589995547163aa1e4 : SUCCESS in 1h 59m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/e0157c7d0af749cea2fa445316a6c499 : SUCCESS in 2h 01m 03s","accounts_in_message":[],"_revision_number":3},{"id":"1a6c931851518b9873e0eb788f67e41097097f50","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-26 09:28:59.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck\n\nGive CI another pass for greater confidence in this patch.","accounts_in_message":[],"_revision_number":3},{"id":"bcbad0da131e87e4de75bab63f7a228ca24b6553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 11:20:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26813ff00e024a62a14e714b684593c1 : SUCCESS in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc17137258b34aaaa7cb3e96f044ae75 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8593deb0736a4fc6bacf66ec47fb0862 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7191429e93d0488682d42359181a2280 : SUCCESS in 5m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ce2fd84ae3374bb4ab45af4b6f1cec2f : SUCCESS in 1h 16m 50s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bdec6320cc3b4eb096004be579329b84 : SUCCESS in 1h 22m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d2368ac3e8d479da8e087c125ec5ae3 : SUCCESS in 1h 38m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7fa8095a841c46c08de1d35dd923f3c9 : SUCCESS in 1h 20m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0381bd52df84680b07b2a9f726ab441 : SUCCESS in 8m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/280a3c6562e74bf1993a25f33461a630 : SUCCESS in 6m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a9d1f994407449cc82e488cc00add01c : SUCCESS in 6m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa96b643200a4f5a8019b59058128a68 : SUCCESS in 1h 07m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dfabfdb1940e467aaa16edbb2ec18e12 : SUCCESS in 56m 57s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a41444798f104997b9303cd914f3f4de : SUCCESS in 1h 10m 14s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/59ae03b9d0da4e30bbbb4c34260fa10b : SUCCESS in 1h 11m 39s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/5d5f81142030473a8afccbcb3a76e801 : SUCCESS in 1h 12m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/baf0874814514222beb864e1adcd6c36 : SUCCESS in 43m 46s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/fa547c0da8eb49d19b8b80a7222bf844 : SUCCESS in 50m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a930e87b5e0a4e789161f3f18781ac00 : SUCCESS in 1h 43m 57s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5ce6ba9a526646a3b6ac35ba00cd2131 : FAILURE in 1h 16m 40s","accounts_in_message":[],"_revision_number":3},{"id":"d748daf03b8e844019c54f9c2549a80fc89a6d9c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 13:19:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"095f92642fdef78b5d555691e031e7f7eb760bd1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 12:17:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"75761edaf5f9d3f23d711a3cffad97330a53a35b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:15:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f8295f3d4b4e19a6ee86dfa1037308 : SUCCESS in 9m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e14e0fdbe1a74c6b9c68c93796143ce0 : SUCCESS in 7m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3b352bfbd37247aa9120e764584cee55 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81140404e7b4cd2bd444f63bcead0f6 : SUCCESS in 6m 37s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ed84cd0d981e4f9eb8e2a41fdcc18b8d : SUCCESS in 1h 21m 25s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/47ea1b05d0e34591907b82408597a125 : SUCCESS in 1h 31m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec682676ecee4fddb429a17870b1ca70 : SUCCESS in 1h 37m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3c189af6f89744a9aabdb140a233f818 : SUCCESS in 1h 16m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aea003fb925b4534b15a0a40871c7372 : SUCCESS in 10m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c28b280c89574a6989ff921e1a2f63dc : SUCCESS in 6m 40s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f311052de5f84447b081c317c013f36c : SUCCESS in 8m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f8996e01cff46b99d441e8be1cf7be7 : SUCCESS in 1h 17m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a9af15d7b921427dbb8121a8ba494fb3 : SUCCESS in 53m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/edaabd60f2e34f8e8567aa4e5fdcd8aa : SUCCESS in 1h 08m 58s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b88424553d5a4e079afb04a7326a2d0a : FAILURE in 50m 45s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/2a69c0625bd94b85ba1cb6ef71e87fe5 : SUCCESS in 1h 11m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/213409718e4440d18777e2dc4040cbf4 : SUCCESS in 46m 26s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/f813475c6fa5443a88455f9769276717 : SUCCESS in 51m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8566141a0b4347b698ef3545bbe94bb6 : SUCCESS in 1h 46m 22s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a4f804f991ff48cea8e70056e04864bf : FAILURE in 1h 36m 26s","accounts_in_message":[],"_revision_number":3},{"id":"64bbaf17d20c5af8005cd4223d6df27fa9df48a2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 15:49:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9f0a9cf39d5684c4ca6d96029a6965ed865e09e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 17:39:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"23d93ba5ecf54937c1f050fdadd6574b3f6e2ee8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 18:42:49.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"01fc677730cb96f2d35054e290718abb8d1c6a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:58:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97e0cab0ef64534ab1aac930b54b784 : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/731a23ab36b94b0db30f53cb17a7c5c7 : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a3a0aed2512449e9b6abebb3450e9161 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa350bacaed42ad86628e6e12475b0d : SUCCESS in 5m 58s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/73353d36d93b4e9b91dc304dec6bb47e : SUCCESS in 1h 23m 32s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/52668873b4b1468eae5d46754e0bb262 : SUCCESS in 1h 24m 08s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edc1f3cf0ef74026aa9c35fd686ecf71 : SUCCESS in 1h 20m 54s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/15c75273a05f458e97ac961d581dab32 : SUCCESS in 1h 35m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/467e17fd51964f61a25d8ffe16d2de0b : SUCCESS in 7m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8ee1e2da93654ec4ad5e6df2574d9381 : SUCCESS in 7m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/bb93b74d32804cff8998c8e663aaea5a : SUCCESS in 7m 03s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2795153870f249a9acfc149139000eb9 : SUCCESS in 1h 00m 56s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a644cb8abeff448fa8dfd0d8739eaf46 : SUCCESS in 58m 45s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/edab09b10e0b4f5f8dff6df4cd7a0af5 : SUCCESS in 1h 15m 12s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/bf7ef92f5ba044878e320950aeb69f5f : SUCCESS in 1h 18m 48s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8754f9fa32f04903ab3ff0b79e04e20f : SUCCESS in 1h 00m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/021a47da9f874d948fb9137f50901dd8 : SUCCESS in 45m 24s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a8cf184785954ac89899b5a2433eac54 : SUCCESS in 47m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ec77255d0fa4946a8a8d5b3429f65b2 : SUCCESS in 1h 35m 31s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/d08f06a361a5428a948b85e78c5f4d00 : SUCCESS in 1h 23m 20s","accounts_in_message":[],"_revision_number":4},{"id":"f0931c1ac20df35bcdfbd1108f9cded73c5cc73c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:58:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a698c52f55e482c41ef7567a874ee92ff105bf47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 21:39:09.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6134b355ce84bf1b73c90b12a063464 : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6143e9c9b5714118a03b20d770d65c79 : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f0ab8d688c6e4d0c9252496b05bff976 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cc5fe7d2b54f74b068198e0705bef7 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2d843c4abc4600bb24563f808368a0 : SUCCESS in 9m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16ae2bcf8ad04607ae32790fcff8e7cf : SUCCESS in 7m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/138acf3d3ad4487fa8273a7d3ad92ec7 : SUCCESS in 6m 40s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ba1e6c7803994dd296680cf6e87ea88e : SUCCESS in 1h 19m 46s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8e8214dc256e4ddfa37f0a6ec71bc8a3 : SUCCESS in 1h 14m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7471540ec4f4a36ab50f914170ae2fe : FAILURE in 50m 26s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d341fff199a04caabd9fef9c3630bcbf : SUCCESS in 55m 49s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/12a4cf3ca0d84e568e47a236e4810605 : SUCCESS in 1h 03m 15s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9bd141f73bae4e7f8efa993e7930196c : SUCCESS in 1h 10m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f1d66f3c01d0455fb2ef786b5c4dea73 : SUCCESS in 42m 35s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/59b2258a46d849d7a0e1fb8d90da317b : SUCCESS in 45m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/410966f085cd43408a04d1eb54fb5653 : SUCCESS in 1h 21m 50s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2af8dc10c3524ab69ccb3ba8d3e064de : FAILURE in 1h 34m 53s","accounts_in_message":[],"_revision_number":4},{"id":"60dce82a2361c657dfbd36704dec6b39b57004be","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-28 05:15:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d0c69ca5a67adfc29b0cf830cf559b2104220e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 06:42:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dcfc3b8fad4a8490e8042a47cd0646 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d28e837d27c6447db7bf4d36368ab824 : SUCCESS in 4m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/52db57b199e143a4979a5fa7dd9c60ef : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7348cce936634e53b83c938d7b6d17db : SUCCESS in 5m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d68dfec633894ab4a780a02c13cc2e27 : SUCCESS in 1h 15m 19s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1fb0125910354b9bb33572e0a1816b64 : SUCCESS in 1h 17m 20s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbc8424282094d4b99f7d2658af7ab6c : SUCCESS in 1h 14m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/805efa3cc63a475287d75ca311ced183 : FAILURE in 1h 14m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f38071fe5b3485a8473cdb36a4f9dc5 : SUCCESS in 8m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7de62568fdc649018583786413625ab2 : SUCCESS in 6m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/595c751e44204cb5a7855675358fd50e : SUCCESS in 6m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6fdf0c94a02a4599a5a41adf46527dc3 : FAILURE in 51m 37s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/28a428200591491f924206e2e98d8094 : FAILURE in 49m 43s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/8e0ef0707c0d41259b7631555978a451 : FAILURE in 1h 01m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/51156aec281348cb82685ed5c96e2c0b : SUCCESS in 1h 07m 22s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/106fcc7cf2fa435293172df5e1958d76 : FAILURE in 56m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70dab9fb9612411eac37f51888ce0aaf : SUCCESS in 50m 53s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/130007155def4dc2b780888807eaee5b : SUCCESS in 43m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa0aa3f5c03f4e9a9f597dffa7ebb9c5 : FAILURE in 1h 23m 27s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/bfc970c0d8e343e9a8288e5544291d3f : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":4},{"id":"7a4191011cb09981033a5d33ee1baebc1760ba80","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-01 10:05:23.000000000","message":"Patch Set 4: Code-Review-1\n\nStill seeing RTNETLINK messages.\n\nNov 27 21:11:19.781982 ubuntu-xenial-rax-dfw-0013103201 octavia-worker[15329]: ERROR octavia.amphorae.drivers.haproxy.exceptions [None req-19df0705-57f3-46b7-bf0e-de307b1d6493 None None] Amphora agent returned unexpected result code 500 with response {\u0027details\u0027: \u0027RTNETLINK answers: File exists\\nFailed to bring up eth1.\\n\u0027, \u0027message\u0027: \u0027Error plugging VIP\u0027}\n\n\nhttps://zuul.opendev.org/t/openstack/build/2af8dc10c3524ab69ccb3ba8d3e064de/log/controller/logs/screen-o-cw.txt.gz#5503","accounts_in_message":[],"_revision_number":4},{"id":"1fe4b8d3738c3f74c624547921270b12329ae1ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-02 00:53:59.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"6ac6605fd3e4fcd2179719a79f31db7a67ec3bb5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-02 00:54:04.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"9ec9d293383c6cf5ae3f6d87f44a56121560039e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 02:35:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf79ea0014340a1b6303af9edc306f3 : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74fe45eb046448e39f879155292f9e98 : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f1235225b6374a0fa55560d1bf271891 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ebc09f9d28469d895f369d7cca9f79 : SUCCESS in 5m 33s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3cc901dfc0ad4fc88ba07fec42b261bc : SUCCESS in 1h 15m 37s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/eed99d59bc29465f86d0a8bba7cd5034 : SUCCESS in 1h 14m 21s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbaac47c317449e5b9daec4b84aa9c20 : SUCCESS in 1h 29m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/22feabf1c5c44390bcc9f0f2c463a09a : SUCCESS in 1h 20m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdea20ae1cb24b07ac3142712e329e46 : SUCCESS in 8m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3ecfc221e7b543ed99a8b29053fe334a : SUCCESS in 5m 55s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/da1cde7a63c2499994368225eb8c39c5 : SUCCESS in 6m 09s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ccc6555025c345e9b93bd23694685a3f : SUCCESS in 58m 58s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5bbeef2fd2ef4866b2c01c1d38e87d53 : SUCCESS in 53m 22s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d7b94af5ff3740bb91a5d325b22ec3e6 : SUCCESS in 1h 07m 24s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cd41fe29ab0b4b2888b73519924262cb : FAILURE in 56m 51s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/11ddc81ad3de4c47bcf744037708ca78 : SUCCESS in 1h 09m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7e749f060504cd19fa55bb2206dfa0a : SUCCESS in 48m 37s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/7670917d81d34f1ab8768d10a295519e : SUCCESS in 42m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa5ce89056d24e46aa4ec57016a1eb46 : SUCCESS in 1h 25m 14s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2e4a8ec6b6d84c288ecf007f777b81fe : FAILURE in 1h 32m 13s","accounts_in_message":[],"_revision_number":5},{"id":"162df15892d73b54f255651d501a01c3d6b82612","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-19 15:49:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d0653bf8b3aa36d60b30c774ba483826d53a3e12","revisions":{"97792ccbc9c9bef66ba96005d528357df56c9482":{"kind":"REWORK","_number":1,"created":"2019-09-20 16:36:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/683537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/683537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/683537/1"}}},"commit":{"parents":[{"commit":"189938ee635b3c597b15e978f7a59a2e157047ea","subject":"Merge \"Work around strptime threading issue\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/189938ee635b3c597b15e978f7a59a2e157047ea"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 16:34:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 16:36:20.000000000","tz":-420},"subject":"Flush the eth1 default route inside the netns","message":"Flush the eth1 default route inside the netns\n\nOn xenial instances we are seeing error \"RTNETLINK answers: File\nexists Failed to bring up eth1.\" when bringing up the VIP interface\ninside the network namespace.\nThis patch attempts to remove any existing default routes that\nmay have previously existed before bringing up the VIP interface\nto avoid the potential for duplicate default gateways.\n\nChange-Id: I78d75afd872fe27e9e4227ea8220f74d83595aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97792ccbc9c9bef66ba96005d528357df56c9482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97792ccbc9c9bef66ba96005d528357df56c9482"}]},"branch":"refs/heads/stable/queens"},"b4a1ee7f25d062d3a76b6fd2c690cea75ad07164":{"kind":"REWORK","_number":2,"created":"2019-10-29 16:49:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/683537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/683537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/683537/2"}}},"commit":{"parents":[{"commit":"189938ee635b3c597b15e978f7a59a2e157047ea","subject":"Merge \"Work around strptime threading issue\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/189938ee635b3c597b15e978f7a59a2e157047ea"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 16:34:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 16:49:36.000000000","tz":-420},"subject":"Flush the eth1 default route inside the netns","message":"Flush the eth1 default route inside the netns\n\nOn xenial instances we are seeing error \"RTNETLINK answers: File\nexists Failed to bring up eth1.\" when bringing up the VIP interface\ninside the network namespace.\nThis patch attempts to remove any existing default routes that\nmay have previously existed before bringing up the VIP interface\nto avoid the potential for duplicate default gateways.\n\nChange-Id: I78d75afd872fe27e9e4227ea8220f74d83595aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4a1ee7f25d062d3a76b6fd2c690cea75ad07164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4a1ee7f25d062d3a76b6fd2c690cea75ad07164"}]},"branch":"refs/heads/stable/queens"},"0a4c78b0440c2551f62ddbcd2c985230743ca533":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-06 21:39:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/683537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/683537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/683537/3"}}},"commit":{"parents":[{"commit":"6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae","subject":"Fix Octavia stable/queens pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 16:34:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-06 21:39:15.000000000","tz":0},"subject":"Flush the eth1 default route inside the netns","message":"Flush the eth1 default route inside the netns\n\nOn xenial instances we are seeing error \"RTNETLINK answers: File\nexists Failed to bring up eth1.\" when bringing up the VIP interface\ninside the network namespace.\nThis patch attempts to remove any existing default routes that\nmay have previously existed before bringing up the VIP interface\nto avoid the potential for duplicate default gateways.\n\nChange-Id: I78d75afd872fe27e9e4227ea8220f74d83595aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a4c78b0440c2551f62ddbcd2c985230743ca533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a4c78b0440c2551f62ddbcd2c985230743ca533"}]},"branch":"refs/heads/stable/queens"},"3ecae097f1a17ab81fcafab3d531b9e207801ef7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-27 17:39:17.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/37/683537/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/683537/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/683537/4"}}},"commit":{"parents":[{"commit":"96e5962e05615df37041723ca173873f11d55ea2","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96e5962e05615df37041723ca173873f11d55ea2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 16:34:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-27 17:39:17.000000000","tz":0},"subject":"Flush the eth1 default route inside the netns","message":"Flush the eth1 default route inside the netns\n\nOn xenial instances we are seeing error \"RTNETLINK answers: File\nexists Failed to bring up eth1.\" when bringing up the VIP interface\ninside the network namespace.\nThis patch attempts to remove any existing default routes that\nmay have previously existed before bringing up the VIP interface\nto avoid the potential for duplicate default gateways.\n\nChange-Id: I78d75afd872fe27e9e4227ea8220f74d83595aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ecae097f1a17ab81fcafab3d531b9e207801ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ecae097f1a17ab81fcafab3d531b9e207801ef7"}]},"branch":"refs/heads/stable/queens"},"d0653bf8b3aa36d60b30c774ba483826d53a3e12":{"kind":"REWORK","_number":5,"created":"2019-12-02 00:53:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/683537/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/683537/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/683537/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/683537/5"}}},"commit":{"parents":[{"commit":"96e5962e05615df37041723ca173873f11d55ea2","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96e5962e05615df37041723ca173873f11d55ea2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 16:34:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-02 00:53:29.000000000","tz":0},"subject":"Flush the eth1 default route inside the netns","message":"Flush the eth1 default route inside the netns\n\nOn xenial instances we are seeing error \"RTNETLINK answers: File\nexists Failed to bring up eth1.\" when bringing up the VIP interface\ninside the network namespace.\nThis patch attempts to remove any existing default routes that\nmay have previously existed before bringing up the VIP interface\nto avoid the potential for duplicate default gateways.\n\nChange-Id: I78d75afd872fe27e9e4227ea8220f74d83595aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0653bf8b3aa36d60b30c774ba483826d53a3e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0653bf8b3aa36d60b30c774ba483826d53a3e12"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
