)]}'
{"id":"openstack%2Foctavia~683968","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I6ad0349dba62b1141be07bfb0e40171e9f7a91b9","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"I6ad0349dba62b1141be07bfb0e40171e9f7a91b9","subject":"Improve the error message for bad pkcs12 bundles","status":"MERGED","created":"2019-09-23 14:05:26.000000000","updated":"2019-10-18 03:07:34.000000000","submitted":"2019-10-18 03:06:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683968-1571367977892-f62c7b6a","meta_rev_id":"ceb4ba7b0f2481e0ec0bd42f18673350c4e645bc","_number":683968,"virtual_id_number":683968,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 03:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-18 01:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-16 19:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-10-18 01:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 19:25:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-18 01:11:53.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-18 03:06:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a60b962bd6dafb3e01c873738e49339f544b30ca","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-23 14:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fde1cfe906dec57117c2ef3a46599effcb23ce66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:23:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd7f9b5c95a4125abb933bd8c4a9739 : SUCCESS in 9m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c983675ecec43399eee354a1837a289 : SUCCESS in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8e1d5da3c58f4915a1d61157fc2a7301 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0e8630c62c4e31bc577520a678a23c : SUCCESS in 7m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/abc1e55042a945c9bc31b65b9bbfa2f2 : SUCCESS in 1h 25m 37s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6ea424e00025483681801d76ae01b5b7 : SUCCESS in 1h 21m 24s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4cf6c624a334ffe949197ce5af3e86a : SUCCESS in 1h 37m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/34ea6d98905a422aa460cf38faf5c4a9 : SUCCESS in 1h 36m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bbd20880fa8497491d8395facf2d18a : SUCCESS in 11m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/27262724762d4bf4a778a0a93dc155e2 : SUCCESS in 7m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7cbed4ab59084ae991d3e54b5b62daf9 : SUCCESS in 7m 49s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d0abf21e293415d9368b7fc4752a701 : SUCCESS in 1h 08m 52s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/34d470bfcdf540f3814127c0a672ef1f : SUCCESS in 1h 05m 14s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/40bb17339ac549ab9d6c1a3719b2672c : SUCCESS in 1h 17m 56s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d40dac83fba2434c90433b9ad5627c1a : SUCCESS in 1h 22m 16s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/04cbecaa0b3a4e3cb013fe4dc01b72cd : SUCCESS in 1h 11m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b27cb6e66d8546dc84bc524fe376b7dc : SUCCESS in 48m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/c1c3c7ebaa3e43b885add69c7a3155e9 : SUCCESS in 47m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c798d7bd088040beab44314e29d69df8 : SUCCESS in 1h 30m 56s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9613eb2079dc46749596bb21449e7a93 : SUCCESS in 1h 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"75f037ed1f1b360eb44a20c45acbbc7b94e9325f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-16 19:25:54.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"3e483bcb86cd2ed5887c613d8139480da6a6a1a2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 01:11:53.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1\n\nI think we can backport this.","accounts_in_message":[],"_revision_number":1},{"id":"e315b4ece72868dee4e4d7886a49a6b02ef8cb89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 01:12:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78ab9697f40713d27e41558f29d82931fe61f843","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 01:12:07.000000000","message":"Patch Set 1: -Backport-Candidate\n\nOh derp, this is already a backport, missed that. :D","accounts_in_message":[],"_revision_number":1},{"id":"19bab9eadd14f637a18ed496d0376323cf7f1523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:06:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce144485b5e4655bb21f834d6c179ad : SUCCESS in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1da0e0cc32ab42eab3e6a3d93d3a08d3 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/91f3213c3deb48bd87ace23e8cc23960 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d840b656c14f6c9fb5fde95eac3cf9 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d251780691fc4f5494a39c3498c5c6fe : SUCCESS in 8m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d4872034f57466aa28c1d2cd436b558 : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4fd4bec821154bbc91a984bcae9bfd39 : SUCCESS in 4m 48s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/78189c92a5e140228acc52a9df001ade : SUCCESS in 1h 19m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/19bf497f76444fd28863959228fc868f : SUCCESS in 1h 06m 11s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6497aadfd27a47cf8737acafa92db277 : SUCCESS in 55m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/51e231ec3cc747fa9cb5e6c48b1b14cb : SUCCESS in 55m 01s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/91a69a22e1e84010bd8fc734b9a7ec00 : SUCCESS in 1h 12m 05s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/27c2910ab4c3435d90138099fe1e4cfa : SUCCESS in 1h 04m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6e39e96966b432a97d47b73a1675fdc : SUCCESS in 48m 26s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/17b47f37481f4ee8b8bd5cf4cf0bb55f : SUCCESS in 41m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a7c49210bb54c59b35267c9e564989d : SUCCESS in 1h 31m 30s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8f6f22eb3ab049739d1e031b18d4dbe8 : SUCCESS in 1h 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d789b52af120a5ce01555d552eba56ca65d9460c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:06:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ceb4ba7b0f2481e0ec0bd42f18673350c4e645bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:07:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9541d588864fb38804327ef5df809d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec76e763076b845f3808bd6eaa94c789ee0f74cf","revisions":{"ec76e763076b845f3808bd6eaa94c789ee0f74cf":{"kind":"REWORK","_number":1,"created":"2019-09-23 14:05:26.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/683968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/683968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/683968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/683968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/683968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/683968/1"}}},"commit":{"parents":[{"commit":"431d9c9b95f7a7ac8b15315ddd39131f065d69db","subject":"Fix l7rule API handling of None updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/431d9c9b95f7a7ac8b15315ddd39131f065d69db"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-19 23:16:35.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-23 14:03:33.000000000","tz":120},"subject":"Improve the error message for bad pkcs12 bundles","message":"Improve the error message for bad pkcs12 bundles\n\nWhen a user loads a bad pkcs12 bundle or one with a pass phrase into\nbarbican and then uses it for a TLS-TERMINATED listener, the error\nwe return the user is misleading[1].\nThis patch improves the error message to point out that we got the\nbundle from barbican, but that it is unreadable and/or protected\nwith a pass phrase.\n\n[1] Could not retrieve certificate: [ ... ] (HTTP 400)\n\nConflicts:\n    octavia/api/v2/controllers/listener.py\n\nChange-Id: I6ad0349dba62b1141be07bfb0e40171e9f7a91b9\nStory: 2006587\nTask: 36713\n(cherry picked from commit a0f4335c38cdf5d5c91877975b8f3c7b28e169eb)\n(cherry picked from commit a501714a76e04b33dfb24c4ead9956ed4696d1df)\n(cherry picked from commit 08916abd2b6c24dc207dfe3560138901661c2922)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec76e763076b845f3808bd6eaa94c789ee0f74cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec76e763076b845f3808bd6eaa94c789ee0f74cf"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
