)]}'
{"id":"openstack%2Foctavia~684700","triplet_id":"openstack%2Foctavia~master~I301cc9fc6b41e9e97f051df29d768b172cade636","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I301cc9fc6b41e9e97f051df29d768b172cade636","subject":"Fix new pep8/pylint errors","status":"MERGED","created":"2019-09-25 10:37:50.000000000","updated":"2019-09-25 21:48:54.000000000","submitted":"2019-09-25 21:47:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684700-1569448050698-94a46b85","meta_rev_id":"5349309cc345f1638a67da30eb16d412c227acc0","_number":684700,"virtual_id_number":684700,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 21:47:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-25 19:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-25 13:38:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-25 14:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-09-25 13:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":1,"date":"2019-09-25 15:55:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-25 19:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 13:38:43.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-25 13:48:28.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-09-25 14:12:56.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-09-25 15:55:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-25 19:32:48.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-25 21:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e631d8124fe0467be7a7cd7a4f4a2ff29966c9de","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-25 10:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cb8996c551dee4b4a9571278bd2ef34b6af6c5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 13:17:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbc4204b99294b8aa0e65c5023a5d5bd : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07d8d9bfb5df49479ef798ba628b24e0 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407d892cd79941a899922a8789cbc95c : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/304c63075a344920963cf16a7e3cf3d5 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb29dbbe33564655b359e28bdbe8a1f3 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5a733ee0a524026a6310c8b21e2a8d8 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105cbd1d1a7d4c87b7e8b976360ed4f6 : SUCCESS in 11m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f2b790722c16443aa0ca4afc72bcac19 : SUCCESS in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e7ae694bf1d4464ab0c14413ff525b80 : SUCCESS in 5m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a8a90e64c23e49118b626fd5b0e7db5c : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6a641ce155649f1902fe4c69bd7ec65 : SUCCESS in 5m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93430f9c108d49c885745f83242a6092 : SUCCESS in 42m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ed1b478a47184059afd6adc8861dea67 : SUCCESS in 38m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ed5e3eb1c0a466e9da41a766f0e6bc5 : SUCCESS in 1h 32m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2aeb15d4186a48fd9e7c1697b4265c53 : SUCCESS in 1h 36m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/660a5941d9f2478a996316e151544511 : SUCCESS in 2h 08m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/113e39f104f0481eaeb67e6b486e0d8b : TIMED_OUT in 2h 34m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1163ba4911b4540965226f6e6975ab8 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8329da15684b41d0b793036189624771 : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ab86f3c7a474f79bdd2a3ee565d06e2 : SUCCESS in 54m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7cc8a478d7f842fea67f72469a95000c : SUCCESS in 1h 21m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90fefb193ac348af81c633529403d779 : SUCCESS in 1h 08m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/15edf7e0e3d04e249304854d8b0f2e9f : SUCCESS in 1h 18m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1d2be1ca5fc54ea1800c144bdc905e69 : SUCCESS in 1h 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da4eb65a89c0cad33023d87b3fc7e4ece292625c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-25 13:34:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"500351fbe2daa8c04744b1433681dbe11c0adc8b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-25 13:37:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3d577bde4a30daf8946acdebf3c31fa09548fee","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-25 13:37:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd98d964ce778f44b620b42be61df42700047333","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-25 13:38:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b623a765cbe02cb474e43a7e75db4f880754358d","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-09-25 13:48:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"19ac78164efb70f9d1c3fec1a41d592b4eb66f79","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-25 14:12:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aad512d5b983e5edbcf4ef17dfca33c3abd98a9f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-25 15:55:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b655f1464e2481291298b33a7d12ac83219a0ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 16:18:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b643c75ee20e4d13b9bea5c158949012 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/222087e7ea414f8c9f6a555c7212deb8 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf47b5eaaa74d13ba62d1d60af926b1 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b33de4b439eb40579d3c19fb0283e014 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d1e4bafe9ba468d9ae2d57329ee8347 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b21728f1b6d544ffad6884819cb93e3f : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0018c212bd22474b8ab201ec130a6b38 : SUCCESS in 13m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/66186b55ec374f34903f940af0be64df : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/632d1907e9024bb8b41acf50c4849fb8 : SUCCESS in 5m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1ad0e4357a0f4516a121e6dc2f1cd189 : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00fc4c462174488d9de466293f03f527 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c8d5142d25044a81816db0a22a461c92 : SUCCESS in 46m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5dbc4b02757d483b95bc65225a238bb1 : SUCCESS in 47m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbf7d312d3a645d18d167cb4589bad68 : SUCCESS in 1h 47m 25s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b190c256dcd949e1a349f1d1837d6860 : SUCCESS in 1h 19m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/814732be89a341e9b9b61e532b7c8ea4 : FAILURE in 47m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f6cd69924c74e05898599c17944d198 : SUCCESS in 1h 56m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0be553cc9244494eb2a428a8796243a3 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e032ef5879fc469cba8b1b7559fa7390 : SUCCESS in 1h 59m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da32deb1ee104968be647b62711d5b95 : SUCCESS in 1h 06m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/84529429607149e6b43889fccfc6b9d6 : SUCCESS in 1h 28m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbf4b97af67f4de786576f666e475dbe : SUCCESS in 1h 14m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2e09095fb88d45a1bdd7d1a440d07ba2 : SUCCESS in 1h 19m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9a8b8cf403ce4b199deec0dd18d90581 : FAILURE in 1h 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"055be222a1945b443af7cd42441f431986afbd02","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-25 19:32:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2211cd9f48ef24bd112ba13ae66c2418aa30196d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 19:33:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e024b589e59e5ea7214b8904d5eeeb03fbd15d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 21:47:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f97962b10b474e58aec7dec9490cebcc : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca75902f62347559a3d51b7bfc88119 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2be667bd818948139a271c1094db11c0 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5438ad9d96ab4c3984cf8fa20b95013d : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/283105665bbb4f6abf375e53877a29f4 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482c4534e6ac4ed9830fadfd51c75c7d : SUCCESS in 12m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ce39b6845634e3a95bb10afaa4a2674 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/03fdd137f9fa46bb87a36c399d0a362f : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ec06171798e4569949a583f94bccfdc : SUCCESS in 51m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b58ddc0d15c1435ca7dcaf04cf2f5bf3 : SUCCESS in 53m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a755cc3061a4457a987821f3f8892629 : SUCCESS in 1h 34m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bdbbe352abce4905bc23567759c28cbb : SUCCESS in 2h 12m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f4bba3672e8423aa76736c9556c6ccd : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4cbaff0cf1bfcfaff05b11ae78b80f1e7cc350c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 21:47:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5349309cc345f1638a67da30eb16d412c227acc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 21:48:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1741590fff42b4a0ca943812dc3b75 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5ffd2ca4048fe5ec40b63a4afd5109997834594","revisions":{"676a1ba75bf9a245331ef710df5eeed4b02fa196":{"kind":"REWORK","_number":1,"created":"2019-09-25 10:37:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/00/684700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/684700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/684700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/684700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/684700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/684700/1"}}},"commit":{"parents":[{"commit":"8c8f447d291a6ac5047a77f002afb220bf69acd7","subject":"Merge \"Add unit test for failover of spare amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c8f447d291a6ac5047a77f002afb220bf69acd7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-25 09:52:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-25 10:11:34.000000000","tz":120},"subject":"Fix new pep8/pylint errors","message":"Fix new pep8/pylint errors\n\nWith new pylint release (2.4.1), new warnings were triggered:\n- unnecessary-comprehension\n- no-else-break\n- no-else-continue\n- import-outside-toplevel\n\nChange-Id: I301cc9fc6b41e9e97f051df29d768b172cade636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/676a1ba75bf9a245331ef710df5eeed4b02fa196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/676a1ba75bf9a245331ef710df5eeed4b02fa196"}]},"branch":"refs/heads/master"},"d5ffd2ca4048fe5ec40b63a4afd5109997834594":{"kind":"REWORK","_number":2,"created":"2019-09-25 13:37:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/00/684700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/684700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/684700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/684700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/684700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/684700/2"}}},"commit":{"parents":[{"commit":"8c8f447d291a6ac5047a77f002afb220bf69acd7","subject":"Merge \"Add unit test for failover of spare amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c8f447d291a6ac5047a77f002afb220bf69acd7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-25 09:52:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-25 13:36:55.000000000","tz":120},"subject":"Fix new pep8/pylint errors","message":"Fix new pep8/pylint errors\n\nWith new pylint release (2.4.1), new warnings were triggered:\n- unnecessary-comprehension\n- no-else-break\n- no-else-continue\n- import-outside-toplevel\n\nChange-Id: I301cc9fc6b41e9e97f051df29d768b172cade636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5ffd2ca4048fe5ec40b63a4afd5109997834594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5ffd2ca4048fe5ec40b63a4afd5109997834594"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
