)]}'
{"id":"openstack%2Foctavia~684705","triplet_id":"openstack%2Foctavia~master~Idb9e0a9846ff419cd2915865954b19396699e35d","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Idb9e0a9846ff419cd2915865954b19396699e35d","subject":"Fix confused comment in create certificate script","status":"ABANDONED","created":"2019-09-25 10:49:23.000000000","updated":"2019-10-12 01:11:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a1ada2f509fb7c16b137a5c7957aa16909a8c2e","_number":684705,"virtual_id_number":684705,"owner":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-25 13:15:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-12 01:11:04.000000000","_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-1,"date":"2019-10-11 21:17:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-1,"date":"2019-10-10 20:22:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 13:15:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-10 20:22:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-11 21:17:43.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"8dc6b23b054c272e17f4197d8c89975cf9e132bf","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-09-25 10:49:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c26b0b78efa2d0b8e6314336ff2d91476613aef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 13:15:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc669a065bb64a578e6bd6da029a45b9 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ddc23977e804faeb87e48c265cc3393 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df53a81fd0fe4da894c4346a584d4533 : FAILURE in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22abb2d5e7b84d729cc50938da11910e : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19ce96e34d44677b7ac660eecae175d : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2b41ac4ff9445249b6dc4444e34eaa1 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48aedb64cc7545d8b025cd47f31585c7 : SUCCESS in 13m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d0f8cd3b62f4ea49c94d5a869b4f7d3 : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a750642b9eee41e4bc835d02328ef56f : SUCCESS in 6m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4abb9bb56ea7493e899c7a49f05dbb4b : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7524e59193974a55bb15f1680f7f9241 : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa39259916944f648aeddc07abb71e34 : SUCCESS in 38m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2f8de89245ae4576ad349d3df464295e : SUCCESS in 43m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21d3bebdc8084e3386803815b9aa6545 : FAILURE in 1h 59m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d316a3c593cd4ed6951afec587bee23f : FAILURE in 1h 50m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dd740634f8ad42ca9125aa7c7cf42d9d : SUCCESS in 2h 10m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/51392aa7f5cb438abd9729f78b52a025 : SUCCESS in 1h 47m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1e0363b6db04ad3be4075af3a9ad13e : SUCCESS in 1h 50m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3acdd6e23c9e4a9a969fe61d8d26460d : FAILURE in 2h 08m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7346e2a9f174457ebe7084b26f1e01fc : SUCCESS in 55m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5eff98f46dc14770abeb85b184332a73 : SUCCESS in 2h 19m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46bed0ad491f4bdf8fe42162473ba1d8 : SUCCESS in 1h 19m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fe42a03d221a40eebd495aef383da902 : SUCCESS in 1h 24m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/478525513f384337aa3846fdae437580 : SUCCESS in 1h 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5aa3079a511e54512006c4a8162293dba76b9816","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-10 20:22:19.000000000","message":"Patch Set 1: Code-Review-1\n\nper the certificate configuration guide definition, the existing working is correct.\nhttps://docs.openstack.org/octavia/latest/admin/guides/certificates.html#phase-two","accounts_in_message":[],"_revision_number":1},{"id":"a2be27429fb7392caf90ad47c797754e2296657a","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-11 09:28:43.000000000","message":"Patch Set 1:\n\nHi michael,in my understanding,this certificate is a client certificate used by octavia controller and will be present to Amphora.If this comment is correct, I am confused with the corresponding comment in the bin/create_dual_intermediate_CA.sh script, which is \"Create the client controller certificate\".","accounts_in_message":[],"_revision_number":1},{"id":"7e1ad6c98e44186916b34d96b25badd468316aca","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 16:45:44.000000000","message":"Patch Set 1:\n\nWhen I read \"controller client certificate\", I interpret this as \"The controller \u0027client\u0027 certificate\". So it would be accurate describing the difference between the controller\u0027s \"server\" certificate and \"client\" certificate.\nGranted this terminology is confusing. I tried to make is as clear and consistent as I could. Let us please discuss anything that can make it better!","accounts_in_message":[],"_revision_number":1},{"id":"e5d5e9fa9ccb07255471451a2e17dfb7f210f244","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-11 21:17:43.000000000","message":"Patch Set 1: Code-Review-1\n\nI think that a full rewrite of the way we name our certificates would be highly valuable, for example entirely removing any use of the words \"server\" and \"client\"... Instead name things using very uniquely -- perhaps \"controller-identification-certificate\" / \"controller-cert-issuing-certificate\" and \"amphora-identification-certificate\" / \"amphora-agent-certificate\"? Something like that?\n\nThis kind of tiny word order swap does not seem incredibly useful. Also, I believe Michael is correct that the existing naming was at least \"consistent\", though having now read through several times to try to verify that, my head hurts and I\u0027m going to stop.","accounts_in_message":[],"_revision_number":1},{"id":"fed3bec7c7b888107ee63562cc266841df739085","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-12 01:11:04.000000000","message":"Patch Set 1:\n\nI agree with you,Adam..I think that not only the two way communication but also the certificate\u0027s usage should be concerned to name our certificates.\nSince this change disturbed the consistence\",I will abandon it","accounts_in_message":[],"_revision_number":1},{"id":"7a1ada2f509fb7c16b137a5c7957aa16909a8c2e","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-12 01:11:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc6b881c58a7088867349e69bfc237e2d20e1bd5","revisions":{"fc6b881c58a7088867349e69bfc237e2d20e1bd5":{"kind":"REWORK","_number":1,"created":"2019-09-25 10:49:23.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/05/684705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/684705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/684705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/684705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/684705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/684705/1"}}},"commit":{"parents":[{"commit":"8c8f447d291a6ac5047a77f002afb220bf69acd7","subject":"Merge \"Add unit test for failover of spare amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c8f447d291a6ac5047a77f002afb220bf69acd7"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2019-09-25 10:44:32.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2019-09-25 10:44:32.000000000","tz":480},"subject":"Fix confused comment in create certificate script","message":"Fix confused comment in create certificate script\n\nChange-Id: Idb9e0a9846ff419cd2915865954b19396699e35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc6b881c58a7088867349e69bfc237e2d20e1bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc6b881c58a7088867349e69bfc237e2d20e1bd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
