)]}'
{"id":"openstack%2Foctavia~685447","triplet_id":"openstack%2Foctavia~stable%2Frocky~If3c3cf9be085bb95b4ebbaf71e24f92d42b8d6e0","project":"openstack/octavia","branch":"stable/rocky","topic":"2006293-fix-stable/rocky","hashtags":[],"change_id":"If3c3cf9be085bb95b4ebbaf71e24f92d42b8d6e0","subject":"loadbalancer vip-network-id IP availability check","status":"MERGED","created":"2019-09-27 17:54:15.000000000","updated":"2019-10-15 18:32:30.000000000","submitted":"2019-10-15 18:31:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685447-1571164261089-47151255","meta_rev_id":"4c29c80a4d63e149f0f757fad37ac5b7893dfafa","_number":685447,"virtual_id_number":685447,"owner":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-15 14:41:35.000000000","_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 18:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":2,"date":"2019-10-10 15:03:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-10 14:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":1,"date":"2019-10-10 15:03:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 14:51:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-10 15:03:52.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-15 18:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"290a7f0aeedb9005e9cc980bc652c202cee8d1ef","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2019-09-27 17:54:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b6423316d8f99fd447b77dab096f6028c12dd29","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-27 19:38:56.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease propose a backport also for stable/stein, otherwise it is considered a regression when upgrading from rocky to stein.","accounts_in_message":[],"_revision_number":1},{"id":"321eb3d05a91a31c6ec76b8f44759790a11b3d23","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2019-09-27 20:22:59.000000000","message":"Patch Set 1:\n\nBackport for stein: https://review.opendev.org/#/c/685475/\n\nDo the releases prior to rocky need backports too?","accounts_in_message":[],"_revision_number":1},{"id":"1c76194914a04216127302917d49cd4684a1c2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:47:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/445d5b9be01040299f0954b95234fb14 : FAILURE in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c530e77022514a35acc6c5bf56862d4c : SUCCESS in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0780ecfd13b9427eb869cc4ec9b2d7dd : FAILURE in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db437bb4aacb43f48176792083ae722d : FAILURE in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb3951392f44457b5bd807c3b9f7970 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a768c7fd5094e819f14573a4af7c7c3 : SUCCESS in 5m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6d39b1e0e9b7444a8b3e9c963f621261 : SUCCESS in 1h 20m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8636020c992d4df48fd8099f06f34e13 : SUCCESS in 1h 19m 24s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6f70420fee4462590d9e4a65760bef8 : SUCCESS in 1h 37m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5603e0f4f919482ab14b4cea45c80064 : FAILURE in 1h 32m 59s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/10eacfd7ac91447d97e467d71fd57476 : SUCCESS in 6m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7c6f7a4c772b49d2aea712d92540c091 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bb74abff6ef4ab0bf663d1a734590ee : SUCCESS in 5m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f92490374fcc4fd8b2118ffddc820a29 : SUCCESS in 1h 07m 23s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cda75e025cf9487bbc35509511682087 : SUCCESS in 1h 04m 29s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/40e6ca7a1b47443e930f01c1035f07a5 : FAILURE in 1h 26m 29s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9c473109b8694090aabc00b1215aac39 : FAILURE in 1h 26m 38s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ed52320f6782475aa4330b964dfd2af0 : SUCCESS in 58m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e98f9d613f634184a014d30c073cffb4 : SUCCESS in 51m 28s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/3bdbe03728cf4cb781a0d44361b7022f : SUCCESS in 54m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5660d677fb94232abde7972999255c2 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b84ec75bb8b14f7e93407be2eb1c12f6 : SUCCESS in 1h 29m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f911514b56eb4c34b7ec69675a806189 : SUCCESS in 1h 23m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e745ab996da413dbd12cd9fa5d67093 : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8c7ee02970269749201bb65c431abc18b1976c20","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-28 08:19:38.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Backport for stein: https://review.opendev.org/#/c/685475/\n\nThanks!\n\n \u003e Do the releases prior to rocky need backports too?\n\nIdeally yes, but not strictly required. If you don\u0027t have the resources to, I might try to backport it to Queens. The important part was not to skip inclusion of this patch between releases.","accounts_in_message":[],"_revision_number":1},{"id":"ecb6c77c7416303ccaf6a75c551df88a9f44b8cb","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2019-09-30 23:00:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"181725ff384ee14fb86f4cd385c76a591f91fac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 00:44:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38d1ec4b524b412ca78897dc8808e76e : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbcd1a00bd3043e28649902ea1efad89 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6310ea52f73d4454a422fa4c2429847f : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b3c12aaeaa7c4c4ca8801a415813b8fa : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da59c418df641dabc2b6028fc4d9352 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f60617198c54eeab94d76b16f9e7558 : SUCCESS in 5m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/584f74e509c44beea2173a109b5ed371 : SUCCESS in 1h 12m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e6220097c50f4ed59f5a8b14cd1c051d : SUCCESS in 1h 17m 09s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/416faee96c1a4e3b80f8135476200a84 : FAILURE in 1h 18m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/628743e0e6e34d3d900ec064bf18555a : SUCCESS in 1h 02m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d8e7010576746e9a09492f222dc36cd : SUCCESS in 5m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f5366c2e63e14b3aba89d7c7e48c2c89 : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef38c49511634c04a0faacf8d9f3a21b : SUCCESS in 4m 47s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cc175bb46344b66b46cc634efe7c59c : SUCCESS in 1h 03m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b8791fa1a31e4c4bbf27b8bd0d606f95 : SUCCESS in 1h 07m 06s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/43012a57432e40f6a3fbae5e177d4a09 : FAILURE in 1h 30m 42s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/39edc82b53324b558af578b455ce556c : FAILURE in 1h 08m 28s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/f98c7945594e45f9bbfd9cc835a87e30 : SUCCESS in 57m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64b0a7f7cd5e4b97b83fc7cc606c67e5 : SUCCESS in 46m 33s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/f107460ceece49899460fc2b2420d19b : SUCCESS in 43m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ef4490b153f43b494581297e9d418c2 : SUCCESS in 1h 38m 28s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9908192a15af471080172158d3402807 : SUCCESS in 1h 22m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54b7f1c0ba314e3ebf44a06a9b5b9fb5 : SUCCESS in 1h 16m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a850199f5a74cc69930f5bc74f57d57 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8250fbf506908973560961ec230a0fd99f3eb423","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2019-10-01 06:56:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a26e4faebf317fb18a96d209e4181b1c19685e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:02:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1f551cd301b4b1c9cd74744883b3a51 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb29b73269c4fa1baf72793827bf8c1 : SUCCESS in 11m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe71a1fbe0c41a39d29bff51e1b9ea3 : SUCCESS in 9m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/34527736c0ac40c88708e5433de600bc : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af3b12fa4c94b3293ce58ebe7648bdd : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d4a3892b3104957a9dbde527e0c526c : SUCCESS in 4m 50s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a12a78cb005d4f5694e89e14b4a8f4bc : SUCCESS in 1h 14m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/610a855cb3fd4d588e8c72cba4e7d07e : SUCCESS in 1h 16m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d413c3c331c4c14bacf19f0b2096c58 : SUCCESS in 1h 01m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bcece67522dc49b3af5810aac30ca484 : FAILURE in 1h 19m 58s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6f329373d59a49c0a5cb4a5b39ef5795 : SUCCESS in 4m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2a88132e1e4147fbbc5fc25fa3e02fbb : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/461be1ae71274b5594bc47f5827ce44a : SUCCESS in 5m 12s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdb51d6d063c49ac87ecc1343bc571ad : SUCCESS in 53m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6713b771d97b4a92a9adb40d4bdad8c9 : SUCCESS in 1h 03m 12s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/431e518bf63b4ca7a7de3667ae00e9e7 : FAILURE in 1h 25m 25s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/66c9939ce6ed4685b3cae5b63bffe041 : FAILURE in 1h 23m 43s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/9e9ff165e314488099889272dd201cb9 : SUCCESS in 1h 30m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/41a48017f3384917a8b93280bc2771ee : SUCCESS in 49m 51s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/7532a99c7f83492c9cada8f75a914cba : SUCCESS in 43m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3742dc7d2a8b4c0a97ce74d6d455e4fe : SUCCESS in 1h 46m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/64ec1928bf564bf585e630ca41a6517d : SUCCESS in 1h 37m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c78aaa75b1e44ecac1e44e58e0d3ca2 : SUCCESS in 1h 08m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad1b1c5223c84ca284b98b79530211d1 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":3},{"id":"6e0f4a7516f7558b0b36bae10228d6ea93d1191e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-10 14:51:53.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"63a0992de79d03f7fbe44f5c3f0c83316f0bb262","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 15:03:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3c70ed7379cfd45247bf0ce02fe423c3f4045baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:04:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b5462a4fabc369a1db2c295be8dfe1150b55ffe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 17:10:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9520228f4d6d425baf72960d9a7bb473 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88213622c7f14c198c3851f186a96499 : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e5fcfcf5eb04a5db2ec5b570c28604a : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be4b4507c91c4102b99995be2b79416d : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc99a83a7e7426faacdc0f61e9b70ce : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80fdddc65dce46a1a5559d4ded0f6da3 : SUCCESS in 6m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/93e5f87ab5254e74bae5603a1f423871 : SUCCESS in 6m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/08135c869653498789d49c1800fc5cc8 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb43b1073ac04be0896342919826f7a4 : SUCCESS in 5m 02s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afaceb89b7df4c3586fcb145050b0b00 : SUCCESS in 1h 46m 07s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b086c1e455494579b6fce0b14190929d : SUCCESS in 1h 06m 54s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b2dfa1faef28421aac6da894b1c59598 : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af7113340d1541ee9ce6f50d2f5fcb59 : SUCCESS in 55m 21s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e1f5ab6fda644a6e81490b09e3ce3529 : SUCCESS in 49m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/558b9bd8d0834f82996df80adf8d7e6f : FAILURE in 2h 02m 54s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/87c617f8875644f6b73f85f2fa8eefb0 : SUCCESS in 1h 34m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff7ae876d8ce4591993fac42b56b0e50 : SUCCESS in 1h 08m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5397796aee774488ae141a35f7cfc70b : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":3},{"id":"5b7882e46e88a9717ed3208a1f9bb0eeed59c8a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 17:14:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5018935641fbd5b2e59aa6730671e06ddb6a80a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:14:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7794c1cbe68d47318e34e2b7bef0e269 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fb556020c445648bf7c627ce2d0742 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ac90032b00147d4befdcabfab9f6b45 : SUCCESS in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b5cf9ad9e8b41baa44f35b1a8f40504 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6516f0a0484aa6bdd00b9744d339c0 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1198cceb1374a42b8612b5e7be0b612 : SUCCESS in 6m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ce61b169193f4782ad8cd15ddf5d00cc : SUCCESS in 1h 13m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4b1c2d79f29b4b9cbbcbbf542d13ff67 : SUCCESS in 1h 25m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c2a40b1a9b3460bb7138955c5edbabe : SUCCESS in 1h 26m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/73ede37d0bf7438ab0656100c55b1288 : SUCCESS in 1h 22m 58s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0c5102a8084c40f8b06079323bba5d4f : SUCCESS in 6m 16s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/82a5d18df01d4c1496f1fff4afd28408 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc199cbaf5734150809b3b1d3347e851 : SUCCESS in 5m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be541a3f2c854796815dcec05c483b8f : SUCCESS in 1h 03m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6eb7026cbf554dc9ad5943e6f8cc83b7 : SUCCESS in 1h 02m 32s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ff437612ce634726a70cc08fa61fd6d9 : FAILURE in 1h 24m 58s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b5ef3d4b9bc14af9a8f28cb35279bb4e : FAILURE in 1h 31m 34s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d7706d2bd0534840a686dcd57019db07 : FAILURE in 32m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/299449cdc1f74a9c8a797d904af70345 : SUCCESS in 45m 53s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e2e236cb90eb4abda980d90d2b4064b5 : SUCCESS in 45m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58856ce942d34c2e85232f8dd4f34387 : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/337f64bcb44846c78c0cd472efb374cc : SUCCESS in 1h 50m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/092f4a744ac44370a6b5aa90c80e0403 : SUCCESS in 1h 06m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf4a88bf816645f2bb2238ca841bb0e1 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":3},{"id":"57889a8d28ce4be1615a56d4b17a9b10ed27cf3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:14:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e1e056fd9e95a856f4c3bd3ae399a3f19f99114c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 21:01:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a19e85a38fc49f09a623d671877d66b : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381d4784947c4d44ba0d62e63406e0c5 : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4f03cfcf05943618fdcbf7c20f86c33 : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/873432d4ff14492d908e363a39b0a9ca : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2643e38cee6948e7863e792e075372bc : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d2194c6eb474a11914460efbbefb41c : SUCCESS in 7m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e39e1260064840afabd9abc5bf75e7b3 : SUCCESS in 5m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6f0da29dd19946028d6ab0f120dc1445 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6dd5fc6296694c4fbe147f88cc3fc7ca : SUCCESS in 4m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f30eed852b448199695d55e700adb6e : SUCCESS in 1h 02m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/717aa6849856404b902c6579972ea970 : SUCCESS in 46m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/673f9469188545bb979764180b97e809 : SUCCESS in 1h 17m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d2102e47b4541f5893bc202842a6cdb : SUCCESS in 45m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/086b3a083ef14617bd11a9594a13acb3 : SUCCESS in 42m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e299f2d36782486a80dc0da620f45897 : SUCCESS in 1h 33m 29s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/61af0d66f7494cc1a6c5ce790f9c9ee2 : SUCCESS in 1h 41m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4dfe257f7a6444289b5a3ac66b2bbbd2 : FAILURE in 1h 14m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e87b7065046d47fca5189f7abde1a400 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":3},{"id":"988d069037b4ed93293560352cce6023d25694bd","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2019-10-15 14:41:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6596b3468487aaaa7bdf9297d566e8f2b08c3c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:30:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe117754a7714d199c74109fa06e3df0 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7742a1f3340e4e0382fd0e73973c97c0 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/211d52b72fd6411f91c6534308dccbeb : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7258a9e1cb34dbcade6fe2dbe710767 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e277f665f9874f23860abfe08b5fef8a : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8005ddb477647e9a25b7d6ee20f2d43 : SUCCESS in 6m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/da22422213174515b669e7d4d7304c49 : SUCCESS in 1h 14m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/240f0f1c1f894ba6805278024b26049c : SUCCESS in 1h 10m 18s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1901069600b42adbbc70129db76df45 : FAILURE in 36m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1394442deed3482ca95b1d08297eaff7 : SUCCESS in 1h 25m 47s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c0b827a195dc40d1bd60a615374b67a5 : SUCCESS in 5m 10s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2f2b0360462543b4b080c3a08cf164f3 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8617034d07e24c78a2f0364fc09a358f : SUCCESS in 6m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7176d93c49c47e79e5d55db4bce344a : SUCCESS in 1h 04m 35s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e0236af296cf497980ba3958339d6209 : SUCCESS in 1h 07m 11s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ac88716d46784c66a711f2e20e41cacd : FAILURE in 1h 27m 11s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/3d7b57e4fd5f476681c423349d390d1d : FAILURE in 1h 32m 11s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/adaa7721b3704adea677395508a342e9 : SUCCESS in 1h 17m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1def7a625d3d4204af177dd299ef21e0 : SUCCESS in 51m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ddfd5e1e3a0d4f0eb51fccdc91695f95 : SUCCESS in 50m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/facf7bdefd83488b81f772457b5e6b1c : SUCCESS in 1h 41m 55s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2445452679c24d39bcf2cfd54175556b : SUCCESS in 1h 20m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0913416b2a4c41ef9240116dbc4a423d : SUCCESS in 1h 14m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9670d6819c094103b798ff7f903a47a7 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":3},{"id":"3273536d540a764233bd2089cdf33707c3f4c451","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:30:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"791f554e42157bc5c1902a742635f62304409ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:31:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/397f7a761b114790917a3a40824d2931 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/552b5a0c36f34fb184e7a2efb93507fc : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2012ed523a949019bc4975ec2f927d9 : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cbd0ecd542ec437e9abb2b7cb761eb05 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2679d8481d44828987e6b795a4272e : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8173eb95814e47b59c7fb32d006c49b8 : SUCCESS in 8m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/17c7c69c299e425daa1bac2858ffda21 : SUCCESS in 7m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f1b88a2d03ef4754942fd8d8e1092dcd : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0053f2ee99f843faa57bc44e9121c624 : SUCCESS in 7m 17s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b902f7e4fa9a4dc7b90c0fe9fbf401a2 : SUCCESS in 1h 04m 46s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fcfa282368bc46b6800cae6b4aeadedb : SUCCESS in 45m 05s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/de3ce3ed0d354215893e62aa24808be2 : SUCCESS in 1h 22m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6af8383d013649cc8a23dcd5265b0441 : SUCCESS in 57m 08s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/2a9177c179df4c5fb0f65c3d08d781d4 : SUCCESS in 51m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a23333f16f7f48b7b2c80c8e235550a1 : SUCCESS in 1h 48m 14s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/88a6e37bc26a429d9b4718449a6e96fd : SUCCESS in 1h 47m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b08665073dc44563a8517e557551470d : SUCCESS in 1h 21m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d5f20da96ab42bc97e039eed1b22b08 : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":3},{"id":"0757cc2f1535888d135b190a0e7d3f04fe2963b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:31:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"403b08c0d497bed45c0207403918f4d5910d627a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:32:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de53eae2a1f4474ba37e3255fb1e45d : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26ffa3ef73dd4048b664f7173b317d57 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8faa42204ef1da2facca2221a649fad303caf548","revisions":{"d6fbb63018e2f7322f791212f3b0d1de4b790440":{"kind":"REWORK","_number":1,"created":"2019-09-27 17:54:15.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/47/685447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/685447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/685447/1"}}},"commit":{"parents":[{"commit":"870dd3483347bc1bbcc860a9310696f9225a5709","subject":"Merge \"Fix member API handling of None/null updates\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/870dd3483347bc1bbcc860a9310696f9225a5709"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2019-07-29 19:41:15.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2019-09-27 17:54:15.000000000","tz":0},"subject":"loadbalancer vip-network-id IP availability check","message":"loadbalancer vip-network-id IP availability check\n\nThe existing code selects the first IPv4 subnet in the network without\nany consideration of ip availability.  If not enough IPs are available,\nthe loadbalancer creations fails.  This patch uses neutron ip\navailability API to check the quantity of free IPs when creating\nloadbalancer with vip-network-id and skips subnets that do not have\nenough IPs for a loadbalancer on multi subnet networks.\n\nChange-Id: If3c3cf9be085bb95b4ebbaf71e24f92d42b8d6e0\nTask: 36004\nStory: 2006293\n(cherry picked from commit cf901539677d24cac0d9541752b4b3c74c2a1635)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6fbb63018e2f7322f791212f3b0d1de4b790440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6fbb63018e2f7322f791212f3b0d1de4b790440"}]},"branch":"refs/heads/stable/rocky"},"82e7a92bcca86bde3b0b60810f4e6b9dfecbf7a6":{"kind":"REWORK","_number":2,"created":"2019-09-30 23:00:22.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/47/685447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/685447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/685447/2"}}},"commit":{"parents":[{"commit":"870dd3483347bc1bbcc860a9310696f9225a5709","subject":"Merge \"Fix member API handling of None/null updates\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/870dd3483347bc1bbcc860a9310696f9225a5709"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2019-07-29 19:41:15.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2019-09-30 22:55:16.000000000","tz":-240},"subject":"loadbalancer vip-network-id IP availability check","message":"loadbalancer vip-network-id IP availability check\n\nThe existing code selects the first IPv4 subnet in the network without\nany consideration of ip availability.  If not enough IPs are available,\nthe loadbalancer creations fails.  This patch uses neutron ip\navailability API to check the quantity of free IPs when creating\nloadbalancer with vip-network-id and skips subnets that do not have\nenough IPs for a loadbalancer on multi subnet networks.\n\nChange-Id: If3c3cf9be085bb95b4ebbaf71e24f92d42b8d6e0\nTask: 36004\nStory: 2006293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82e7a92bcca86bde3b0b60810f4e6b9dfecbf7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82e7a92bcca86bde3b0b60810f4e6b9dfecbf7a6"}]},"branch":"refs/heads/stable/rocky"},"8faa42204ef1da2facca2221a649fad303caf548":{"kind":"REWORK","_number":3,"created":"2019-10-01 06:56:24.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/47/685447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/685447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/685447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/685447/3"}}},"commit":{"parents":[{"commit":"870dd3483347bc1bbcc860a9310696f9225a5709","subject":"Merge \"Fix member API handling of None/null updates\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/870dd3483347bc1bbcc860a9310696f9225a5709"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2019-07-29 19:41:15.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2019-10-01 06:29:53.000000000","tz":-240},"subject":"loadbalancer vip-network-id IP availability check","message":"loadbalancer vip-network-id IP availability check\n\nThe existing code selects the first IPv4 subnet in the network without\nany consideration of ip availability.  If not enough IPs are available,\nthe loadbalancer creations fails.  This patch uses neutron ip\navailability API to check the quantity of free IPs when creating\nloadbalancer with vip-network-id and skips subnets that do not have\nenough IPs for a loadbalancer on multi subnet networks.\n\nChange-Id: If3c3cf9be085bb95b4ebbaf71e24f92d42b8d6e0\nTask: 36004\nStory: 2006293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8faa42204ef1da2facca2221a649fad303caf548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8faa42204ef1da2facca2221a649fad303caf548"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
