)]}'
{"id":"openstack%2Foctavia~685906","triplet_id":"openstack%2Foctavia~stable%2Fstein~I93e5eb5bc69fe4de4c450c09367b319769ef07db","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I93e5eb5bc69fe4de4c450c09367b319769ef07db","subject":"Fix healthmonitor message v2 for UDP listeners","status":"MERGED","created":"2019-10-01 09:00:34.000000000","updated":"2019-10-04 11:22:30.000000000","submitted":"2019-10-04 11:21:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685906-1570188072481-34c20870","meta_rev_id":"9edcb78a6f79675151b68834dc40f8f66dd925d5","_number":685906,"virtual_id_number":685906,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 11:21:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-04 09:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-03 14:24:28.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-10-04 09:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 14:24:28.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-04 09:38:58.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-04 11:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d30659aeb473ee947dbbd92a137c19df14e72a5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-10-01 09:00:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"8e945c47c2d6c48fa472951ed253f68ffa99c0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:35:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2443c0706374336be86f871a81b6129 : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4694810e291c428a844aa2cdb296d1de : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd439110caca4a2fa7dc9489c642b7e3 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0abf9a72595a4be2884ee32dfa44a8a3 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c3c0fbfcc1e049339e4941c51135e6e3 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b496e41ed24ac98410252b1e409403 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce19a58a509453a973e5425632a9c0a : SUCCESS in 7m 21s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d16cc7afba0c45fc94f618910fa3e2f5 : SUCCESS in 1h 19m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/26e632acfc8445f69570caacb21be1ec : SUCCESS in 1h 25m 53s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fec6562634994fa1ba877057026ab4eb : SUCCESS in 1h 32m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d476d431a47149bfaf192b0ebe50e319 : SUCCESS in 1h 42m 03s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fa38a567a218448c8b3b5fd0b5e9b087 : SUCCESS in 5m 15s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/36ee6395ecad4f6e93da93ac345d6d4c : SUCCESS in 10m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8da1a9a4a7054362b8742dcc6c87105b : SUCCESS in 6m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7c5b4c6029542c7bc36e880a1de2656 : SUCCESS in 1h 03m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3fb91ec1998c44379066d069b3795454 : SUCCESS in 1h 04m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b981506a04b5460da3940024931cb083 : SUCCESS in 49m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/944079a060424eb6b350cfc9730a58da : SUCCESS in 48m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/599984e923c3461594366e57f2168d2b : SUCCESS in 1h 38m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d8d8bf80e8e54139bda116760e17bfd7 : SUCCESS in 1h 38m 50s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a976de45bcc04605a430975ce509b861 : SUCCESS in 2h 05m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/47ae76b979ee4c118505a8f20bcaae9c : SUCCESS in 2h 19m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8ec17f1b7d844149a20cc1517220b9a : FAILURE in 37m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d42da7e6c5747b2be0da1d8311fb709 : SUCCESS in 1h 53m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/63a924b7808048c28ef8cfa50b1fb9d4 : SUCCESS in 1h 45m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a82423ad31694061aad8d94eb44906a0 : SUCCESS in 1h 19m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52b863cb42d8429a816ef7f4e94c7c2c : SUCCESS in 54m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2ba7a68b3dd1aa9339ce99d0c5a299573345379","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-03 14:24:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"739f47b2a97f401da00d3b3d39fa44e073594d55","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:38:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e07590ae118d0d750d525a63cc36b45288478da4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 09:39:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fdb400cf6616f1adc8e3f0d02820e55dce5104a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:21:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d3b2e171565f75bc6c0a76e8a2d21570ec1b7cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:21:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7665997d25c84697a531cf2b9ba950c0 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702a875424e6429c9ae01ec2602faded : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f33b57a53ed4feabc71aa51ab77018d : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2f94c02b512e4c999e0951c3004dfa1c : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a84f1290f3e4a9ea4b0460a899f26be : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc8c0c9a9534a188d58b0548c065b8c : SUCCESS in 9m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5b7a54a63c9d4df2905a540e6f9cb748 : SUCCESS in 6m 51s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/deeaf7152dbb4c0c9ec5aea5a0dd57ed : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e4e93ee43a34579acae8cfb586217a4 : SUCCESS in 8m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae6652c70a924e51b959f1857ba858fd : SUCCESS in 1h 06m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bfda9d02b7c948bab6b0844826e6de14 : SUCCESS in 1h 16m 09s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/28bf337d79d143fcb55c90d5d09cab59 : SUCCESS in 1h 25m 10s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/10010a5cea1f488cbb01c097f57fc3a2 : SUCCESS in 1h 33m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef4d0e12ae264c3696cd6759ed42131b : SUCCESS in 46m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d298a9608d19438e9c926e255dca6c32 : SUCCESS in 55m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8d6f985b4e44842956ef007e4c12c7f : SUCCESS in 1h 31m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0a95cdb408834471b95c9efd7dea7494 : SUCCESS in 1h 38m 21s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0f15b85da07e4734a1a5e22fb6c6b68c : SUCCESS in 1h 34m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8c219489ce040369dbac4d833675bb9 : SUCCESS in 1h 22m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9edcb78a6f79675151b68834dc40f8f66dd925d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:22:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681c276ee66d45978456218f487776a7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ecfa0a5552cffbef4027cb2893ec336c2286a7b","revisions":{"5ecfa0a5552cffbef4027cb2893ec336c2286a7b":{"kind":"REWORK","_number":1,"created":"2019-10-01 09:00:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/06/685906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/685906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/685906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/685906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/685906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/685906/1"}}},"commit":{"parents":[{"commit":"6d49c6992885996cd58fd49ccfb0d8559fa4f10a","subject":"Merge \"Fix building configs for multiple listeners\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d49c6992885996cd58fd49ccfb0d8559fa4f10a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-16 12:57:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-10-01 09:00:34.000000000","tz":0},"subject":"Fix healthmonitor message v2 for UDP listeners","message":"Fix healthmonitor message v2 for UDP listeners\n\nMulti-listener LB commit (Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b)\nintroduced a v2 message for octavia healthmonitor.\n\nThis commit fixes an issue with healthmonitor messages for UDP\nlisteners, they didn\u0027t follow the v2 message specification: pools\ndictionaries were stored in listener objects (v1 format) instead of\nbeing stored as in the root dictionary of the message.\n\nStory: 2005736\nTask: 33394\n\nChange-Id: I93e5eb5bc69fe4de4c450c09367b319769ef07db\n(cherry picked from commit cad80a6c7d0638ede02f7c3d8b37e6e0028573c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ecfa0a5552cffbef4027cb2893ec336c2286a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ecfa0a5552cffbef4027cb2893ec336c2286a7b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
