)]}'
{"id":"openstack%2Foctavia~686541","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I5619f5e40d7c9a2ee7741bf4664c0d2d08963992","project":"openstack/octavia","branch":"stable/train","topic":"amp-cert-reqs-fix","hashtags":[],"change_id":"I5619f5e40d7c9a2ee7741bf4664c0d2d08963992","subject":"Fix urgent amphora two-way auth security bug","status":"MERGED","created":"2019-10-04 08:08:52.000000000","updated":"2019-10-05 11:31:23.000000000","submitted":"2019-10-05 11:29:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686541-1570274997355-fe29816b","meta_rev_id":"e0b58546522e8d4d0ac62d02be54da9acc387d10","_number":686541,"virtual_id_number":686541,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-05 00:11:07.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 11:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-05 09:49:23.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-05 09:57:04.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-05 09:57:04.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 08:10:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-10-04 12:48:11.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-10-05 09:49:23.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-05 09:57:04.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-10-05 11:29:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfa48a7413238cb64c09797307cd9c684ab3e7cb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 08:08:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ea197922659a12a329969c8566ff9928b744e9d6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-10-04 08:10:38.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9c5b007bee82c89715fc539698200e315222bc23","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 08:55:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d27513513d48e9b34a8b16455915c241dac65e0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:11:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5562b2c73a4bc852274323702f025eb5b45e1d8b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:14:32.000000000","message":"Topic set to amp-cert-reqs-fix","accounts_in_message":[],"_revision_number":3},{"id":"5dce754c6c98c28e3bfc6f493c6ab1f583140961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:22:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f92cf9a9d674eb4860442ddf89ac36f : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4df59b715cd4fff9f9ddb980a2e3e36 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1a324ea7d24b1c93aa96a67019fb71 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46010e0201ca40878b150caeb16e5949 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45e172caa23f4037a98ec8565603ee3e : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46be60821802436fad9554301624ffc6 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae944964a92a435fb3cd86ac02e482b3 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa09d0ba7886462a8e5f0462b90a4b77 : SUCCESS in 6m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1c5c007b4be44fbeb49ee1597ce934b8 : SUCCESS in 8m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d47aa976aeff40029d33fd14c7f369f9 : SUCCESS in 6m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3c85aea3ddd14c338b0f305fa5d16427 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7953dc84bc564940bb1bb50eb9287960 : SUCCESS in 8m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a361e4de8784c57b7d008cb0426f360 : FAILURE in 1h 04m 22s","accounts_in_message":[],"_revision_number":3},{"id":"534d8fa8c5c9a8dfb173921acb9f996e05d42bd9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 10:28:11.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4abd719f9c76944bee465bd8431bb8d61b85beca","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 10:28:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0d865da005b7db72796f4f778a099c86570f8b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:51:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be0a96354ed34c9abf94b18689c43071 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9535986ad8843449a402504a008a280 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017966dfe1734b1d8fa17ea8b43bffd4 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0addf51e849450a8a15a18d129dab26 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24bb1ebcf1d74b17a82719d053c9c80f : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/855dc9a6141f4341826239d927865c06 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aad6237a2c44132a672aa282fdbaa82 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b32b0b8ce984fa6ba62c309b3037533 : SUCCESS in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/61c5029573804bff95a14cb0eafe9505 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/839dbb236a114c47933be3ed20c21bf6 : SUCCESS in 6m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a84f65b8170e456199265af5e2414ebd : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c8908167c1d42bbb69a1c170bf953e7 : SUCCESS in 5m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c753ee44f1b341af8277b13f93f8db90 : SUCCESS in 1h 17m 27s","accounts_in_message":[],"_revision_number":4},{"id":"2b40ea1e3b13e6b819c0fb35e3ecd928a3d792db","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-04 12:48:11.000000000","message":"Patch Set 4:\n\nwhy the functional test was not running?","accounts_in_message":[],"_revision_number":4},{"id":"633e12cd799c39f5743455d12d4b0321b8904571","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 13:02:22.000000000","message":"Patch Set 4:\n\nFunctional tests were ran: https://3ced6a6d4c5072144511-6ddd9a666081d58184e95a13f7a762f8.ssl.cf5.rackcdn.com/686541/4/check/openstack-tox-functional/a84f65b/testr_results.html.gz\n\nDid you mean Tempest maybe?","accounts_in_message":[],"_revision_number":4},{"id":"bf310dc5b372f803a63b3bb6b4a7b9a64010300e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 13:07:22.000000000","message":"Patch Set 4:\n\nAsked about tempest in #openstack-infra:\n\n\u003cAJaeger\u003e cgoncalves: the release team needs to branch devstack first to be able to do this\n\u003cAJaeger\u003e cgoncalves: that happens at later time, so for now we depend on good testing on master","accounts_in_message":[],"_revision_number":4},{"id":"4a4532f03b24fbfde6cee821a333440364af24ef","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 13:57:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"315baaa8b7f18728f27d7810c778a9f79773ccd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:27:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78573e001a6d4bdbab03f1002af0b957 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d6ec21b0fbb415e8e60998dfc85fe61 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455c5e162c504e1b83032250d7379aaf : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52042003fd314ce2a26ac760bf70a66d : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5779f53321c045088761854cdbbbc0b4 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cb81f8cead441768a61a7fbbf3855be : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1752379df32d40e8b6054ea501677c8a : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f4cfa674644e819d25efe4b2448f03 : SUCCESS in 7m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d8051481a5724954a0f6be4ad42ca894 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a71e3c68e9d24d7a823d85be63aa5f5f : SUCCESS in 7m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f6c5a58b7c94630b84c7bbfa3a643bd : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccc906cb913d4364af4b0de91b882c11 : SUCCESS in 6m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e72806bb827f4e528dc7ae6ff6f9c39f : SUCCESS in 1h 26m 58s","accounts_in_message":[],"_revision_number":5},{"id":"61fd3e1c5a98bbb5ae69001d4a4878ad0423d311","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 20:15:16.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"143dc2b2e7ec8a9715e582341a0c322eb9ba8ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 22:21:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf415425e0f642d9963fb5bfa7901bfd : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c19ce12d2a3b45408af072a9320e2529 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48c2674486f4b98a16943097717aca8 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bf35a740e6744cc9fca5b56acf6d1a6 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b909ed0de5ee465ab507541ad1ec02b9 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ef7236233314ae090169dd412f23f8f : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3774f127f789411ca25dcf90e9995d46 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80a954f3c834f9fb03c353763462e85 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/33ceec26e977476bad12fab2ff21ac92 : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e75c59401a294f7e8fd6c0c904d1bc1b : SUCCESS in 6m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0953868377fd42ecbb39017c9a3dce99 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34352d8f612b4e2fa3e4160fc2327358 : SUCCESS in 7m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4da61886788140319d9952258cb78e0b : FAILURE in 1h 33m 39s","accounts_in_message":[],"_revision_number":6},{"id":"e3804bd271deef7ab7922b4cb7b711101b621e79","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-05 00:11:07.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3998b635b4bfe18b6f79f37b6a4f415f7617969a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 02:11:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d792f06848c49f0a735cb4bc3fad3ae : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db8a09c83fa240e8b2ec5ed12a981a6d : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df7abc77dfd4e839e5c3f4327c234bb : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2071d003b773409d96abc54830e8b7e8 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a37935912ed64b02812add782fa97109 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea8e0f16583f4af2b78788e621449d01 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da92553d50e943b1bcd7dd1bf4d1b813 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3ca2708ca3647458b758aa80c3e8f27 : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0dfede58eafc48e1946a05a66f964743 : SUCCESS in 4m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5347eb3a32674165bcfb10b2d220ac3a : SUCCESS in 7m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a46ad1c26a7341668b0b814b85ab3bb0 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1645fa8d72954395b787b535f71f1697 : SUCCESS in 6m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0478d4c4f6f6453a8d279850e4188a39 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":6},{"id":"80317927479bd373e760ec142e84a77b3dcff144","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-05 09:49:23.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3e4273586146a07f200fa267af0f77d3eb5eef6b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-10-05 09:57:04.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b306cc86d127e1b846e0d53c7252cf9b0d8e14cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 09:57:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3fd642571f3617ccf4ed0f5cc7d2ce436c964cac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:29:56.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36ae166f710740f39f9644fc1bb9cbf4 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f299c638004a4008b7c6875bf0c34435 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3359de7835254495b42ce320da265c15 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e87da3047bbe45fdaf8225a9318452a1 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8cfb71ae838f4943badaaec1e78d028c : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20e593fa03d49068dbab03e215ffd41 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d870f0dcba499ea1291b68eb8f213d : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a426b613a1ae41a380c2ddbbbca0c43e : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db8e361907a24cf2abf71ea4b3d1e9f8 : SUCCESS in 5m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcef35c5a3774f529a3b2e2d96254a97 : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":6},{"id":"362833bc717bd50fea9c160cf7b2ca35ae60ceb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:29:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"e0b58546522e8d4d0ac62d02be54da9acc387d10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:31:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9636d811061642f3b86019542355ce4a : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/620ef937f6454b75acd3f00597b0e41a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6f54167b05cab1ef2721d63ddee99f92d61ea14f","revisions":{"cbc1fd3feda3091f789931618ed3e0e54aaef9ab":{"kind":"REWORK","_number":1,"created":"2019-10-04 08:08:52.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/41/686541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/686541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/686541/1"}}},"commit":{"parents":[{"commit":"39b8f8211f3838c8d5f8fdbfc290e024316d8050","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39b8f8211f3838c8d5f8fdbfc290e024316d8050"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:08:52.000000000","tz":0},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbc1fd3feda3091f789931618ed3e0e54aaef9ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbc1fd3feda3091f789931618ed3e0e54aaef9ab"}]},"branch":"refs/heads/stable/train"},"939eb5f5a03cef07b2da5abacf23719c8f9896e2":{"kind":"REWORK","_number":2,"created":"2019-10-04 08:55:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/41/686541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/686541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/686541/2"}}},"commit":{"parents":[{"commit":"39b8f8211f3838c8d5f8fdbfc290e024316d8050","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39b8f8211f3838c8d5f8fdbfc290e024316d8050"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:55:15.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/939eb5f5a03cef07b2da5abacf23719c8f9896e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/939eb5f5a03cef07b2da5abacf23719c8f9896e2"}]},"branch":"refs/heads/stable/train"},"9672b810b619c44f59dbffeb6922d741e0867d73":{"kind":"REWORK","_number":3,"created":"2019-10-04 09:11:28.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/41/686541/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/686541/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/686541/3"}}},"commit":{"parents":[{"commit":"39b8f8211f3838c8d5f8fdbfc290e024316d8050","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39b8f8211f3838c8d5f8fdbfc290e024316d8050"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 09:11:06.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9672b810b619c44f59dbffeb6922d741e0867d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9672b810b619c44f59dbffeb6922d741e0867d73"}]},"branch":"refs/heads/stable/train"},"fa4268ac3f229d131f2d804009a06a92d6a927fd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-04 10:28:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/686541/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/686541/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/686541/4"}}},"commit":{"parents":[{"commit":"39b8f8211f3838c8d5f8fdbfc290e024316d8050","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39b8f8211f3838c8d5f8fdbfc290e024316d8050"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-04 10:28:05.000000000","tz":0},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nDepends-On: https://review.opendev.org/#/c/686565/\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa4268ac3f229d131f2d804009a06a92d6a927fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa4268ac3f229d131f2d804009a06a92d6a927fd"}]},"branch":"refs/heads/stable/train"},"998163af97c4f1a6288c7a28c5c9f825ad6dc184":{"kind":"REWORK","_number":5,"created":"2019-10-04 13:57:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/686541/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/686541/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/686541/5"}}},"commit":{"parents":[{"commit":"de49210383c6425da550040a05891c068fa22482","subject":"Fix healthmonitor message v2 for UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de49210383c6425da550040a05891c068fa22482"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-04 13:57:50.000000000","tz":120},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\nDepends-On: https://review.opendev.org/#/c/686724/\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/998163af97c4f1a6288c7a28c5c9f825ad6dc184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/998163af97c4f1a6288c7a28c5c9f825ad6dc184"}]},"branch":"refs/heads/stable/train"},"6f54167b05cab1ef2721d63ddee99f92d61ea14f":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-10-04 20:15:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/41/686541/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/686541/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/686541/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/686541/6"}}},"commit":{"parents":[{"commit":"de49210383c6425da550040a05891c068fa22482","subject":"Fix healthmonitor message v2 for UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de49210383c6425da550040a05891c068fa22482"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 20:15:11.000000000","tz":0},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f54167b05cab1ef2721d63ddee99f92d61ea14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f54167b05cab1ef2721d63ddee99f92d61ea14f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
