)]}'
{"id":"openstack%2Foctavia~686543","triplet_id":"openstack%2Foctavia~stable%2Fstein~I5619f5e40d7c9a2ee7741bf4664c0d2d08963992","project":"openstack/octavia","branch":"stable/stein","topic":"amp-cert-reqs-fix","hashtags":[],"change_id":"I5619f5e40d7c9a2ee7741bf4664c0d2d08963992","subject":"Fix urgent amphora two-way auth security bug","status":"MERGED","created":"2019-10-04 08:11:44.000000000","updated":"2019-10-05 11:39:35.000000000","submitted":"2019-10-05 11:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686543-1570275498205-9e252c24","meta_rev_id":"95716bb6d49e6ff34b833770623f55c1cbfdc905","_number":686543,"virtual_id_number":686543,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 11:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-05 09:49:35.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-05 09:57:17.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-05 09:57:17.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-05 09:49:35.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-05 09:57:17.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-10-05 11:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e9519bfc710b5206ce9e52be6bf0decc6159a26","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 08:11:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e65be8c857aeef7620c500c4b216d5dd987a528d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 08:11:59.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit db5c1509a384c6ff837cb5edc9db7ccd04fff93e","accounts_in_message":[],"_revision_number":1},{"id":"425e6829c68a2e7237c22ba7b6f795ca82ead701","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 08:57:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd45ae861112d1ef4cc3ea90c94265d6e5e2e893","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:12:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"49808d6306af1320bd617bcb459cf000e6f07241","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:14:41.000000000","message":"Topic set to amp-cert-reqs-fix","accounts_in_message":[],"_revision_number":3},{"id":"f75bdf4324a67563c5a9e1e2e0530cbc2595d85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:20:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f0b8d13949a47ac9d2804e05f0ed15c : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db91263257ae4aafb0c42735c5b5552a : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c8daa75adc4abc95c4d38404115d6c : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00de2660ee0e44b2bd19f416d95f6e01 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef7884491aaf4c79ba9b732d878cc289 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce1469eba473448c88514dfc80d35dea : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54a64bce77a40959a97aa52b46dbfb6 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710fdc18296c4a41ac3ab0d5878cb9c7 : SUCCESS in 6m 13s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5a34d2b5e7544db89578fae12043be10 : SUCCESS in 1h 19m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c28095e127ee420081451539a025c5de : SUCCESS in 1h 25m 45s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/beb556de54ae41099cd5d415f9502d3b : SUCCESS in 1h 32m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7101cc02ce014fc5adcadc99317a2430 : SUCCESS in 1h 40m 49s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/844e1aa254a0468e86ab8c836cd4363d : SUCCESS in 6m 31s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5701f64ef0434921978e7e41b85d2120 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bfa9c6b335834c09a640c72ab1cf8bef : SUCCESS in 6m 31s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f488551bcb8a41eaa9917f1e377ecdaa : SUCCESS in 1h 05m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/388873d2788542ef9478a161a1dc9d17 : SUCCESS in 1h 03m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd7fd68bef894e3ca3245b777177e4a4 : SUCCESS in 53m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/fc0ae49da6f14145ad708d74843ad9bd : SUCCESS in 49m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e5c8ad1d7ed4a9da5db720b4fb8a217 : SUCCESS in 1h 36m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ce841c13205248409a45cf138ce824ce : FAILURE in 1h 46m 17s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d79fb9a309b846d5bac326b1367fa169 : SUCCESS in 1h 59m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fbeb428e8e7149de8ac6ca2d94edee40 : SUCCESS in 2h 02m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18bee29b30a246d4bcbfa2e054118f50 : FAILURE in 34m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe22f0f1a6fb43729ff151c6463f91ca : SUCCESS in 1h 54m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3007cb81ec984e2abd284f8c33c4eaff : SUCCESS in 1h 47m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7480333c253246aea92f02c7c66b6161 : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01f2c04abb3f4934afe0f84324f54f95 : SUCCESS in 49m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd055415a7ffeafee75f5eab88d1963dcaf4c8af","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 11:40:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"328f95d723a6f9272e8e169e24772c76ae0db24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:55:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db52bb027c624f0ab1550553f8603f5a : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7833d478b5ea46e5af167643d2e03003 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5feb79f26b849f1b1d80a15655c2e6e : SUCCESS in 9m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/427ad9be247b438c8c731a93e56c3fd1 : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/376a28ae3f704c2986ec95a7b1f3b49e : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/608335d45722445ba8435842af2b89e7 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3af3c4dbd894502bb5ff8cccc8b91a0 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32859a304dda41b38f4cad8b80189585 : SUCCESS in 5m 14s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/183cb67eafc547ba9f4084a52fb38e8e : SUCCESS in 1h 18m 49s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4d1b36059b8d4112949fee8c09379bfd : SUCCESS in 1h 08m 53s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbe567f168454ef88b7377fd56058ecd : SUCCESS in 1h 40m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b4ca0df2a7334fe19eb609b40c0b2f3b : FAILURE in 1h 33m 42s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ffc13a454a7843f98ed95c0c8c6f9a0c : SUCCESS in 5m 47s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/17cf0c55939b4df0af331ffbfcd0e923 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4db6e62eaa3f4b90a6107135f5d6c36a : SUCCESS in 5m 01s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a586045a04044628d349af860287652 : SUCCESS in 1h 09m 20s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/229b91c10e7c43e39e129d9a358df9ad : SUCCESS in 54m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20bd0bc4401b453bafbe10f851714240 : SUCCESS in 45m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f0164edf273742378a0c503b93c3325e : SUCCESS in 42m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0da2ca09f07d4270adb1fa9d20114eac : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/61635c8799f74511b65560f29b8e6c23 : SUCCESS in 1h 39m 08s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8fd4e33efc2d4d0e898a116107e0a2fd : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/76b38d7047b74feba97f254f98ee8d07 : FAILURE in 2h 07m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cf4de9dc1424c29b22a3bf1da929429 : FAILURE in 43m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6760212cbd9946aa808bbe73ef02e3d6 : SUCCESS in 1h 20m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/451eff6d67004c709cfe2ed0d3c8dc40 : SUCCESS in 1h 15m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99897a6978f34858b29cff579d5b766e : SUCCESS in 1h 19m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e762b85052e2473cae52e4ea5929ff76 : SUCCESS in 52m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68cb7b11c1470121c439da5d2204368946f25fd0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 13:59:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c79973a8e57bbb81d2b7831e12a0a4b6aaff37a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 14:25:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"56122fe9501aafad7d85178ba42aa6577f8adfc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 17:07:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82b846feddfb4ff8bdcdc1523ae64817 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d54f81d42cc47839a6aa94c6a362694 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb5992b4369492ca3d2d9b022f5b591 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb308aea3ccf41dba31c7fdf918c3c56 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e013227e82dd46e3b852be3297510eb4 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2e9c10a20445fd8000094a4e8dbc4b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef736ef494842228f9583dfcc75ce0b : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e840efcac7c4463aa0abe552ddde479 : SUCCESS in 6m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/97bfb88e9514468099d2059a0ae2ebe2 : SUCCESS in 1h 20m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/53fb1bee8d194bdc89418be00255bc2b : SUCCESS in 1h 22m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/694247f583b34fafa48da8f74845c922 : FAILURE in 1h 24m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9c1b4db6fb8e41f192a4edff88d48644 : SUCCESS in 1h 40m 49s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d72f71f12ee24c68b09f0894f6115dea : SUCCESS in 8m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e77008eea93b4d259fb5eb87ded10941 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c2fe30d5e9c4126bf3a2438f4ea803c : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/695cabd2ac7e41ca80b5e304c8ad49ab : SUCCESS in 1h 07m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f9a43fc14d10452999b97ba77df69e6f : SUCCESS in 1h 10m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e70c9d41c9c5467eb52b2d9888f72fd5 : SUCCESS in 53m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8dabdc4ec4e0428a87d8db9c81c787ed : SUCCESS in 43m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72650423ee224b7e933cbb2bff22f2ee : SUCCESS in 1h 41m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7d869970c46948f7b5f85805d002c45e : SUCCESS in 1h 32m 10s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/351fa1d2adba41ceb5afe52254f8f20a : SUCCESS in 1h 50m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/86b4e73d4df5405191cffa14612fedfc : TIMED_OUT in 2h 35m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59e3420fb3dd4c1db6ca1b7ed41d5b57 : FAILURE in 38m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/942927ad37844122924bb12193dd5725 : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f8646fff5294654ab8305bbc6aa2fd5 : SUCCESS in 1h 16m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd06f0618cbc44c3858edf2f5c7174d1 : SUCCESS in 1h 20m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5750357ea3b44ca8a24c9a651247dbc7 : SUCCESS in 57m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2bc85b6c1c82c07bf1ae6dcdeec1fd948facbe54","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 20:16:05.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"9c6f8bf1db30d9c1396e4cf0d804ed20bfc85aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 23:18:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac4263d3b35a4adbb510f1f7b2790249 : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d78d62a8e4ec4b698e79250fa8b06d55 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40682e00abe44564a40e47d9815b458a : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81277e30568e4ce99068fd08267b2123 : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a0dd5ed2a69841528e11246707d453b5 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0866b39a0ff04f32a3e491b6023767cf : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ae23f43f63471b8d78a7cc24e0d8eb : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e41bf3e3b6442ae91bc56bdf0f90bc5 : SUCCESS in 6m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/20ec41d21c89431b8f71a1cd8c1d960d : SUCCESS in 1h 17m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f8b5ea41fb8148e38365792e5f63166a : SUCCESS in 1h 34m 20s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60748704250a4999979494b299eb3c68 : SUCCESS in 1h 27m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8c6626a66a3f43f289d8e452a4bf495e : SUCCESS in 1h 50m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e33185d07ab47ed861f434e4f605ccf : SUCCESS in 7m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c6d021a199dd40969f32327b8ae9b2df : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc42311874654653b24f608f1698789f : SUCCESS in 6m 47s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4bfe4c01e81c4d8c81ba419a411d1308 : SUCCESS in 1h 03m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e8f394bf58f64cb6a28253400adcf010 : SUCCESS in 1h 02m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c721537433254588ac382bedf1f10e2b : SUCCESS in 46m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/442e4c80c09847f08e16fa4425ec3923 : SUCCESS in 45m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25fc57aa364d424a8cfa6ee1f9e50e16 : SUCCESS in 1h 38m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/62a57a3bb2e34ec1bbbd286ec5c3d641 : SUCCESS in 1h 37m 07s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3d9bd209cf124ba78875b544a5fa89ee : SUCCESS in 1h 37m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8e6b50a37e1c40b7b75ca1990238717f : SUCCESS in 2h 25m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3f1bc42b3904b769faaf647a4b77509 : FAILURE in 35m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86a55e4e81ec460cbc4040a8e2417279 : SUCCESS in 1h 49m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a948e56a3f77496db095fbda46df50f7 : SUCCESS in 1h 26m 57s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba0107b22e0f4f95a25bfb6ba6a3e450 : SUCCESS in 1h 17m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4ab28a7833248b39fb0a5e54a28de78 : SUCCESS in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0795af5693e473479479b92ecdd16c03781b9bf0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-05 09:49:35.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"100e3ad8f722443561eda3141ab55845278720d9","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-10-05 09:57:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4e1ccb0e9f1fe42bc904c8a6e8eaaae324192c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 09:57:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3e42ae5a0765171d504b7abb93f9b8091207c444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:38:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b60607846e7546c1b4a019ab1006a2c3 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec99d4410a2e4cd0818c04797454f939 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5744171d1d84d4e8594847249b79aab : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/201e329693154e26b74512fe8124e56f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/761491ca6d8d42908cef6611a3ef8c29 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ac2dc251f1496daabc2fc7792e06c7 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/945dea0196a1403eb54fdf2c6b89df31 : SUCCESS in 5m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8581f35948624c74a19e30b09eb13d8a : SUCCESS in 5m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d68798d89f384ad59671f9a96bf98eab : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23b4b3d8c9714c3d8c14fe966ed149ad : SUCCESS in 5m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b63c3ee990847cd903037964d7127e4 : SUCCESS in 52m 32s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/95b660bd6fef4475a8dfccb280f51df5 : SUCCESS in 46m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1f6eef8d67ea4577a58e907470244370 : SUCCESS in 1h 11m 00s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/cd289347106e480a8d4b9c4ef9da77df : SUCCESS in 1h 10m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47e72bde7dd24c008dd3718b57172216 : SUCCESS in 38m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/492b42832c5545a2b4269f43105baa68 : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a29201025feb4502aa1faee2b224f654 : SUCCESS in 1h 17m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/77d4871f64984de3bd5bec50ff7c1783 : SUCCESS in 1h 19m 26s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f247fbfc0fa04b348813c0248de213fb : SUCCESS in 1h 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4146400e78144196835bef65f1370f8a : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":6},{"id":"30bea8b96cb19154f788af6f936b498b1c4260c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:38:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"95716bb6d49e6ff34b833770623f55c1cbfdc905","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:39:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c44e764e804aee81ff0f311928fad5 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a40057bdadf4abdb920fff98b0a10fa : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1725517d1d209f26b2275306d83e49c099dcbe1a","revisions":{"271a1d25ffbf74b18132e338fc1dedf46cb9f839":{"kind":"REWORK","_number":1,"created":"2019-10-04 08:11:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/43/686543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/686543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/686543/1"}}},"commit":{"parents":[{"commit":"6d49c6992885996cd58fd49ccfb0d8559fa4f10a","subject":"Merge \"Fix building configs for multiple listeners\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d49c6992885996cd58fd49ccfb0d8559fa4f10a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:10:59.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/271a1d25ffbf74b18132e338fc1dedf46cb9f839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/271a1d25ffbf74b18132e338fc1dedf46cb9f839"}]},"branch":"refs/heads/stable/stein"},"75d032f46d37bc507d3bac854e41b1f3d50273ca":{"kind":"REWORK","_number":2,"created":"2019-10-04 08:57:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/43/686543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/686543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/686543/2"}}},"commit":{"parents":[{"commit":"6d49c6992885996cd58fd49ccfb0d8559fa4f10a","subject":"Merge \"Fix building configs for multiple listeners\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d49c6992885996cd58fd49ccfb0d8559fa4f10a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:56:51.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75d032f46d37bc507d3bac854e41b1f3d50273ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75d032f46d37bc507d3bac854e41b1f3d50273ca"}]},"branch":"refs/heads/stable/stein"},"5f72c1418acda635f27e3d954666b02a85b15c65":{"kind":"REWORK","_number":3,"created":"2019-10-04 09:12:22.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/43/686543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/686543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/686543/3"}}},"commit":{"parents":[{"commit":"6d49c6992885996cd58fd49ccfb0d8559fa4f10a","subject":"Merge \"Fix building configs for multiple listeners\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d49c6992885996cd58fd49ccfb0d8559fa4f10a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 09:11:57.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f72c1418acda635f27e3d954666b02a85b15c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f72c1418acda635f27e3d954666b02a85b15c65"}]},"branch":"refs/heads/stable/stein"},"9111c35dfad1e7455e22b2caebc3a42b083a6254":{"kind":"REWORK","_number":4,"created":"2019-10-04 13:59:20.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/686543/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/686543/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/686543/4"}}},"commit":{"parents":[{"commit":"5ecfa0a5552cffbef4027cb2893ec336c2286a7b","subject":"Fix healthmonitor message v2 for UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ecfa0a5552cffbef4027cb2893ec336c2286a7b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-04 13:58:58.000000000","tz":120},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\nDepends-On: https://review.opendev.org/#/c/686724/\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n(cherry picked from commit 998163af97c4f1a6288c7a28c5c9f825ad6dc184)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9111c35dfad1e7455e22b2caebc3a42b083a6254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9111c35dfad1e7455e22b2caebc3a42b083a6254"}]},"branch":"refs/heads/stable/stein"},"672990df699aa4d2f990d149ea4c0f425d5ce764":{"kind":"REWORK","_number":5,"created":"2019-10-04 14:25:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/686543/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/686543/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/686543/5"}}},"commit":{"parents":[{"commit":"5ecfa0a5552cffbef4027cb2893ec336c2286a7b","subject":"Fix healthmonitor message v2 for UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ecfa0a5552cffbef4027cb2893ec336c2286a7b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-04 14:24:58.000000000","tz":120},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\nDepends-On: https://review.opendev.org/#/c/686724/\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n(cherry picked from commit 998163af97c4f1a6288c7a28c5c9f825ad6dc184)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/672990df699aa4d2f990d149ea4c0f425d5ce764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/672990df699aa4d2f990d149ea4c0f425d5ce764"}]},"branch":"refs/heads/stable/stein"},"1725517d1d209f26b2275306d83e49c099dcbe1a":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-10-04 20:16:05.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/43/686543/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/686543/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/686543/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/686543/6"}}},"commit":{"parents":[{"commit":"5ecfa0a5552cffbef4027cb2893ec336c2286a7b","subject":"Fix healthmonitor message v2 for UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ecfa0a5552cffbef4027cb2893ec336c2286a7b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 20:15:41.000000000","tz":0},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n(cherry picked from commit 998163af97c4f1a6288c7a28c5c9f825ad6dc184)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1725517d1d209f26b2275306d83e49c099dcbe1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1725517d1d209f26b2275306d83e49c099dcbe1a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
