)]}'
{"id":"openstack%2Foctavia~686547","triplet_id":"openstack%2Foctavia~stable%2Focata~I5619f5e40d7c9a2ee7741bf4664c0d2d08963992","project":"openstack/octavia","branch":"stable/ocata","topic":"amp-cert-reqs-fix","hashtags":[],"change_id":"I5619f5e40d7c9a2ee7741bf4664c0d2d08963992","subject":"Fix urgent amphora two-way auth security bug","status":"MERGED","created":"2019-10-04 08:12:56.000000000","updated":"2019-10-05 11:42:34.000000000","submitted":"2019-10-05 11:41:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686547-1570275679881-9cc037c6","meta_rev_id":"73bcf9af8326aac8af219f44241ece0980ffd714","_number":686547,"virtual_id_number":686547,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 11:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-05 09:49:46.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-05 09:58:09.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-05 09:58:09.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-05 09:49:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-05 09:58:09.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-10-05 11:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"716877ddb0b41cde9bbc55698e65e9e8d1a60fd3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 08:12:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/pike.","accounts_in_message":[],"_revision_number":1},{"id":"432a850bf15fb61a40c9ea1ec14da539f8e84540","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:00:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c269b7196b7a37d4b70daca9d92c1e970a58ad3b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:14:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4cf33e822cf8ba2a7943c9b2c33513b0f4b16585","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 09:15:26.000000000","message":"Topic set to amp-cert-reqs-fix","accounts_in_message":[],"_revision_number":3},{"id":"0b35d18f7b7785f953406345e69644b5284caa67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:49:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9647f4f78004cc8ae6ad40bd03280d4 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2bacc6eeebb48008bc18549aa390670 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f5d599b1000246b6b02dac0d7819a76b : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4666b11c1de4c0b8d2072bd354f8a80 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29edfb5ec6f84daa9638dadb6f208fe0 : SUCCESS in 6m 07s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/60ac350672e34d2e83886000d34d5074 : SUCCESS in 1h 12m 04s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7cb78d7c3a24e60b5e9ef91d7b79c56 : SUCCESS in 1h 27m 49s (non-voting)\n- octavia-tox-functional https://zuul.opendev.org/t/openstack/build/63145b084d8e428dbeae7f7261652490 : SUCCESS in 5m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5abdd624e83b44c0acd7f408ac1c63d0 : SUCCESS in 54m 35s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/fbde2b43cb46474fb39ee9539268890e : SUCCESS in 1h 00m 10s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e5a2fc9a69b84e888760f7b1b20e9cc1 : SUCCESS in 57m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a19dc42558543f6a415f9dbbb5173cb : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":3},{"id":"18cf5cc7c570a3a5c80b8f7379d7b1b2737323bd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 14:01:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37303396c14cb50b62198dd0f03a4f2be4ad5768","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 14:27:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aa8878604b6afa0acef729bc54d644d27c4dfd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:06:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0442305829d4c1c9ea92562af44334c : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96872eb335e441cdba7c28ea14510dc8 : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/10b6f84947674ae3abe7104f00417bc2 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd99eb21f6ec41f6a62c97b6efb122c0 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3134b3f99e24ae1b5f573aa08beefa6 : SUCCESS in 8m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/58dd4c3ee13645ed8f6094ede9e1e382 : SUCCESS in 1h 12m 06s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bde022f93b14ef39dc9dbf2f0f6c0a7 : SUCCESS in 1h 30m 14s (non-voting)\n- octavia-tox-functional https://zuul.opendev.org/t/openstack/build/9555f476ff36411fb4f75437c71e74b2 : SUCCESS in 7m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03d8101b99eb4bb0b40d34aae7b72120 : SUCCESS in 1h 02m 28s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6ecd972d9315485ca055a73472fdc89c : FAILURE in 53m 30s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c0ced85129c8483ca3fa0f657a5b7356 : SUCCESS in 49m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7795536ee42a471db1f61259bc233af0 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":5},{"id":"83e4cb7710cc56b7195b863ad1be0819caabc4f5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-04 16:25:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"30f6feb16c4921ed1e4d50708fd9c32080804400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 17:54:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46451b02b508420ab1603f5019304e87 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfeb4e2aaa0042b78f791db5559d62d2 : SUCCESS in 5m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c2c209cfc2b94963b0dc53b7f438d119 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d2f98dd8bb45128cd35251b4cd5f29 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d58159dc1ce243e3b09d337da86e8a86 : SUCCESS in 6m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/637408ef61c547cda11739c828f436c2 : SUCCESS in 1h 15m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c6b7d5050ef4550b023069145c0e5bc : SUCCESS in 1h 22m 36s (non-voting)\n- octavia-tox-functional https://zuul.opendev.org/t/openstack/build/81c1e2443633468a91511cbe6d39b46e : SUCCESS in 6m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec0723bd8a0a4dd298e83c6402e4fa76 : SUCCESS in 53m 30s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f067386e9f774304b684d593f6afd278 : SUCCESS in 59m 48s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e51423f33f6641f993791e9dc4d83084 : SUCCESS in 47m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6451da98e0ec49bbb1283b42a858e26e : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":5},{"id":"6addf27b1150de82a28fedce6aa3a83dd5dc1f1a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-04 20:18:56.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"3672aa32d09db60df09a02c8ebd7c0fe40aacbbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 22:46:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dcffcae98e24a9f97e3c985824884d1 : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1fc62381e7443deaf876734600f869c : SUCCESS in 5m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b644328a52f42bc9ea40ee8d4992c48 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c48be275c0f4bd496c2d02773d939b4 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256f21d6984447b1b5950b30bb039792 : SUCCESS in 7m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2062c4d3011b4804b02c6ec031e3be55 : SUCCESS in 1h 07m 20s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f7aa0d6ebfd45acb55cc06b1c877942 : SUCCESS in 1h 29m 30s (non-voting)\n- octavia-tox-functional https://zuul.opendev.org/t/openstack/build/6c2e540972de4faf8cc12653facff63f : SUCCESS in 5m 00s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53a43f6fd31a49c9957744b4a2a7c281 : SUCCESS in 53m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c658ae95f97e49d1a58c665dff38568a : SUCCESS in 1h 10m 23s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/069befa507eb4ae88a4f7100f603ada9 : SUCCESS in 48m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24111defafb3492b8fea443302eb542d : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d04438fd7dbb131d795c01d871057e8bfeae3d74","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-05 09:49:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e39f744c26732a955a5426477b7eff98a4e61c62","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-10-05 09:58:09.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"69dd0b245bf15648135f9bcd9956676e7a318b69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 09:58:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9f24ad06fa01661a880f26527e042472464c24db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:41:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a8dd51bd5f2f99d2b4d9821758ef66c04eb06cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:41:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8110f8b25f1c4266a1cf16bb654d84b9 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0131fd9c99241a6b367ee66f657af2a : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f185ccd91890411ebf6cc396f75c7e9a : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beab3e791af14f289d91bd5857a5e107 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce7d64bf020b4261953d46510ea75ace : SUCCESS in 5m 52s\n- octavia-tox-functional https://zuul.opendev.org/t/openstack/build/96623df3839e4752865f209a2a4f9b15 : SUCCESS in 3m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36539de22067479cb9915b27d5aca497 : SUCCESS in 47m 43s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7e0bda31b37149a383e73eb4543ccfb4 : SUCCESS in 50m 09s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1f194b9e6f3e4e2f90a798f13462ed68 : SUCCESS in 1h 12m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/196b3c64c83b4d98a43ddbede313cb48 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"73bcf9af8326aac8af219f44241ece0980ffd714","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:42:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ffe833a8b646beaf371bca57f92401 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22a3eaecd2f047c9a793ee950932f6f5 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c2fdffc3b748f8007c72e52df257e38756923b40","revisions":{"a9e92509ffc5ead0c375ea9d7c67737a004933b0":{"kind":"REWORK","_number":1,"created":"2019-10-04 08:12:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/47/686547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/686547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/686547/1"}}},"commit":{"parents":[{"commit":"168cdce13cd951743edf1bd4aa898f447ce14cd5","subject":"Add bindep.txt and ignore sha1 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:12:56.000000000","tz":0},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9e92509ffc5ead0c375ea9d7c67737a004933b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9e92509ffc5ead0c375ea9d7c67737a004933b0"}]},"branch":"refs/heads/stable/ocata"},"94bce5ee9a61c2258afde3983a786de3cb952cbc":{"kind":"REWORK","_number":2,"created":"2019-10-04 09:00:34.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/47/686547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/686547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/686547/2"}}},"commit":{"parents":[{"commit":"168cdce13cd951743edf1bd4aa898f447ce14cd5","subject":"Add bindep.txt and ignore sha1 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 09:00:12.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94bce5ee9a61c2258afde3983a786de3cb952cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94bce5ee9a61c2258afde3983a786de3cb952cbc"}]},"branch":"refs/heads/stable/ocata"},"a1bf9af264dbb554e0852a4697d31d046cb9a755":{"kind":"REWORK","_number":3,"created":"2019-10-04 09:14:06.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/47/686547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/686547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/686547/3"}}},"commit":{"parents":[{"commit":"168cdce13cd951743edf1bd4aa898f447ce14cd5","subject":"Add bindep.txt and ignore sha1 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 09:13:47.000000000","tz":-420},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1bf9af264dbb554e0852a4697d31d046cb9a755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1bf9af264dbb554e0852a4697d31d046cb9a755"}]},"branch":"refs/heads/stable/ocata"},"6439cd94da334a5d377316691cea16e7bfee6ae4":{"kind":"REWORK","_number":4,"created":"2019-10-04 14:01:50.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/47/686547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/686547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/686547/4"}}},"commit":{"parents":[{"commit":"168cdce13cd951743edf1bd4aa898f447ce14cd5","subject":"Add bindep.txt and ignore sha1 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-04 14:01:40.000000000","tz":120},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\nDepends-On: https://review.opendev.org/#/c/686724/\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n(cherry picked from commit 998163af97c4f1a6288c7a28c5c9f825ad6dc184)\n(cherry picked from commit 9111c35dfad1e7455e22b2caebc3a42b083a6254)\n(cherry picked from commit 56616515d06f61b984b0d0fdfeda4688f3d36ce8)\n(cherry picked from commit bce9dede8d1f4c0214d21bb1b8a0303941a685c8)\n(cherry picked from commit 9d154bcccbf99a9dae6c1256c8b08d9be864454f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6439cd94da334a5d377316691cea16e7bfee6ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6439cd94da334a5d377316691cea16e7bfee6ae4"}]},"branch":"refs/heads/stable/ocata"},"5591c57f2dc40a14532d4da180f1e9c9de7c380a":{"kind":"REWORK","_number":5,"created":"2019-10-04 14:27:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/47/686547/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/686547/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/686547/5"}}},"commit":{"parents":[{"commit":"168cdce13cd951743edf1bd4aa898f447ce14cd5","subject":"Add bindep.txt and ignore sha1 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-04 14:27:25.000000000","tz":120},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\nDepends-On: https://review.opendev.org/#/c/686724/\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n(cherry picked from commit 998163af97c4f1a6288c7a28c5c9f825ad6dc184)\n(cherry picked from commit 672990df699aa4d2f990d149ea4c0f425d5ce764)\n(cherry picked from commit fd289ff7590a738690064dea8d96a9ada7519db0)\n(cherry picked from commit 60a277111a427a004e82d6cf0c591d61dc217a6a)\n(cherry picked from commit afa4e7b3bf8324afb994f307081c0e74a6d71971)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5591c57f2dc40a14532d4da180f1e9c9de7c380a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5591c57f2dc40a14532d4da180f1e9c9de7c380a"}]},"branch":"refs/heads/stable/ocata"},"c2fdffc3b748f8007c72e52df257e38756923b40":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-10-04 20:18:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/47/686547/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/686547/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/686547/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/686547/6"}}},"commit":{"parents":[{"commit":"168cdce13cd951743edf1bd4aa898f447ce14cd5","subject":"Add bindep.txt and ignore sha1 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 08:04:20.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-04 20:18:33.000000000","tz":0},"subject":"Fix urgent amphora two-way auth security bug","message":"Fix urgent amphora two-way auth security bug\n\nThe value of gunicorn\u0027s option \u0027cert_reqs` for client-cert requirement\ndoes not take a boolean, but rather `ssl.CERT_REQUIRED` which is `2`.\n\nStory: 2006660\nTask: 36916\n\nSecurityImpact: CVE-2019-17134\n\nChange-Id: I5619f5e40d7c9a2ee7741bf4664c0d2d08963992\n(cherry picked from commit 2c9af84bdfb443adb1c708d2e6d277170ad2b6a7)\n(cherry picked from commit 998163af97c4f1a6288c7a28c5c9f825ad6dc184)\n(cherry picked from commit 672990df699aa4d2f990d149ea4c0f425d5ce764)\n(cherry picked from commit fd289ff7590a738690064dea8d96a9ada7519db0)\n(cherry picked from commit 60a277111a427a004e82d6cf0c591d61dc217a6a)\n(cherry picked from commit afa4e7b3bf8324afb994f307081c0e74a6d71971)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2fdffc3b748f8007c72e52df257e38756923b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2fdffc3b748f8007c72e52df257e38756923b40"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
