)]}'
{"id":"openstack%2Foctavia~687370","triplet_id":"openstack%2Foctavia~master~Ia30435b3bb19acd1d2f9fe217251c393d4f7480e","project":"openstack/octavia","branch":"master","topic":"remove-py2","hashtags":[],"change_id":"Ia30435b3bb19acd1d2f9fe217251c393d4f7480e","subject":"Stop testing python 2","status":"MERGED","created":"2019-10-08 17:56:27.000000000","updated":"2019-12-02 22:26:32.000000000","submitted":"2019-12-02 22:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":53,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687370-1575325445238-51762757","meta_rev_id":"48bfcbdeda02721fac03a700e86b76b19d941cea","_number":687370,"virtual_id_number":687370,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 22:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-12-02 17:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-02 17:09:54.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-12-02 17:10:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2019-12-02 17:09:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 17:23:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-28 14:16:39.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-12-02 17:09:54.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-02 17:10:16.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-02 22:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc9005902a0b59ee57fe2cefa5b085083113433f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-08 17:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c48ee090d7f5e619a0f5d252ef084b979705d4f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-08 17:56:58.000000000","message":"Patch Set 1: Workflow-1\n\nBased on similar changes in neutron and ovn repos","accounts_in_message":[],"_revision_number":1},{"id":"643a25c9effa88e279ac77e35792271e1299d67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 20:20:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5a9766b1dc39442684292f4955d48ad9 : SUCCESS in 5m 04s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/40db809bcfee4851961833fb5083ad0b : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2875117e1a694a7289404de240383cc4 : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b986d3af9fad44d68634532546a21715 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e107bdd47c6434b812735acfb5d000f : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08adcd94dd4048e3ab0b0cf0a386a87a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae5c2d01216346679271ed4229750106 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f1df5003c224a3fa5cf58bcae8cd7e0 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3807553be0b34d26ab17f1e3d795b3e4 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/005fa7883d564db28241d42c4867ac37 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/192a166c746c4848a041be2aacc84719 : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4711190444e849cb97a473682e79a767 : SUCCESS in 6m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d7e9bde06eeb473996107f792db72a58 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9445fffbac74ecd8885998bf3017bd2 : SUCCESS in 5m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/408dba1a258b4e8586ba6eef3a781aa6 : SUCCESS in 41m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3201ec0414744167815ac28b4ea40d27 : SUCCESS in 2h 08m 26s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d779b02ef4a54836a8bfa827bff361e0 : SUCCESS in 1h 51m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5eefa812cf40452ea445ae63fcc2902f : SUCCESS in 1h 26m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bb71b438ac9454990c0aeb40edd0cb0 : SUCCESS in 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ae123b010bfe477e8dd12b2bd5137a55 : POST_FAILURE in 44m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edbe67c82a1d41c693809f2516b5e3be : SUCCESS in 1h 07m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99c2523c825f88a5d7a1cd5d356d0af3a30b1aa1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-09 12:38:25.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"780d97ccd8d392df9c88c08c0c78c328444cc24a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 16:39:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nYes, more please!","accounts_in_message":[],"_revision_number":1},{"id":"d830dbbdea24588720161ff35ddd1e6ab1c8ac09","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:26:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5c5507c1cf354a156a1f90500e3c49367c5792fd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:27:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b417bba93f0852cfc36def8e70900b388e8a3774","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 17:32:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28a17d919446cdea8cd3ddea07b8aae1bdda312e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:37:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2206ff31c1e8fe2f3bff634a35cbac844c56bac1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:46:46.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0a76f8e1a183e15e1ed419697e00fc5bd672833f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:52:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"71b8878306df3f584756ae9b9b6d1c8397591a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:00:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e8fdce26165b441d83835937b2076e84 : SUCCESS in 5m 25s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c674f248f8c4f67bba3c28eb674da5c : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c04e4e7bd3bd46939fef6a52f62b8e6d : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7026d904c05a4234b945c4543f2c372f : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9c8278f3194715af63314c7d1c1b34 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e39c18d4aa774d7a9c51c61d2f6e44e5 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/defdaaf7ad5a46ad91a5411429746208 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8216911f52474658bd29103689c09087 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0a7e6b695a4e689001cde47fa29a23 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b2c50fe497453190225352ee68ce21 : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3ec351cf36e147c29250faa9ae2b7574 : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/55c2b9da3e21414898a8318f2f17b9e7 : SUCCESS in 7m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8575edab2380473d88c6e7cd696c6713 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be65920ecd9e4293aac5ca7e44ade945 : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da5946343aa74ce28437f55c136701c9 : SUCCESS in 46m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6635e920cad148918427046c0e195661 : SUCCESS in 1h 46m 51s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ef10ffd14004146ace669c32860045a : SUCCESS in 2h 00m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4391fa889d83409e84e0fd9842339a73 : FAILURE in 1h 21m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5dc7c03f0d0546edab502a0fabd87dc6 : SUCCESS in 54m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/21dd2dc0aa20418884a55a0f75a463af : SUCCESS in 1h 30m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/961bdb8b7da1489c8fba145bd88d1cca : SUCCESS in 1h 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3a00fa7ef092bf2e6637eedadb8b578732e99eaf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 20:01:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2d4d0cb5dcb290a30af8849cd7efcde401aed99c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 21:22:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"01c835da708f03c7ef840213c6284734666384a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 22:58:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d1c8f1807bb4bbeb16c488f96ba9445 : SUCCESS in 4m 21s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/086ee8db940042079f55d8d25e5ed024 : SUCCESS in 3m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8cbdee5a2dc54ba4a87dfc9e38894065 : FAILURE in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1416124e406b465792cf727455365cf1 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6c5b7e297cd468ba2f4af725992fda5 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f598d737839245938986e612190ad525 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bd7402188364d7d9d166b5b64fe95f2 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44dbf5f3cca240719c2cb0dea4f44e86 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6966584dc1f4251836eab763a8dba16 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e792afc2254977b1138dfb1e0c4683 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37f7e168973e44e2b70c6ad7e4644979 : SUCCESS in 5m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e79f08a2256042d594899293579f0044 : SUCCESS in 4m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6f394abd50c14b839f15a34d51400d16 : SUCCESS in 6m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bfd70266648540a8b8798d473cd926d2 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45f2e5104d014a43ba0304ce2345f8aa : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a74ed3fd1f94b44876c0c53f02ca8ba : SUCCESS in 50m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07e8ca3d0dd248199bffa4060dfc02da : FAILURE in 55m 57s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/257889e14d944cb0bf4b5806ddb9835e : FAILURE in 48m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/689d3d45b573412ba101e17f121be32d : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee787f138bb44a6fbeac735d452e06b1 : SUCCESS in 57m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c46295ff5e8a455399ce81eed95a0cd5 : FAILURE in 43m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a963395dc114bdaa0d60c97fc2e6e06 : SUCCESS in 1h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"02ec4575d99a5aa0089dc3ce71d671192abc2926","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-14 19:43:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3dee667fd44450b3dc252beb33819155f2854b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:26:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/199a95a348924fc49251326b1ce4dc4f : SUCCESS in 6m 50s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36c6daee5e954db098ce9e93cb37b131 : SUCCESS in 2m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/34efe63391634c53a63212cee0f5abe7 : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7919725a3c3422ba803520d2ddd26cc : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfd1f68e097b469a843d7ce43e404639 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81aea01e47864a9eac8d15155176d7bd : FAILURE in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96e9fea7b13445baa356b5c6592e0fde : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c86eda64e48944859fc35af1f6f7c073 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/831a7d848f424ec2945b59719695f088 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4882d00366f54262a2c060bd497b645a : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48f3e4b869684c6cbd6d03c8cdf59261 : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9479b3615ca74878825fd34f6a905201 : SUCCESS in 3m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0aea259f64414f29b20790beb10ddf38 : SUCCESS in 7m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/805dec92bcb9453783c0853ba3708f0a : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c69a67628fc2402ba928050873ae9d9b : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ebd4360391946eb9e01e08355a7bdb6 : SUCCESS in 44m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/375a2db1a4564b20947d18d3083ceb1e : SUCCESS in 1h 38m 19s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/481d4f5bb24045439e654219d555fa01 : FAILURE in 1h 22m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/26976890ee094867a9e53985fc1a6349 : FAILURE in 51m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b007a59748b7464bbf2d7e45f3b3579e : SUCCESS in 52m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb52a0cbb9ac4118885aca50d4ec2b65 : SUCCESS in 1h 29m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dcffa6776834d6181e6e9e4a975e1d7 : SUCCESS in 1h 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ff16044fc7c1353238c3a398a5ca8572ceac380","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-14 22:00:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0b48fd1da6aacfc56e0e6474252671be5814918d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 00:02:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3bf8e4de06334923a69c1a9bf384eb7e : SUCCESS in 4m 37s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/199d23ac0f5b48b0b13f77f4f54ccc24 : SUCCESS in 2m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1cdcd564cabb4eb08054c8bb0b72cc72 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a72e79152e8f4d90929f273bfbcc68ad : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40a60251e0cc45f9acfa280f75668886 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963f95d737ec4bd5a07a6346ed461f56 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c527efc1db344f1aa4dcb5cb56d49f5 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87a46cbd199044d49b800b01be0b6610 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47a9d5d5ad7b43b69f07c5bda29cdd22 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90dcb1cf90424256bf5df9a14eb99445 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e13386ebe042b1b94d473663beb792 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5cb87f794a55426aaf62a0963bfeec7c : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/069de7260b8b48829120f56069c518c2 : SUCCESS in 6m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5494da19c9d44d03a435db788762a481 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f5878c9034e4883bbd3780e02b9924b : SUCCESS in 5m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb44810bb8594635aff747881ac9c6b7 : SUCCESS in 39m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbf960ca172c4ac6a7b6c8d8226f0feb : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/44741423afe241a48c91ff5f0f4cbdb0 : SUCCESS in 1h 55m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e89c4c3451eb4a17a4945fa9aa671eae : FAILURE in 48m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73dd9477802042fcb16e06cf838c2d7d : SUCCESS in 48m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/31f8267cfb7b4f52b7cd41351b6fd287 : SUCCESS in 1h 17m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2046663d41c4e4dbf19a920fd1348ca : SUCCESS in 57m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ac7bd66af966e57d5599dd013fff0f7743bb9572","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-15 16:42:25.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"85cb6b2461bc5536de6cd14bf9eb42b89d0a20bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:42:37.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"3da8226013a41f2d86d120c98d3c19073b8d6883","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-15 19:22:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e60ce5949f5acebfd40012374cd03f334f91e347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 21:09:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/115027dcc8024220b4d7da20e202596f : SUCCESS in 4m 47s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4cf3e4b8751548c3a37af7e5a4517d28 : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dcd3cb6f5af34591b767892dde455a79 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80d408cdaaec4535accebdcb66ce3772 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49c04fcd5a0144b0aec69dc4d0b9fc74 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cc314598d64044a235184f427be52b : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84d637e00f5248ffb994ce8ccb017d03 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f78ce5dfd419456a8001af8e24fc7819 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfef8978e0d04cb5a555004273d021f1 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a833bc99aa4f48dcb3b33b7857afa102 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7cb5e2c0f244e9b985f7daf121771de : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a72191b2c2a242a4a3bfffaf12b450da : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e18df18098c94df48ac80cbd7a9b49da : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9db07aba67c34d89a3fc599f76080215 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e58c160f12f49e2855c7fda4744524d : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11fc3e4e81ef4a6fa81f17a0e07052e5 : SUCCESS in 47m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cab20a58426405486be848cef5fa95c : SUCCESS in 1h 43m 01s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/590ffb514c3c4c9dab78c63c09a8c379 : SUCCESS in 1h 35m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/94d7c85b48a94a949738d83d1ab2ddc2 : FAILURE in 56m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dccb2e103c7a4551904201b16f1e37fa : SUCCESS in 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/48f677c0fc644cd8b9881b31a7ad1deb : SUCCESS in 1h 37m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bfdc1dda2394539baa30f22cc41f625 : SUCCESS in 1h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f75155f8221a508f3252b66ce8401183bc2568a0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 14:32:06.000000000","message":"Patch Set 9:\n\nrecheck\n\n[Fail] Couldn\u0027t ping server","accounts_in_message":[],"_revision_number":9},{"id":"23f85e44331f0074273d409c54c086342868a026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:13:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/28038d735c5d4662a7287a71778b08c1 : SUCCESS in 4m 34s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8342dfed40104c6e884a1b6f3e6effa6 : SUCCESS in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ea1b7dce78d4892b7e5b478d318a9aa : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93650a08279f479d8a68eeee92c86b41 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ff3d0afadb2441ea4f41b3f391d75fe : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e7cc62191442e5981acbb6e8db4eea : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d36f112eb734922962c505c977c5588 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3fb612c4cc747d0b7f63c7e55c6bd6d : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d9f87fbf77f41ad8424b4285e9795af : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4084d712d89a4950ab020d61c83f476c : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eadab65ba5943b3a8a9707586c6c814 : SUCCESS in 6m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/68a729f76e104bd8bb73f797ffda19a5 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fee5ff48917f41bd899e0247dbc71854 : SUCCESS in 7m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be96a77eea3247d6aa4cd206aadf10da : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c4dd7628b1b40b090cb1fc297f22fe6 : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/faf832e9a4a54301a8e1e7ff499d629e : SUCCESS in 47m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3afa34f6f6bb4d34ab46868e38562c91 : SUCCESS in 1h 44m 12s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/55276ecc468f40e7950eebde27ab17d9 : TIMED_OUT in 2h 34m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a04a529adf5f4f65a2884482ce6cee2f : FAILURE in 59m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/314e3560b5474dc890c452489f3a1079 : SUCCESS in 50m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a5d408581fd94ae8a6ba9e5459453240 : SUCCESS in 1h 32m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/095f8ad6aac44f1b805a9bbabdabe1a2 : SUCCESS in 1h 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1b98abd7adf24ae95a4a8a9ba860774fa7e5f089","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 18:29:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0d55f8ca053ea6961393fee85d86f4086f5ebd9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 18:29:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3f528a1c30483333c69670e2b9dbf547744b5e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 22:12:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8eff42e0fef74b228b0371a377ebca60 : SUCCESS in 6m 28s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6464aec9ba314d6ea2090e2aee27e0b3 : SUCCESS in 4m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e37176afc8b24098baabcf1eb6526425 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2355a99d2c2439191eb52605070d2ea : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdbb25aa24a24e779ba6b56e599aa792 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53c5c97f3154f09af6e166c72d679e1 : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edc5079f3fc3488993d784a11b30517b : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01640481f13a497988f029fe40e853ac : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71d9005b9de64644a2308d9aec3b451a : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfacf8edad34bed8e37c9d1a7daea00 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf16ebb68a2475b944b7aa3a968942b : SUCCESS in 8m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/28c8414377114a089e2503ac0db3f4b6 : SUCCESS in 5m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9427970411d04b24b8e081d24127a1c1 : SUCCESS in 7m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d06ac3df6a56499ea162b93cb5ff828b : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/edfd909df8b0421e97cc767b6766fb9b : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a971a527fad24fe58df680b255e411a9 : SUCCESS in 52m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7ec1d9776a54b5f9373e048a43ee9fa : SUCCESS in 1h 40m 35s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c2f47a0874b4916a9f7aee5b1d93127 : SUCCESS in 2h 04m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/07c38eb7e20c4093b7c5b7c04e919986 : SUCCESS in 1h 25m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17adfb4275964858a539ccb4f6711f6b : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3091780740ac4c63af1a2f243946a4b6 : SUCCESS in 1h 23m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb35dd5d881a483d948831ad34ab5e95 : SUCCESS in 1h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"92e291fd5a6b0a999830ec2b2b80f87ef83355d7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-24 15:29:58.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"bc5d13b8173315bae090468268c63f7eec4276ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:35:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/05a0237bac7748b1b02b558aab4e32b2 : SUCCESS in 9m 03s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0d16573302d4b5fa482edc6900c8d8c : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c76a964e317e4be68e09fcb0b0c6cfd4 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98de835a01a5443d8dbac6f0b9ac19d3 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f3d62f4593b403e9f5199e20e174476 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a550e636b04ac3aa0f8a46d666d939 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a424e3a027ba4317b7e4ff1302df15f6 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3adc28e2f1dd4738ba8a969d299e5bf9 : SUCCESS in 7m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/990f53c6600e4cf28a99edb5313f8da0 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a3d01f63524174bec40030bdbb583d : SUCCESS in 22m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20de77218a7a469d9e701b8fc9d1d5bc : SUCCESS in 6m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/abc5372eb8984f3a84891a42fff5bd38 : SUCCESS in 6m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e1f89144e33f425381c7a249571cc2f4 : SUCCESS in 9m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f0d36a84f9e64297a3e91a78c2f2d32f : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a5e17cc2f5a42559b9b3a5813e42480 : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c7deea1d06c4fcbb15200f8187cb013 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/667ce7d01d3c4b25b3db628ae7d8cbae : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d89862374bb64898aa8c72759d87b5d9 : SUCCESS in 1h 50m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10cde54667b6412780386ed9661d8c28 : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af17d5ae95da4132adf299a10dd23a03 : SUCCESS in 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/54884f7ae05c4ec5980aee1942bc03c6 : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b640fbeee9f0425899367dae62bf16f3 : SUCCESS in 1h 06m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dbb1753375e8ea469513a73c1467fb2b20bbceb0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-04 17:14:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"54efc8604495031f0488572f9d0659e1fdecd3e4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-04 17:16:11.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM,\nWould like to see a story against the grenade job failing under python3.","accounts_in_message":[],"_revision_number":11},{"id":"a8ca3ba891422f10cf118f4a1d2b28efb8edec94","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-04 17:23:59.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cba1abbf63169a6225fe3927f81f263a0cd5805e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-06 06:59:27.000000000","message":"Patch Set 11:\n\n(3 comments)\n\nMichael - did you mean we need to add the grenade-py3 job to Octavia?  But I\u0027ll file a bug by end of week.\n\nI also missed some six things and a release note.","accounts_in_message":[],"_revision_number":11},{"id":"d40687d747b1e3c4508fc843227f148fd9d2062b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-06 07:03:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5da971b456422975d24af127f3d0a8e92b830fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 09:17:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/98d4fd875e164b7b81659d11000919ad : SUCCESS in 5m 41s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8b202e72e48493dab05a37a84cd3a0a : SUCCESS in 2m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b293cf733e8b4e259184e5d4862eeeac : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4c6746413184167bde7c41bc772337c : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f079cd7ba4c40089a5a2c2cf92f52ce : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bcdc6e08d7e443ba61f55e08cc4f0a8 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c95b6746edb4def97d2f9ceb8f2f587 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4fcd7b5fc9a47b99420ea7b87b0dd90 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1556004b1dad405b87fa6f1204a179ca : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22581d64c70d4cfa9473b6c5effc0e92 : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e18a8d53aac54fe190fb5db1350dd59a : SUCCESS in 4m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f8de1ca4cd3442d09845a92017a25d0c : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bfca9e3fae3c4a788241ef54593e3070 : SUCCESS in 5m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f384a1b444054e47a64a18aaa2ee3666 : SUCCESS in 45m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf1799eb065346c4898bd061fbf64a63 : SUCCESS in 2h 11m 30s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6555a05934e4ee6bf0d039f831f3c84 : FAILURE in 38m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/62c108622d2d49f0859ee98450d17021 : SUCCESS in 1h 39m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3fa15466571d49d395f667546fd761bb : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/440d483f772f433dbb14677cc315234e : SUCCESS in 1h 15m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5701e6497a8d4e6fada5d0c5e0aba9cc : SUCCESS in 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b04834e1b0392317b2686546e7ba16ff715eaa25","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 15:36:05.000000000","message":"Patch Set 12:\n\nrecheck\n\nLooks like nova broke master by moving to py3, so this might be necessary sooner than later.","accounts_in_message":[],"_revision_number":12},{"id":"349a9ad908b21d669648aed3dbeb26eb787ce387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 17:20:13.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e15e125829cf4a59b6f4be671f6f1081 : SUCCESS in 7m 26s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f158a5486a7641eea8fedb5c501f657f : SUCCESS in 2m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/97e53943db4d41ebb7d802f279fc655f : FAILURE in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0a9005842f400b9d46994c59de0818 : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b9aff8f412c432d8b7721ded59e77ff : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81a88f0174c4c19ac036a35ca499b30 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46dd2d83d08d4c4d860fb79b4ea0e3dd : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62b1d6b26e0848e1b1c6ec37feeeb820 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c8372070fc488a881fbd6b7e2da870 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24c8f4de4128447c8bfb2d3cc2f5daf9 : SUCCESS in 6m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d85cb60f54c8474680a4c769dd02b711 : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f56d3e458a6b4de3a1237623f2ffb7d9 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c80adb0e1a1494e853365613ef11da4 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b107255ab4242d5ace457fedf90f064 : SUCCESS in 48m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30c2c9d727f741d38cdc18f61da3805b : SUCCESS in 1h 40m 51s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc93198467d7402b96467ea9b8628e3f : FAILURE in 21m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6cfb41f39694be183774a7458ce1359 : FAILURE in 1h 20m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6d200fbc02b4266836a7df949008932 : SUCCESS in 57m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d2cee2ba03841e3a1df9c56c646d817 : SUCCESS in 1h 39m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bd18a82b4814aef8b04ddce6858ded1 : SUCCESS in 1h 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cb6ae6258d9f6a698ca19d068e62607282ce3d51","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 17:37:56.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3450ee19b6ea4db11541cf8a6cbadd4b635b5e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:22:37.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6bafbd43b86d4d9aa8ceee9328491d18 : SUCCESS in 6m 04s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e33c906b28f4834ad1691bf660f90d3 : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8abcb363ad074587ba910dcaffc7e6c1 : FAILURE in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc846ec4c9e44ffcbd611a81e670c86c : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad2f799491664cfb94e84c612b7b80c1 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b709436227c459ca51b86f81f96186c : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b33c2235db8472bbd004a0db8bdb213 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26fbbf4496b84f91b075ed3aa35ff9f2 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d18d1f91634516b62edd23dce72605 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dd3276a510e4ccbbf06be40c9b431af : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/63915d15b40848988d6889131faef199 : SUCCESS in 4m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b7d19ef038c84e099f1d7818d40c039e : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a717b8eda7b74fc7b49952d100bda13e : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f8ec3ff11b747b8900375b62aaf5206 : SUCCESS in 53m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7af383438ff94cd0931805be453e355d : SUCCESS in 1h 33m 08s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07f62129396e47ac9ab56e38e548562f : FAILURE in 40m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/80823d98c1eb4204ab8c181d04434523 : SUCCESS in 1h 19m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c065aeef6794b4d9fabdb56377d47c9 : SUCCESS in 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/77de67b844d24e229fd026a9421c0bb3 : SUCCESS in 1h 40m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc3843ebd44d40c2a8b57de75fcfe778 : SUCCESS in 1h 07m 58s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fbab8100cce599211764d22522dcf6011f412421","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 21:55:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"450c57fb96f8260789b2b53a7b9423236ff44440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:34:21.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/80079630b9a549eb92be85b7949ae2e1 : SUCCESS in 4m 36s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04066a58d32c4a189f04abae9f8d3fb9 : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/438d1be5a4b94c0994744487b85868a8 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1153c664ff5e46b9b2fad04e69a9d5e7 : FAILURE in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d7708d1bb364f4fa29aaf2d16f84f5e : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60ffdd5cac114397881bd20613467666 : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/335a196396b040608393ce9c39989095 : FAILURE in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c31cd74d9f604041a5e4f30de454f2c1 : FAILURE in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d79fd0779e45e98f87102ddc32f076 : FAILURE in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3f4343e50ce4ff3bd3449242738e5fb : SUCCESS in 6m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/df0e1863774b41d3905fd273c9ae4b10 : FAILURE in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91279f8aaf2347338d2a5172b6f8264a : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a791ac4293e4fc5bc91e805e677ebde : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7234fe3ac6b8443d95a79b2a62007f56 : SUCCESS in 43m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b6085c3114a4066b549506abb49d8f4 : TIMED_OUT in 2h 33m 24s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1f5d3f81916448595ed26c8fbd5211f : FAILURE in 33m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e83f515a519b4976b49eab788ab3af0c : SUCCESS in 1h 33m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67725fe161ed47b6b66ef449f73c97b7 : FAILURE in 1h 07m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a3374849e2674038919d101315688dae : TIMED_OUT in 2h 33m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32d7c4a3e6a6478fadc29eb3b02f37b3 : FAILURE in 1h 04m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e4a62db2c4894de5d7d60250311648827f41181b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 16:47:52.000000000","message":"Patch Set 14: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"163328a84c7a61d89d6f5c2fe3692b5127886f64","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 17:08:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"27c7297df62ef7fe5b928c01520b710715e070fc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-25 14:03:03.000000000","message":"Patch Set 14: Code-Review-1\n\n(7 comments)\n\nSo, you\u0027re going to revert all the `six` changes for now and simply make this patch turn off the py2 tests, right? I think that is what we need ASAP as py2 stuff starts to fail across our dependencies...\n\nI do think we need to restore some of those requirements you\u0027ve removed though.","accounts_in_message":[],"_revision_number":14},{"id":"9749133b4f068610db207b83e7c2a1733d285c7b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-25 19:44:33.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"a12b1dc3f6fc0b56b7ad7ddb10dae157682e9707","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-25 19:46:03.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"bdc47d844c6c9df06a08688378f09bb0ba569fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:41:32.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c56a4fc9f8bb486dba7338b9d701bedc : SUCCESS in 4m 56s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f7a5221a52742f682e9a5b7791aa17d : SUCCESS in 3m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/05f581a90a2047f79bc146362af8692e : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d7e83f2869145a9a8ff55bdea532471 : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e32b4ad01e104dda936d126e69f541d6 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4314569462f5444f8c2aafb6c496f8e5 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef556c57f2ff43008a5fccbcb2c3c5b2 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7eb40220779b4cd6b883d4fe6f96d213 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e277173cdb435999c9eac40a1b6f87 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7801623074824330a34e766d3f1ecfe0 : SUCCESS in 9m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/161b1bfcb190410dbf65fd288addc00d : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f1aab360503a458987037c556247f8f5 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ca18a6f1ecb4a87bcbce0af7eb5c9a8 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e50984300c548adb56f4495e781395c : SUCCESS in 47m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52ffe6ac5d5948d9954e8ae46d2b9ee7 : FAILURE in 33m 17s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/159d21ed1ebe464e84b392153b122bed : FAILURE in 31m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/07a284360dd74836998b51a5c16dd505 : FAILURE in 34m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee94d06634224dbf95d52a752f280fe1 : FAILURE in 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1498a4e6a008435c80e4473320087db7 : FAILURE in 32m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40648cffd902438fa8b4b7c6f352260c : FAILURE in 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"920988f0446e6a16d82615cbe63f6be23f4e2914","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-27 16:12:22.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"d1455aa718ee983c244e25c15c1247cb80c32387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 16:12:38.000000000","message":"Patch Set 16: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  The key \"voting\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  voting: false\n              voting: false\n          # TODO(haleyb): restore to voting once python3 issues fixed\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 25, column 13","accounts_in_message":[],"_revision_number":16},{"id":"b7f14cadd53af55b64d8cdb8add774e04eebb05d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 16:17:15.000000000","message":"Patch Set 16: Code-Review+1\n\n(3 comments)\n\n+1 for now while I think","accounts_in_message":[],"_revision_number":16},{"id":"4208e32555a60966279325b16560f5655e7170ed","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 16:17:50.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f3bba5e5a27e9dbd85440278fb4b1b720f04f623","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 16:32:08.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"52b5b2d8ba2387db48c16ee2dde47e561a5e8864","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-27 16:38:23.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9312122b1d47dd9310e8964f06d1fd2a2c4de6f5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-27 16:38:30.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"61fbc0eed2b1ada9ff1d0ebbb755ad74e8ca23a2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 16:44:39.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"83eb938724852009555aa79f9b8473040e3ea760","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 16:44:56.000000000","message":"Patch Set 18: Code-Review+1\n\nBut, probably no one runs that full test list so whatever","accounts_in_message":[],"_revision_number":18},{"id":"56c99422e3ea9dc9652de20b5eb2c97e18427e27","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-27 16:48:42.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"0d82028d3d91cd9558edbea74064ee2faa4839c0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-27 16:48:47.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"274d9f7779c4997e5ef78da0c5f8c92c9b2d839a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:11:03.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d6fa6e8ef5114994afa74d150ec5b590 : SUCCESS in 4m 55s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a757883f121435083b5d5d2f7b871f7 : SUCCESS in 2m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/adc4e379141d446b91f5fe879f8b59c0 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c1e98e5d0ac40798ee91dd0635aaf6e : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/783a49e04ff846d098fbf8475ec7d7a4 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c02f418d0f4143a9b614c4966822c4 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d831823c7c9451886535a1e47114b67 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac74aa70ec774a099eb528a1e7d4aae2 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8168547c8f8a4c0098e2a6559b7e1884 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4938870773af4fcdbcd0742ee2b3c0df : SUCCESS in 6m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/400c37bfd732425f94ba652033189286 : SUCCESS in 4m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8bfbcccb6a6e4d369b782e3f97be4021 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3588348c90448b583ab6ebea05315a3 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2c85f9dba4441068ab5db7407947fc5 : SUCCESS in 43m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fe79f3c44f3457691f99dae22ad948a : SUCCESS in 1h 39m 06s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/932406273c7c458d919d4f853bd3944f : FAILURE in 39m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33173ceb53e746c49c1c6a51f018c136 : FAILURE in 1h 04m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6dba53fb5aa54ba8861cc0b95dd7b454 : SUCCESS in 51m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d2a2274c06ff469e9b2ebf819d5cc691 : SUCCESS in 1h 50m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b9d1057f5ab4dc2a86aa6e841bab5a0 : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"8a45ff3591372be9c45aa155c1f64ed627b58733","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-28 14:16:39.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"a9f8a163bb58e5164b0e1b2c8208d66bfdd1e0b7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 16:03:15.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"3dbd68f7463cccb82888ed7e90446a86981be7ff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 16:10:26.000000000","message":"Patch Set 19: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"20624bea7ac0aeb588c112a434bba9626d5dbdf7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 16:12:02.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e09f21cc54043a78dac7da1e712c863dc7f75c2d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 16:39:21.000000000","message":"Patch Set 19: Code-Review-1\n\n(3 comments)\n\nI think the ONLY change this needs is to keep the cent7 test. The rest is as it should be.","accounts_in_message":[],"_revision_number":19},{"id":"dc01d5afb65c7dbab9a5e5609094276f13d7e3f7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 16:41:15.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"5baa20ceb1e4e2f8a7f058e4d890ce6d48f7209c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 16:43:55.000000000","message":"Patch Set 19: Code-Review+2\n\n(2 comments)\n\nOK, I\u0027m back to +2 unless johnsom really wants cent7 to stick around too","accounts_in_message":[],"_revision_number":19},{"id":"dce5150c22dafd9ebf4f7380bfaa670c8a770e64","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 16:52:03.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e5f9c2334f967fa6f5e4a7ebb8308273e0e83505","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 17:01:09.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"033347b037974aa55ad3fbf4e4e43b47df1b08d3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 17:01:51.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"c0e46d40e0df1bafcb47ee9a366cc77cc984da22","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-02 17:06:01.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"fd8082daaf1e6f3d920857b8c3c09df5e55628c2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 17:06:01.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"0c14091216399ab7ec3ac60982c5e7f77a28d3d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-02 17:08:29.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"99d73218facdc842209b230c2590554d38d5e116","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 17:09:22.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f094a3f057b42835ccb12ba4718b4521b9ddce40","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 17:09:54.000000000","message":"Patch Set 21: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"922a00d22a952e79a5a396e860218e935e17ac03","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 17:10:04.000000000","message":"Patch Set 21:\n\nThanks!","accounts_in_message":[],"_revision_number":21},{"id":"78658a5677b0356ee8b2863cdfd864f0beb236c7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 17:10:16.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"0eb2f931c9baa5f039001335c817bbd23fe63291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:11:34.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c70671b5c734c679c4fd8d038b8e331 : SUCCESS in 5m 26s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3dbd3202a6594d26b8d0987a3f5d7dcb : SUCCESS in 2m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d820e04858c4c5f9fc108106c43e188 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e7e400c8c34f698d4c37fe93acfc24 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d83d0c88a6784455bd970a05ecf0aec9 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6ac99dd4114837901d42c5e4e1fba1 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0962c70fffc9405583870d97be34a20e : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed5be98846aa49e4ba59bd986a8b8239 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b537d5cc9ca4460b83b86efb45069aa : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d686a6b00f8348acbae5f9266f33dc4b : SUCCESS in 5m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/72cf928bac6448ac9a9fd89dedc372ae : SUCCESS in 7m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b6ec93363cc549f0ad3989a4d95f5d44 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73c6602778894b2c899bc0e010437cf4 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3de1dd4278e44ebfac1342708ae9a524 : SUCCESS in 49m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58cc19e839464c308656acd10c910b72 : SUCCESS in 1h 26m 51s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/807e6449f059440b9c9594ffc4232989 : FAILURE in 31m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/322f85800fac43d59264b770ffbc8b31 : FAILURE in 1h 02m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4bbb6e33190413d8fad1a74cda4a9ee : FAILURE in 52m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73227253f34b4c41995b35c31677551d : SUCCESS in 1h 43m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29a17bac57c447ba80f0fbeec341c103 : RETRY_LIMIT in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b9834b59e150816891d028a0a98c3885ad2cf59b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:11:50.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"a8bc3982ed031b503da1aceb7e13b84083f8fdd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:24:04.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b396ef6c4b4a4bd09e384b315686fd40 : SUCCESS in 5m 30s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be40b176d4df40c489431c1a3cdb891e : SUCCESS in 2m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f4404bb3c3043c1be28a4452fd0fc3f : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b6102069a3f4b899619a7e89d7ac890 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1dfe496d8fd43ea826f4f276579e660 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c06ee445b4048dba45cb4ab2928afda : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46d90e949e9f453096312bb5a4008b75 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4165447a03748c98d5124f6dacbc26c : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/946bfb82ec04413e886b9ef4c1fa752b : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de537d389c764c0e80b7716046fd85f7 : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b70ba6cedc894a7180adbe18b6f40281 : SUCCESS in 49m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecc5331b565641c3af52193f6b45eb55 : SUCCESS in 1h 44m 56s","accounts_in_message":[],"_revision_number":21},{"id":"87015450b520f55f5ecb4f59fd865f49296394d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:24:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":21},{"id":"6c648eac2172e05144e9bf8d5eaab7795804dd0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:26:32.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/68a4d8c2f4344367beb8dbe97029bbfb : SUCCESS in 1m 06s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fb729ad0f540a39bb0ebcd4bbf534b : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec0a790b9fbd41cd8cd783ac8ded153a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"bc94789dc9e8418a05d92d22f66f90922ffd1ada","revisions":{"60df503d45ef50604ac163e1309d74f1324f0104":{"kind":"REWORK","_number":1,"created":"2019-10-08 17:56:27.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/1"}}},"commit":{"parents":[{"commit":"b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:55:57.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\npy27 unittest tox target is kept because the job is defined\nglobally for all projects. We can remove the py27 tox target\nonce the job is removed.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60df503d45ef50604ac163e1309d74f1324f0104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60df503d45ef50604ac163e1309d74f1324f0104"}]},"branch":"refs/heads/master"},"1ec456d6e6c01285cce0cf5c65572baaec6bdb3e":{"kind":"REWORK","_number":2,"created":"2019-10-11 17:27:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/2"}}},"commit":{"parents":[{"commit":"b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 17:24:24.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\npy27 unittest tox target is kept because the job is defined\nglobally for all projects. We can remove the py27 tox target\nonce the job is removed.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ec456d6e6c01285cce0cf5c65572baaec6bdb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ec456d6e6c01285cce0cf5c65572baaec6bdb3e"}]},"branch":"refs/heads/master"},"9ce96e068168ba2bc8a04fa28d47fbbb143bad7e":{"kind":"REWORK","_number":3,"created":"2019-10-11 17:37:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/3"}}},"commit":{"parents":[{"commit":"b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 17:36:58.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\npy27 unittest tox target is kept because the job is defined\nglobally for all projects. We can remove the py27 tox target\nonce the job is removed.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ce96e068168ba2bc8a04fa28d47fbbb143bad7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ce96e068168ba2bc8a04fa28d47fbbb143bad7e"}]},"branch":"refs/heads/master"},"478fe6a22ca9230cea38e8de2c7be7756617e919":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-11 17:46:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/4"}}},"commit":{"parents":[{"commit":"b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 17:46:32.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/478fe6a22ca9230cea38e8de2c7be7756617e919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/478fe6a22ca9230cea38e8de2c7be7756617e919"}]},"branch":"refs/heads/master"},"b43467df3c13b1e2fd9f405b9badfb6f6cbd2e20":{"kind":"REWORK","_number":5,"created":"2019-10-11 17:52:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/5"}}},"commit":{"parents":[{"commit":"b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 17:52:26.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b43467df3c13b1e2fd9f405b9badfb6f6cbd2e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b43467df3c13b1e2fd9f405b9badfb6f6cbd2e20"}]},"branch":"refs/heads/master"},"3754fa35e642998b76ac669df753d2e44eac7677":{"kind":"REWORK","_number":6,"created":"2019-10-11 21:22:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/6"}}},"commit":{"parents":[{"commit":"e6cd43d6cc2710266a1990c952201c5714a03393","subject":"Remove duplicate keys in sample config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6cd43d6cc2710266a1990c952201c5714a03393"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 21:22:20.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3754fa35e642998b76ac669df753d2e44eac7677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3754fa35e642998b76ac669df753d2e44eac7677"}]},"branch":"refs/heads/master"},"3c8f7b93ad7dedc7ccc54e9678b050017eb27640":{"kind":"REWORK","_number":7,"created":"2019-10-14 19:43:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/7"}}},"commit":{"parents":[{"commit":"3177aade971dbab78e3d02a9888b9003c8d51773","subject":"Merge \"Clean up requirements.txt from tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3177aade971dbab78e3d02a9888b9003c8d51773"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-14 19:42:37.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c8f7b93ad7dedc7ccc54e9678b050017eb27640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c8f7b93ad7dedc7ccc54e9678b050017eb27640"}]},"branch":"refs/heads/master"},"094737c4f32e887bae133d0e10cc8fa9eb55c32d":{"kind":"REWORK","_number":8,"created":"2019-10-14 22:00:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/8"}}},"commit":{"parents":[{"commit":"3177aade971dbab78e3d02a9888b9003c8d51773","subject":"Merge \"Clean up requirements.txt from tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3177aade971dbab78e3d02a9888b9003c8d51773"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-14 21:59:41.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/094737c4f32e887bae133d0e10cc8fa9eb55c32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/094737c4f32e887bae133d0e10cc8fa9eb55c32d"}]},"branch":"refs/heads/master"},"026579699c7a80c9bcac6b373581aa27144bf79f":{"kind":"REWORK","_number":9,"created":"2019-10-15 19:22:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/9"}}},"commit":{"parents":[{"commit":"25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf","subject":"Merge \"Use bandit block in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-15 19:22:28.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/026579699c7a80c9bcac6b373581aa27144bf79f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/026579699c7a80c9bcac6b373581aa27144bf79f"}]},"branch":"refs/heads/master"},"f041ca2173508c4e27a99fad17faf29e0851e300":{"kind":"REWORK","_number":10,"created":"2019-10-16 18:29:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/10"}}},"commit":{"parents":[{"commit":"25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf","subject":"Merge \"Use bandit block in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-16 18:28:27.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f041ca2173508c4e27a99fad17faf29e0851e300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f041ca2173508c4e27a99fad17faf29e0851e300"}]},"branch":"refs/heads/master"},"4bfb573b8585da834bf3938b1c84a8f0d73305c2":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-10-24 15:29:58.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/11"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2019-10-24 15:29:58.000000000","tz":0},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bfb573b8585da834bf3938b1c84a8f0d73305c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bfb573b8585da834bf3938b1c84a8f0d73305c2"}]},"branch":"refs/heads/master"},"902e0c932c4a5054bb3f00620e6c388c902daf04":{"kind":"REWORK","_number":12,"created":"2019-11-06 07:03:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/12"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-06 07:02:16.000000000","tz":480},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/902e0c932c4a5054bb3f00620e6c388c902daf04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/902e0c932c4a5054bb3f00620e6c388c902daf04"}]},"branch":"refs/heads/master"},"f5dc5ee2fa792c58e42ca953dd9944cb9a58158c":{"kind":"REWORK","_number":13,"created":"2019-11-19 17:37:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/13"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 17:37:38.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5dc5ee2fa792c58e42ca953dd9944cb9a58158c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5dc5ee2fa792c58e42ca953dd9944cb9a58158c"}]},"branch":"refs/heads/master"},"47b021d2ceb38956663b3f4d4e1f8ea0290495aa":{"kind":"REWORK","_number":14,"created":"2019-11-19 21:55:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/14"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 20:25:10.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47b021d2ceb38956663b3f4d4e1f8ea0290495aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47b021d2ceb38956663b3f4d4e1f8ea0290495aa"}]},"branch":"refs/heads/master"},"db3f88e546605502a9865e565a7c5569427f0668":{"kind":"REWORK","_number":15,"created":"2019-11-25 19:46:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/15"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-25 19:43:43.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db3f88e546605502a9865e565a7c5569427f0668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db3f88e546605502a9865e565a7c5569427f0668"}]},"branch":"refs/heads/master"},"2aa31325adedd62284b77ede4ec39bd8bedf8c13":{"kind":"REWORK","_number":16,"created":"2019-11-27 16:12:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/16"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-27 16:09:20.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa31325adedd62284b77ede4ec39bd8bedf8c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa31325adedd62284b77ede4ec39bd8bedf8c13"}]},"branch":"refs/heads/master"},"e523a54c64e4aa91cbf7b195ef98edaf7a4ee6df":{"kind":"REWORK","_number":17,"created":"2019-11-27 16:32:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/70/687370/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/17"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-27 16:31:47.000000000","tz":0},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e523a54c64e4aa91cbf7b195ef98edaf7a4ee6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e523a54c64e4aa91cbf7b195ef98edaf7a4ee6df"}]},"branch":"refs/heads/master"},"ce734bc8bedf11f0905bfc56d72013306a5e1aa8":{"kind":"REWORK","_number":18,"created":"2019-11-27 16:38:30.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/18"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-27 16:38:11.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce734bc8bedf11f0905bfc56d72013306a5e1aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce734bc8bedf11f0905bfc56d72013306a5e1aa8"}]},"branch":"refs/heads/master"},"4c79ed5df04ebb0eff7e4e42362fe076e6d0a40c":{"kind":"REWORK","_number":19,"created":"2019-11-27 16:48:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/19"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-27 16:48:15.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c79ed5df04ebb0eff7e4e42362fe076e6d0a40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c79ed5df04ebb0eff7e4e42362fe076e6d0a40c"}]},"branch":"refs/heads/master"},"6fd4570b19f69521ebf6ac5319a9f6088dbc90f6":{"kind":"REWORK","_number":20,"created":"2019-12-02 17:01:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/70/687370/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/20"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-02 17:01:42.000000000","tz":0},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fd4570b19f69521ebf6ac5319a9f6088dbc90f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fd4570b19f69521ebf6ac5319a9f6088dbc90f6"}]},"branch":"refs/heads/master"},"bc94789dc9e8418a05d92d22f66f90922ffd1ada":{"kind":"REWORK","_number":21,"created":"2019-12-02 17:08:29.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/687370/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/687370/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/687370/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/687370/21"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 17:54:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-12-02 17:08:19.000000000","tz":-300},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nChange-Id: Ia30435b3bb19acd1d2f9fe217251c393d4f7480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc94789dc9e8418a05d92d22f66f90922ffd1ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc94789dc9e8418a05d92d22f66f90922ffd1ada"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
