)]}'
{"id":"openstack%2Foctavia~687760","triplet_id":"openstack%2Foctavia~master~Ic194a030edd84106217eb5ac02f9b3190d3a7ba6","project":"openstack/octavia","branch":"master","topic":"duplicate-keys","hashtags":[],"change_id":"Ic194a030edd84106217eb5ac02f9b3190d3a7ba6","subject":"Remove duplicate keys in sample config files","status":"MERGED","created":"2019-10-10 01:21:36.000000000","updated":"2019-12-30 08:58:15.000000000","submitted":"2019-10-10 09:46:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687760-1570700802303-01385176","meta_rev_id":"2fd719b81a864f17f9674afac43763a636ed6082","_number":687760,"virtual_id_number":687760,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 09:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-10 07:54:02.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-10 04:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-10-10 07:54:02.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 04:25:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-10 07:54:02.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-10 09:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3ba0dacf2dedf989846ba220e508a13147228cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-10 01:21:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"823f32be9e3343b1a547bd12a176b855df40f69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 03:50:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/734aae5d67e24e66982dc1598b478c98 : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/215c446724a94f4bace758a83dc060dd : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc5b8a0d91847fba290a04c488d2fef : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/727dbc1c44b44fb2b0383a1d3f5e0697 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636b0bd2148343d98134ffb82462bf82 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94ebd6b899244f8ab3db100706a6d0d8 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa2e430c9d14f0381a4a43a16e51d1e : SUCCESS in 9m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb4145d0d4a9449290a1c7090eda247c : SUCCESS in 4m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24bf87e569b94b5aa068cb913fff38bd : SUCCESS in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b8e6f87df5a4443eaaa0b42982417a6a : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/033aed8384f741ab854f133eecb41fa2 : SUCCESS in 5m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6232ccbd7a140b2a5a324d7bdea95b2 : SUCCESS in 42m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c01d548c897e4c2aab02ff7227bdbb55 : SUCCESS in 43m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/704a883e17f34aaaba89019931fdd3cd : SUCCESS in 1h 44m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/375434deeb994718919dbfda57fef0a4 : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1e7680b7d8f646768bef96f64fc80d89 : SUCCESS in 2h 18m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4349663d34514e1ba9e1e3a59c02aab6 : SUCCESS in 2h 00m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b0e6ff5dae043289e8ed1e640785b45 : SUCCESS in 1h 59m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f91a431a5907490d9ccd65bdbf2878ab : SUCCESS in 1h 04m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/217343a26066455ab9df98ed7fa3b8cb : SUCCESS in 1h 02m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7e11eb608f0a44c6b8777be8a6a91989 : SUCCESS in 1h 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f883c6cc3f8f4788183e11587ae61e36185d0eec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-10 04:25:28.000000000","message":"Patch Set 1: Code-Review+2\n\nSigh, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"9767682bfe86d36ebcd654fbaff8e4ef839eb38c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 07:54:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"424c815f38b9350ced87b68116d5d64a67cf3316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 07:54:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd8a2b405fa83ce1cc1c07b1d6ce5f60d3bc7e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:46:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c09b29bcd9034367ade7c7540caa897f : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec28a19fb12423d9b579562167ab6aa : SUCCESS in 8m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e47477afbb5404ea00cbd8139583c74 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aacc254dc470435eb3b1fce21f7dcc3e : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd5c3291efb54e3784e335b2cb84a550 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac33e7be7f554a61a19c7ec9f97adda6 : SUCCESS in 10m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d93106604914763a5c290bcf1853efd : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/898ab256f7054902af2fe1ac66729e06 : SUCCESS in 6m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/054d09ef96be4787ab323041df6d4fa8 : SUCCESS in 48m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7a7d4ec2aa074b83a54ddffe155eb938 : SUCCESS in 47m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0851d11c15d4816a9e13bacfff7256e : SUCCESS in 1h 42m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e10470ef66924b82bec03ab2902b9185 : SUCCESS in 1h 44m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ea2a6aa2fec7049b8248e746dc93f4e55a19ba0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:46:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24f1349930d6363aa26ad9d3dc4b3aa36a4683a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:47:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe5467b3ac64d8eb08aa42b022784cc : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"7cc6c783bf68932c82dce11b731974129a1e4024","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-30 08:58:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 201760ca9ebc545315aceabc6ad90f659da98723","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6cd43d6cc2710266a1990c952201c5714a03393","revisions":{"e6cd43d6cc2710266a1990c952201c5714a03393":{"kind":"REWORK","_number":1,"created":"2019-10-10 01:21:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/687760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/687760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/687760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/687760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/687760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/687760/1"}}},"commit":{"parents":[{"commit":"ca80bc9e03368fa202ef108536c119670dfada93","subject":"Merge \"Bump diskimage-builder minimum to 2.24.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca80bc9e03368fa202ef108536c119670dfada93"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 01:14:28.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 01:19:58.000000000","tz":-240},"subject":"Remove duplicate keys in sample config files","message":"Remove duplicate keys in sample config files\n\nW0109: Duplicate key \u0027protocol\u0027 in dictionary (duplicate-key)\n\nTrivialfix\n\nChange-Id: Ic194a030edd84106217eb5ac02f9b3190d3a7ba6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6cd43d6cc2710266a1990c952201c5714a03393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6cd43d6cc2710266a1990c952201c5714a03393"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
