)]}'
{"id":"openstack%2Foctavia~687936","triplet_id":"openstack%2Foctavia~master~I8a0cdf7ee912d39d9ac24244728094a1d1e15c63","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8a0cdf7ee912d39d9ac24244728094a1d1e15c63","subject":"Do not set log format when \u0027no log\u0027 is set","status":"MERGED","created":"2019-10-10 15:23:42.000000000","updated":"2019-10-23 08:42:58.000000000","submitted":"2019-10-23 08:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687936-1571820056679-34f7ffe6","meta_rev_id":"1d39918dbb7018480191a3f14f959efaa8b739e7","_number":687936,"virtual_id_number":687936,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"date":"2019-10-21 10:56:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-10-23 04:31:59.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 08:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-18 01:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-10-11 16:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-14 04:29:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-10-18 01:15:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-10-14 14:11:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-10-11 16:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 16:34:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-14 04:29:30.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-10-14 14:11:49.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-10-18 01:15:04.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-21 10:56:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-23 08:40:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49b19d76f610e9b745f7cdc72670a1b77e397e48","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-10 15:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"353d864ae9d41451a5926463a1d3ccb6dd70232c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 18:06:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c454b5edb5e6472ba6e6c551dae16f6b : SUCCESS in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64c01aca9afc4e0b85431eb72253db54 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532d36fccf53463395efd2b0f517ef8f : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/204e09c5f454418c9f5dbd7f451bf35c : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a2d73e0140444d3b44a1fbc3cd9fd37 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa52611efefc4f859f1dcef13eb64cd2 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe71ab85cb98489ba96261d43a0d0b6c : SUCCESS in 10m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/021e716ac7c74f23a2c2c7ff762ab64b : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24341731c1f84cf99e2fafa326360ae1 : SUCCESS in 8m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fed5e0a794d44dc2b42477c49b46a97e : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11842c4ac6384882b8d1d3941f1847b7 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9799779b4d9b45c487d68cd1c16b539d : SUCCESS in 42m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/61311b725bb745fdb27e62fc0c281069 : SUCCESS in 45m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/372a5d274012410fb3dc76ae6b7fe7fb : SUCCESS in 1h 49m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b0d7c4dedc3e4839a7424d6bd679817c : SUCCESS in 1h 49m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ef97b64de99e482982cd7129be99e712 : TIMED_OUT in 2h 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b50cda58d1174af9813379527ab9bfea : SUCCESS in 1h 54m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c96cf4445e07496a9d3d7e39d40e3ef1 : SUCCESS in 1h 48m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46c486b6081847e6b7c92f74c09c1e45 : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e137ff37a4ef41d7af889bb4333911df : SUCCESS in 55m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a5f29d0a08864ca6a3c926de279af61f : SUCCESS in 1h 47m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0d25dd04ba14ff3867b07fec3a506ac : SUCCESS in 1h 20m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6b99d8bcfa624cda83f0fe13976db3ac : SUCCESS in 1h 21m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fcd65d6b71df4feda0186729aaf36be0 : POST_FAILURE in 1h 21m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"139ffe11750a5b9571b6249fb3a32835fb18b066","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 16:34:49.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"26632e41a3b8d8a7a12f71fc8ca234d021e20279","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-10-14 04:29:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6302b97692e1f7ae33f1151a1948070e4488b6db","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-10-14 14:11:49.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"a2f6bd1046807eba6170bc452a1786916edb1c19","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 01:15:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d69406db74bedc8bccd8e1a6ebb6daf423f8e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 01:15:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1968ebe42fec63eb6eded00495956a111a348f38","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 01:15:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit d64576e3facebe0811677d7cf9a53a2310c555bc","accounts_in_message":[],"_revision_number":1},{"id":"e726823f74a540b0e60372d6673f678f127f0295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:06:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4b89cf5f13c414b86e545e4c9e32297 : CANCELED in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb724dd52c94b1c8282cae9e550f46e : CANCELED in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35060a3e42854b599970cc22caccd8dd : CANCELED in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9adb9396b0e948c59bdaff18e286bc2e : FAILURE in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93d76181ad2c447c82f2cc8e55b06814 : CANCELED in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da44a478532430993e319fcf84d85a4 : CANCELED in 4m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/282e8748c8c844d9853caac9eadb4440 : CANCELED in 4m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a95f13c6496e471ca2a566b5722f38df : CANCELED in 4m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9363a3feec4c4aa095f965b69e46f668 : CANCELED in 3m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0b937e58a7dc4d37aaffa4cf8f9058be : CANCELED in 3m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1385c7f43a7346d19173e27fb85ad3ef : CANCELED in 3m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d4142627019742e1832c8604a98671e1 : CANCELED in 3m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/977b9f090d7a4c7cbaf30b9ce3cf0baa : CANCELED in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c191ad24678616e4958344ddfebc7c039f27190f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-18 13:12:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d19b1a14239e14681989b3491ec00ec7f0efd229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:36:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b118451888fe49468a6e558530a6a6df : FAILURE in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e403ce59259f4599999e9395f6212dff : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5806c1df1932455881ed07d71487f964 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6a3f113e9cb4a32ba3a092f3f3c0ed2 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69e7e889ea63403d9720741ccc9dcbad : FAILURE in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a897331fc04643a281327a7f4628ead4 : FAILURE in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e602a34d8baa49d8a6b80321c34b8134 : SUCCESS in 10m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d1ef8fc41f214cbf8abd36a30a408f72 : FAILURE in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9e74dc11579d44e1ad2b32a43e0956ac : SUCCESS in 6m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/966c87ccf5024f20ab76873fd0bfdb3a : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/602e13da8d7e4710aa870e2b19b73aa8 : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/884225eb5e104c8db18deafd544d4370 : SUCCESS in 47m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f5545566c9994d1895bf3f752bc61ff4 : SUCCESS in 53m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72fe94a736dc46e789910871c23039fb : SUCCESS in 1h 52m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/01e1f76ce220406caab9c2c2711ca039 : SUCCESS in 1h 42m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0442926a07ff4479a272490a66ad705c : SUCCESS in 2h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed215c7cf8bf4816828a1d6f952d1318 : SUCCESS in 2h 17m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9abc1b55f42e42b1a3fd572f4264c7eb : SUCCESS in 1h 55m 01s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9442dd4e71c14c8a94a80d646884445a : FAILURE in 1h 33m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b722613948714406bf9f7478d789cf81 : SUCCESS in 1h 12m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/125d15019bfb405cb42ba5650715517d : SUCCESS in 1h 22m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e72ee3d0cbc401ea8d87daa93a99b64 : SUCCESS in 1h 32m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d132a8a13a59489f92dab0637de14108 : SUCCESS in 1h 06m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bd423ca5be5e46b8ad26d1e842f69638 : FAILURE in 43m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1063cb9295741dbc228236b0650412fc1a24a065","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-18 23:36:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11458bc15fdbd23d5e4143890365bd0d00104de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 01:57:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9809dd8bd7554aff87682f86237213b3 : FAILURE in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91ee620422684bc9921262b5e8d35461 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ff67dab06342b08c7a0502b636a3f7 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11590579b57b4e8d94ea854e887c5280 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/232db85e754f4cbf957718eab107680d : FAILURE in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f3e4ee410c04c1ebb573bf6931c666b : FAILURE in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1a07df8a3a45cbba4faff753145dfe : SUCCESS in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fd5b338261fa4cf09db994cae923db6e : FAILURE in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cd5d2ee5f9b3474d94e5b0d7503ea73b : SUCCESS in 5m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9d2bf12af8b44a6e9399013e5df971a1 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1eb96a2a2fa47a383c5a3f7dfe60b10 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/283330809e55493e9dac0f23e91717bc : SUCCESS in 46m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cf3535f6757f4357a37d382bd74f8de3 : SUCCESS in 43m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44928f198a974e21aea5fd6a341b6653 : SUCCESS in 1h 46m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5ddafd5b3673456f81187f72b0e67032 : SUCCESS in 1h 46m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d9114241d1bd430eb4d58e6c4b9defcf : SUCCESS in 2h 09m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f4a79b6898a04de08c6379f969761904 : SUCCESS in 1h 59m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d40e4188a1994eb5ad5cb600cfedad9e : SUCCESS in 1h 54m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e5303a1f9e04e64b3a39fc05d586240 : SUCCESS in 1h 27m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2d622cb08f2a44999a29534f0c81aef0 : SUCCESS in 52m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b0dd6ff9c64d413a9413f39d351459e3 : SUCCESS in 1h 38m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/084f98c5a6cd49449d84aa1e0d4eca22 : SUCCESS in 1h 10m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bc63ea1de11b4dca90f1e12017c0a928 : SUCCESS in 1h 04m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/176fdc8755e544c299482602d4e00ba8 : SUCCESS in 1h 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1ed7596686f6b50bb90a1d743e6bc04bef1b208","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-21 10:56:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77a86904f0030fa35cb6f0324255c4045ea5f96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 13:19:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e10aac12842a45c8afbb86f2e95a9ebe : SUCCESS in 7m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5dcb15e5ed842e9ae7fc83eea0126fb : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abcfe908a724aff9a1a46931d1ba96c : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/993a141d348747f7bb0c455f233be7ac : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cffbbfc8d39d44b098977264ced331ea : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93a59a8ffa6b402a8a9a6bd01347012a : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301e94d4cf614828bda483aad42ce4d4 : SUCCESS in 9m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/69a7d32df4c64617b28f7d2138f65dcc : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f839ca279d1d4b359ddd50fdd5de6936 : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4886150d39f645ada892f712115069e3 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c58995014bf4706aa7cb32ff6545f7f : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac8018643dc24ba9953397530e7d67d2 : SUCCESS in 46m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3eb0e2953b4d43f1b1419cc257b0123a : SUCCESS in 49m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db907d8c8c2f40f69d03d97b0f6f5531 : SUCCESS in 1h 43m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/17ac641968b9417486425fb0e67417d6 : SUCCESS in 1h 41m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/506091b15e7a4f23bb94a06c7c5a145b : POST_FAILURE in 2h 17m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/39c954df9221438b81852d727e6162af : SUCCESS in 1h 03m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5adb3338f0d143fca2d2ab806232ed97 : SUCCESS in 1h 53m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ef59911e8b44c62b9665c2351d5555e : SUCCESS in 1h 26m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8d16ab0a33644daacde2a4a73de8442 : SUCCESS in 50m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/035177c1475b40adb7fac2acb4f471d0 : SUCCESS in 1h 33m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49a98507720443b0935122dbf6d87c46 : SUCCESS in 1h 16m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ffe1b78335944aa982e3261cb603e371 : SUCCESS in 1h 10m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b5c0f47647c4da399301abcb411bb36 : FAILURE in 1h 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb4b3b57011d8144106cb2dab86dfa2f9e7aa043","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 13:19:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0187075c2d50bf024bd0d1f5128bd34e9c329f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:46:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a718ddbaaa4b4421a7e4cf6b362e0982 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ea743f0a224ee8b81128ddcf3719e8 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e76b04bee0194840a48d6b207a92decc : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/164f9966b9714d209b456840ab0a456d : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52281c9610e74926a6c4ab0fb3d06d57 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a16f62e7fd14b86b4d3edd44300f8e2 : SUCCESS in 12m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d23607b28934495c91e38405393308e6 : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71af583ff9d04f81b5fe7f3211af9e25 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67bf3a89ffd5428dbff04b3448940729 : SUCCESS in 48m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1bc4e14d53e34f8193d1dfbe61697c72 : SUCCESS in 57m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35a4b280ca0a4c1ca1d6cdc796721538 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2faaa174437349549a088a5e7e9dce6c : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/226b5bebd6f4485eb9efa7171c1195ee : FAILURE in 1h 24m 49s","accounts_in_message":[],"_revision_number":1},{"id":"12fa0b7a52b9e07d330df77e9220ac07916f04f2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-21 22:54:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ad6e73cdfdd73aa4bddeb3f3021950a30a8d1813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 01:21:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/275e2c01230941c28d8e59f11d840c1a : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09673dae53ef4c5d92238700720bc195 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/271f422ada2e47b0b37e63a70f5b9307 : SUCCESS in 9m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4d882cb6e8b492d9448da1d6f122af6 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92363803c4aa452793d62742a14e3169 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9685167005d42d1af75b7d6c2f77e47 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd18da511384166bb0250c45a6ae95c : SUCCESS in 10m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5ce3755b3b6a4f41ae109e0f478d9c1b : SUCCESS in 4m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c09e165789ed4bf7b0eb2c06795ff3f7 : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/868230bce5e94daa9228f77270d1fed1 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a49c66b29d8948768732de94ad6744e1 : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b33138e953254427be7b104de0979b4f : SUCCESS in 40m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/db477283e26d47ab9e297a39c436acb6 : SUCCESS in 53m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20b09cbc687a427a83432a49ed4d3ca7 : SUCCESS in 1h 38m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7ec16394d3264181b27f12dcaddd7b6f : SUCCESS in 1h 35m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ba37eb0da6a4ba3a3b35121e020e3cc : SUCCESS in 2h 21m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c9be5a9cbc548bfb0262f0c85c629ef : SUCCESS in 1h 50m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ca79691b76c46cb82b4e5150eb5592b : SUCCESS in 1h 50m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ded6a6ced2b43d98dffc95cfd6a8b16 : SUCCESS in 1h 13m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/793919858dd54ddfadc5910122778260 : SUCCESS in 48m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e84f71189cc4ffba765cae8ad7e63aa : SUCCESS in 1h 34m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bbd7c75f5734d88a3299d10ab82d8b0 : SUCCESS in 1h 02m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/404d24aba5544b7799cd72957152e632 : SUCCESS in 1h 00m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/87d10e07f18f4d91a116846d47262714 : FAILURE in 1h 07m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebe36d6c62117e7464b49a585d325f6efcff5984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 01:21:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3129622df11645f833c434ed7dc0381e2e850c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 03:29:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afaaff5b93074e7a9076ed11ba374e11 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c94fb373a9141dba0eff45d1d7d8c11 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dc8a3575167442c941a0b54e61ca4b4 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c051bbac2ca54f14b3e83b0617db7ce4 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86003b328dc149eb933ddf47fd567136 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392ab8ac858c4faa9b0a69b168fe3cba : SUCCESS in 9m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9a79af3986124c9a9b3d7f8507147a88 : SUCCESS in 4m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f964002e690466ba48f052e45b298d9 : SUCCESS in 5m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0637d9d810cd4c93a95d2b85e46c6591 : SUCCESS in 42m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0b180220dda343379845b6bcbab5e0d6 : SUCCESS in 40m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99de5baed11448139e6bc335c78e2947 : SUCCESS in 1h 36m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/98f6742cbb914fbd9eafd7e15cbba58c : FAILURE in 1h 40m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9cd90173f8e84b819a6854e2c5ba254d : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bd6378e7d3127dfa27080f43c7ba3753c9c25b2c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-23 04:31:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dfe51d2bd8324759955815fc4ca90fb29823fa7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 06:56:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd0923fe75d448e6b2d34384de0c13a3 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9802a1ed2394e169e1ab4f5f120e008 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1357b7035eec48aa81ef8bf6a74b674c : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47636bb56d4f493e8f5dae66b6769127 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e97213bcf3944a06a56cff0c7a12178b : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbb1de2813484e5da7bd2fd4bed93dae : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a32858273c84fd28965e47c818849fd : SUCCESS in 9m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5022ec5cb0a4b1b8ae46581e6892865 : SUCCESS in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7d0ba6d50cb54558988cd740e25036fd : SUCCESS in 6m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5b9d47c81543476ea86d10e02a30ac5a : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5393fec2af54c24a35b9b71236b7e2e : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fabd57745e242729a21642562d29ae0 : SUCCESS in 43m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9cfd03f93b014659bfc3b4e0412afdce : SUCCESS in 45m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63c02724b3e54b15852825d8aac28678 : SUCCESS in 1h 43m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f13855ba169346d2812d5722c0d12582 : SUCCESS in 1h 45m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c4d07048029c4e7088ba8d04dfe8fdf1 : SUCCESS in 2h 05m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53eaab96314141228943c208bf3b55de : SUCCESS in 2h 23m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/62c9b278a481444dbbdb7575798fd3e5 : SUCCESS in 1h 56m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b9ec909a2e04d72bfd0dd5a8ad986b8 : SUCCESS in 1h 15m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/889670b2f08d43e5a2b829e6e5b6ad08 : SUCCESS in 55m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/91bddf9607a94640910d47448f0d5f44 : SUCCESS in 1h 29m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1758def8a3f446ffacf28ebd847b7e50 : SUCCESS in 1h 16m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3374f6468dff4e5abc4358e8ef16af2a : SUCCESS in 1h 13m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/acf36e8d60ac4d1cada81aaba0e02074 : SUCCESS in 1h 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd2ec0d1c552419736dbeb0835329e691d6bc3ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 06:57:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae917d3cd99ec826b5b81ddb2c72686323f3a1d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:40:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdae1d256a7a4ea585d98db217cfbecc : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debf79d27cb34a83a9627b251643846b : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/909995f26c7144f89601de3025124454 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc254d60eaba471fbbbf8118445aa98c : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1635d3416e3a425a872625d438f32602 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6edbc9a393d34cee975445c6f0373f84 : SUCCESS in 10m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e5e66905a614cf3959f1347be3cc0d6 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0978d9c1f6354bf7882dfd00d7046590 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/739ee365fd0a46be8a485fd481afcfdc : SUCCESS in 48m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/71445cb7d33845f4ad039433b69982c4 : SUCCESS in 47m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7973938cd3864850949d2eece1806035 : SUCCESS in 1h 32m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e060288054c44901858f299314193f7d : SUCCESS in 1h 41m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70958e8f59b64fe5af473d96f805d61a : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f71169c16cff42d2dd0af268a057b75183a74719","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:40:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d39918dbb7018480191a3f14f959efaa8b739e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:42:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96dd6895a054bd9bd1b97feabd63628 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9b2bcd04edef6f09eb64427a7d7d05eb5153d82","revisions":{"e9b2bcd04edef6f09eb64427a7d7d05eb5153d82":{"kind":"REWORK","_number":1,"created":"2019-10-10 15:23:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/36/687936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/687936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/687936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/687936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/687936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/687936/1"}}},"commit":{"parents":[{"commit":"b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c2cd7b4c835c391cfedf12cf9f9ff8a0aabd17"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 15:04:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 15:12:55.000000000","tz":-240},"subject":"Do not set log format when \u0027no log\u0027 is set","message":"Do not set log format when \u0027no log\u0027 is set\n\nRe-factored code to not insert a log-format line in haproxy\nconfig when \u0027no log\u0027 is set in the template. Adjusted tests\nas necessary.\n\nStory: 2006181\nTask: 35706\n\nChange-Id: I8a0cdf7ee912d39d9ac24244728094a1d1e15c63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9b2bcd04edef6f09eb64427a7d7d05eb5153d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9b2bcd04edef6f09eb64427a7d7d05eb5153d82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
