)]}'
{"id":"openstack%2Foctavia~687963","triplet_id":"openstack%2Foctavia~master~Id0bf1c6b1633ffb4143c7628b722434faf433d7d","project":"openstack/octavia","branch":"master","topic":"bandit","hashtags":[],"change_id":"Id0bf1c6b1633ffb4143c7628b722434faf433d7d","subject":"Use bandit block in tox.ini","status":"MERGED","created":"2019-10-10 18:03:49.000000000","updated":"2019-10-14 22:51:51.000000000","submitted":"2019-10-14 22:50:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687963-1571093427759-fe29816b","meta_rev_id":"6830ef4d84a43b85f7293dea5a7a8c3031dc0f94","_number":687963,"virtual_id_number":687963,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 22:50:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-12 09:37:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-12 01:42:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-10-11 16:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-10-12 09:37:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-10-12 09:37:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 16:32:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-12 01:42:46.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-10-12 09:37:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-14 22:50:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a3019f17b3b6d874cc151f9f99e8528ee8200a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-10 18:03:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7871e0137412c322940030a8aa61733036065a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:03:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/92c4098f69f14582ad2fb6d21bd3098d : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f00db1fb1004dd1873bb595bb90e7f2 : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a049c50c58042098b21ac12e3e1ae70 : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ac20ddd407d4e9a91e28e0890010246 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c86e6fa4a64053b5242c711d37b757 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71ebbfba06aa4f6e8118e5000102fe7b : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f312851db534ac7abe28de9efc2a3d3 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1d198853859411198eb2f9bdaf85574 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1848d51fcb4a42e3a19557cf74f67689 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e7bf8c348a40ba933491bb2e78c619 : SUCCESS in 8m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9902e44e0b594a6ca62378dab8773082 : SUCCESS in 7m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8970d1cc8dc847068ceb6a22b18a55b7 : SUCCESS in 6m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9e15869376784dbb985cdd376e78594f : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e434e2633674dea89d9eed4e459aff0 : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef05184629bc415d992ab1d2c509266f : SUCCESS in 46m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f567ba56426b4245bf8dfe54f6bab512 : SUCCESS in 41m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14d2c76f2608407db6a57909644fd252 : SUCCESS in 1h 56m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/edd328da66a143ae985d2d6b49024785 : SUCCESS in 1h 45m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cb4f252164a7468b8afc7de5b6274b9c : SUCCESS in 1h 54m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d3586f211d248d19e4e18628b534ab7 : SUCCESS in 1h 52m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8dbab9b86e0341f5ae1c7b5214634dae : SUCCESS in 1h 11m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19a667c36d474b2f8a49cb702f73cd2c : SUCCESS in 55m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df4469a023894ad38513583e7fb2d56f : SUCCESS in 1h 34m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e78ade242250423cae8cfee3451ec0e8 : SUCCESS in 1h 08m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/be01a90aa14e4b57ab88b17e84bb8df5 : FAILURE in 1h 02m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/874a8223730f496091f4ea9b6ad7ec32 : SUCCESS in 1h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd24d4597331f47fd0580903ab7f245fdef5a46a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 16:32:18.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f1e39a3bcc7897e05caa2fc5ea9eb99d90be2e38","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-12 01:41:01.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"278574efe59921126409f8c189cd200964bbe842","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-12 01:42:46.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fad532c48179f4baf341ce03d8f95b91c3a0a3a2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 09:37:44.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ffe3f90918dd9407a74c667c3e63c4853a44e81c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 09:37:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7f93344cfafd2a9aa31354e741cc0616f7c0865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 11:20:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eaadaa1049184507b8bd9623ccc64f8c : SUCCESS in 4m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b58ebce757646e4a5dbae70b34b29ac : SUCCESS in 1m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e13dce9c2bf24fda96d9c9ee88187c01 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f01a7f83eee429dbea914c54407184d : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab9327643b5b4fe5b1f5b9538bd47f3d : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f35f91f6bef497790c593e2eb3615ae : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1dd10fbb5204ced8547c0a9bae92e86 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0cce7e16e04f3ab2c41d5ad5d8ab9a : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c915ce04c3b645eba19744114066d359 : SUCCESS in 6m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be302e82b70c49bea39f78670dd3bb4f : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4fdaac3c01c4869bf790f23198924cc : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0242f01397594510a1a659f7f132dbc4 : FAILURE in 26m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bebacb3398a741d5a06b959ec604f372 : CANCELED in 26m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/abca0ef5edd045acabda0553dc0b70f8 : CANCELED in 26m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e72fa8dc575b4adbb060571bf2d61534 : CANCELED in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c5ef0813eba63d89e6db644845a48d1659055489","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-14 18:28:04.000000000","message":"Patch Set 1:\n\nrecheck\nglance failure","accounts_in_message":[],"_revision_number":1},{"id":"106a3b4e8ecfbe238e379491754a8e5bcad7872f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d31b8636cc084c40bfda0b1de8adce8a : SUCCESS in 6m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/23be79e727334975bc07334fc777a269 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1f52f93f2b740de8e9f7c7e805cddc8 : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b61056dcd2af49319002320c78431397 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abb6d5b9093f486892fc8fe431e5e136 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85c7a6d85c714d17b3fd425dfd8d0e84 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afa2e99afde84035a5ac833df95197ea : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe4e70fcac1e4a6ba001dfb25ccfc5df : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbbeed4c5ad42e9b98cfe51ba8546ae : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c566c73647fb44d29b1b15075571592d : SUCCESS in 6m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7942340f4e514c9c97f86554636a7e14 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d2e12c3661494bc59ce2010e475af665 : SUCCESS in 8m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e2785c183baa4d9380ca1c5f988e7888 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e621e9653836434a8fab002fc3a85e96 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f65fa67839a14bb399ad9e1d3724ea6d : SUCCESS in 45m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4def5098283049b4aa5016e8c2f6487e : SUCCESS in 45m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4c989d4ae8b406f84250cdf1719ae8d : SUCCESS in 1h 49m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8d35ed19b5e64834b7b79fc28c80011c : SUCCESS in 1h 51m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/90b5d00ff5a34f2e98d7f19989757f0d : SUCCESS in 2h 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a305e864e20d4dfca9e2a345d4c4ceac : SUCCESS in 1h 28m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ec89d2430664f6d9f1cfefbde62f36e : SUCCESS in 2h 17m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef5c0bd354ce423481ebe3607fb5c2af : SUCCESS in 58m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64fdf203e79a4fde902af20452c91e52 : SUCCESS in 1h 42m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10c0ac3c94f44116a472185f23208e81 : SUCCESS in 1h 09m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b3e10e74ee88423cbe0f2674104bea0b : SUCCESS in 1h 13m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c4ca798dcae04a1baa9413610f271676 : SUCCESS in 1h 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf2e9064ee924c8d481727e3667fcbb03080f1c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:05:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b29d3f0c5cdaae9eafcea3ed49c5657e0d1392f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:50:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/65aad7263bdc41ce9e9c17c00448292c : SUCCESS in 6m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f14fa942c6247c5b2c062daa2b9abac : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72866df0d0ca4e6e80f29c233cad37bc : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272a65a4a591454d9418a20c2f4cdaaa : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a230b21797574f8a970314a592b86995 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6643a5efc6184483ac5431d7fe7a41c6 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63e9ab3e7382424fa11eab0207bc86fe : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dcd86b5cd204d589fa1d93fdf9f00c4 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c73a1d4973d49deb29e2afe68971b7e : SUCCESS in 7m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3cf1d005d92845f9b99db64afdf7b641 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/174d51cfce7843b1853085c0c52ec20b : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d7a2d974f3b4e4b86abff723a98af6f : SUCCESS in 46m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7537d3fddbd0410db4364114adcf691d : SUCCESS in 47m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12cc837ada8046c589fa362641f6a810 : SUCCESS in 1h 23m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b67e65b013d3416b9b1b645f051054e1 : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":1},{"id":"11345b05ccb2d122b649c714a069a3dfc6dcc14c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:50:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"40fd25f5e7a8b941f247a9bea5ce0070e9e6811e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:51:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/365c5468748d43dc99e40ffce780cbc2 : SUCCESS in 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5748a872f0c41e5860241207eac49b1 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d6bd45d928f40eaa8c175fd0d9ab204 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65905cdb998aa9a74d364b2b46da4304c19beb07","revisions":{"65905cdb998aa9a74d364b2b46da4304c19beb07":{"kind":"REWORK","_number":1,"created":"2019-10-10 18:03:49.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/687963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/687963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/687963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/687963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/687963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/687963/1"}}},"commit":{"parents":[{"commit":"98448dce44a45e03ebea471c2d1b4398b0aca301","subject":"Add bash script style checker to pep8 check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98448dce44a45e03ebea471c2d1b4398b0aca301"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 18:01:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 18:01:40.000000000","tz":-240},"subject":"Use bandit block in tox.ini","message":"Use bandit block in tox.ini\n\nUse the bandit testenv block in tox.ini instead of inlining\nit. Also changed the call back to \u0027-x tests\u0027 since that is\ncorrect syntax, it was just broken in version 1.6.0, which\nis now in the blacklist.\n\nChange-Id: Id0bf1c6b1633ffb4143c7628b722434faf433d7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65905cdb998aa9a74d364b2b46da4304c19beb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65905cdb998aa9a74d364b2b46da4304c19beb07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
