)]}'
{"id":"openstack%2Foctavia~687978","triplet_id":"openstack%2Foctavia~master~I31bdb2956ae37d1116069c2b37656ce2ee3c2dd5","project":"openstack/octavia","branch":"master","topic":"tox-cleanup","hashtags":[],"change_id":"I31bdb2956ae37d1116069c2b37656ce2ee3c2dd5","subject":"Clean up requirements.txt from tox.ini","status":"MERGED","created":"2019-10-10 20:17:02.000000000","updated":"2019-10-18 20:55:45.000000000","submitted":"2019-10-14 13:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687978-1571060619916-9e252c24","meta_rev_id":"f9954745c25913b6829ba3abe9bf8d685776e8ee","_number":687978,"virtual_id_number":687978,"owner":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-10-12 09:30:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 13:43:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-11 05:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-13 14:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-10-14 09:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-11 16:42:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-11 16:42:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 05:07:33.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-11 16:42:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-12 09:30:00.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-13 14:36:11.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-14 09:32:51.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-10-14 13:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7cf732668f63240819960af007dccbac44e47fe1","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2019-10-10 20:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22afa5cd3c60558b74d723275ef01a22a8f338ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 22:44:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6bdb28b7cb96482386f8376e1d700d0b : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0f9be6189c949fdb7d7557b52c5e7ca : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccb39d2f79224419871ea1a5e9be919b : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a394a803dfc6468ba2c625590998ef01 : SUCCESS in 8m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/023e7c3335164ed18f6f82382d5c1bf9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d8179c8edce4220bde72a2f421a491e : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fd4dca2393047259b5bea9a6b5a5882 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece198ac43b143d5b22fdf5ae80490b9 : SUCCESS in 14m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e68290865744c45bc5b41a788f9f937 : SUCCESS in 6m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f6f6d28ddda7480bbca41f79dfa57e24 : SUCCESS in 8m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0a16ebd6e86b4cef855bb0896b207c15 : SUCCESS in 8m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f39c3c4f84f346ec94d30db1b5a80129 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/557beab705b245fe8c53aabf0f859b54 : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9dbcdccf18ca46cb9058963ef419140a : SUCCESS in 45m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1914ea43c106438dbd2309528b7b3be2 : SUCCESS in 44m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e54cba913e84ab2aac582e9ba017838 : SUCCESS in 1h 58m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8ac52e0378dc426f948a121ddf1a8204 : SUCCESS in 1h 40m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1599193fd90c4477bece7c788e3bd58c : SUCCESS in 2h 22m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c4b8999d77464502b7904400aa018c69 : SUCCESS in 2h 06m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b08a0771b244ec0b2a8fe34f6e62383 : SUCCESS in 1h 46m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7338e4a9f22b4b32851af96bfb071876 : SUCCESS in 1h 08m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f2bf041a1b84efba4d8fcd935f4d1ff : SUCCESS in 1h 32m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5c6cf1c40d04401a32626ac1f1d2b1f : SUCCESS in 1h 11m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/eb25722282fb477d8a2f37f1ab2b8645 : SUCCESS in 1h 04m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e999b1d07f4f46a5947a8f4fa39a3986 : FAILURE in 1h 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9281e00d2bcfa76d836ba07f0937a17c88a73bd0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-11 01:41:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fabb5d19521133ea9663428d4ee1d60d6342b93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 04:28:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8d0c044569f944ba98e7241fb5beabe8 : SUCCESS in 4m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec2bb31787ec438b87b75cd44296b876 : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ffe9f9205e14be0b606b5c1199e052d : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24cceb7b62141119b29d9044a1b9e7a : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36da32956a1249818fddf20e04334c47 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/729508e11b714380aa78d56f3c6cccc0 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16ce891ab061430c8991b7ec76ad569c : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6693ab2f0140463fb9df0cf55738e918 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53b00e5d5d1d411781ef8876aacb641a : SUCCESS in 5m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5b9b4b278cc64c1ea4ef69c54e1fa2a6 : SUCCESS in 4m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c60f3b47e5d348b4a34f3640765c7741 : SUCCESS in 6m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/28fedbaa96cd48768de4cb72cac5936a : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a317f6ceda9d4bd38836943757f1c607 : SUCCESS in 5m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08c5a8444d9641658fcf084b9a52b74e : SUCCESS in 45m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/80aab46e1245489183be5a43b73ada7d : SUCCESS in 37m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8530bf793c9b44f79380cc6ba7a841f4 : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/90cc31cd1f22460f83a31ae0a27b8ed3 : SUCCESS in 1h 45m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/031805b83f894fd3a7285fc0d2602694 : SUCCESS in 2h 13m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74bc1543fa894abfbc4de9b2ac7e23d1 : TIMED_OUT in 2h 33m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/88c66b2e4c6448bca7d55d2db2e77526 : FAILURE in 1h 27m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68864bc035ac47308657448faea97959 : SUCCESS in 51m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a130a9ece91840a99a17638a8f68356d : SUCCESS in 1h 34m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/157fbd543e8447eda0c284491336db6e : SUCCESS in 1h 04m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4d894af145474a3fa7ff036f3f3c8a79 : SUCCESS in 1h 10m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3d421a57e8224ddba19b1cf1637510b4 : FAILURE in 1h 07m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8eb55ea32ce8537d00005a077db27f798cea538","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-11 05:07:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3939683d18b71dc8215caa28ca58316872b589ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 16:42:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"69974c650123c23d7368305e7fc6091cc722de85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 16:42:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3899c2ef9029c716fae5fc03ff7d9fa36d331326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 18:33:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3bea25795f0f427faae0c58864d93103 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9eed9541344a4bc7bde31d4c51789cc5 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36727c1e53a042ec956b23295143748a : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1014863ffff14fdd93d7fc4e45e14259 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fc1c77719c5413a9a95fd49b2765174 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb88a011ef274931829b43318a3c7c48 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fba6feaf2d4988894de5fca4b30a5f : SUCCESS in 20m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1995a9aaa4a241439437d94d50aac75f : SUCCESS in 13m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f6a86caa36ad400d9c79487cd2b204f1 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae5991847a904f17b914532460fa4f33 : SUCCESS in 5m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8de97985c51d420c861deb7e3dc973c9 : SUCCESS in 57m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f2c3daff9f08454599a45cfb53d4f7c1 : SUCCESS in 40m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40da9c379ede42a1b75d190e8ab88201 : SUCCESS in 1h 43m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/899353513fac4c2fb65408cab0ffee25 : FAILURE in 1h 47m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4b19de8aa10267adafbeaf8f2971150489e698e9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-11 18:45:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1cf86c8f164dff5c6d69a68e430f3354df3aa43f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:49:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/40fd52fabf2443918152535d5ea88a57 : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc8cc3627a04ba6bc06d2d7006a9a3a : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83f79a8bb8e04c45ab09cc3f98021169 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6069c20b41f4f5f8279fc6473c3c225 : SUCCESS in 8m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a776ca16fed24af4bdd4d00996ba0cab : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b15677826504bee8a01d811be403aed : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26d953ade7e446108a3083940f2f4e0c : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23aa7b79c4b44c79fcf8fc887b70139 : SUCCESS in 15m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e84efcdabef944459b59017303e0c233 : SUCCESS in 6m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/43e727f1074d43a893fb1036582994de : SUCCESS in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3f455c4e5047476fa781b0f17d5df556 : SUCCESS in 7m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5c1d83be43547cab08e4c15f0b61ba2 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1ee7d4653a84fc499745e3c456a952a : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6bae9fd64f4f4ec09f6b191dd8b3dd92 : SUCCESS in 41m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/09ccf352f12e4be08ac8a8f2313e321e : SUCCESS in 48m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a414da6f82e64a9db5f0e796262507c9 : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1e949aeaffd84255911c9a97ca4c9b7e : SUCCESS in 1h 47m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/39b19cbbc968452890211c6d121c7547 : SUCCESS in 1h 58m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db3334e6c3c74da9b356bbaa9ea5f438 : SUCCESS in 1h 57m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/01cb8e3228b94c3db97f282f568090f5 : SUCCESS in 1h 17m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/607a3264b44d4b6bbda2f946e71feead : SUCCESS in 57m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b62f224ad0794f71a1b4cdde31fa6edd : SUCCESS in 1h 36m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d696f843bd9e48be991b367d0e662469 : SUCCESS in 1h 12m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e96f8dfe495d4ae5b9c66b0f5a07bc66 : SUCCESS in 1h 06m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a089d687da1642d183118a51c3b008b1 : SUCCESS in 1h 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69b536c0521e0ec4142d4fe50631a57ae70d2d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:49:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"317de1dbdfacb4e5ebf8b3dcc3468de3ed91b97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 21:00:33.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"06ce4cff02a66646449a6efb4fca6b17fbe8b385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 01:14:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/417ae730c31649098fb4a9116146e129 : SUCCESS in 13m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1853395b94c84507ad26e7b9a938e10f : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b7773766a34237acdbeebc816e9683 : SUCCESS in 9m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d674204e791451c86a4b7a9056ce174 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d21014e899146d990851bc495df384d : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b74dfe0b8b4b49a1b960ab92056a3515 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/223fb90602954b3c93d1aa87204c9ec2 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b71bc85ca24404e811ec8bbda8f40f7 : SUCCESS in 5m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6669ad3272cb42279346a6fa95b9a506 : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab4180d3b2834560af4bfd06069a6930 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/12fbcce63ade4304a0b0da3e7b30eda3 : SUCCESS in 44m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6e8e3e06c8504479b6c845e0f7069604 : SUCCESS in 58m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27dd787921b344d28b983339ca77c2d8 : SUCCESS in 1h 41m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57ad7324000241d1a8d9cbc34b9d06f9 : FAILURE in 1h 53m 37s","accounts_in_message":[],"_revision_number":2},{"id":"577b1fdfd28a13e32dde29e35a487a1d7c79fb6b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 09:30:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8ff8bd56ccda2ddb795fa08f3712f7a8e1a4e19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 11:36:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c0346228b4cf4b32a77ded2b8c4ebcf3 : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d824caaf06bd41888a9b934648d05485 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78c5b148838f4537811d736d940b6ef7 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68dc83daf8b54e99b6f4581d2fac4c8b : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bfbbd3549cd4e4ba99bb076742923fc : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38f37cb61eee4feb80bfe252d8bca9b8 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dde7cfa891d4c3489526864d9c1f6ec : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea5bbfeb73464afa9bb04b8b1d80370e : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faecbc1d993e413baa4fcc7f15665960 : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/50fb3be709ae4e7d83be401f331c6392 : SUCCESS in 4m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/81496b3a3c6f4947b394db6278b6d40b : SUCCESS in 6m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7b445632af3a4dcbaa1720c783c212f8 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2dc31c60d744a2b9c9fc251e9b42cf7 : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d80c6ee23164a7687cdf289b3da397f : SUCCESS in 37m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/905cb1759f8d4a12b1317106555bbbae : SUCCESS in 40m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cb348bcdb804b62b587537563bec6a8 : FAILURE in 1h 32m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/77188d6ad49d44f1bc9de478fd2f995f : SUCCESS in 1h 42m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/df67d6e76d73455bbd264c722afd9bb7 : SUCCESS in 2h 01m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22c77b8f476744e597adf41f9e688a97 : SUCCESS in 1h 53m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/253161bb0ced4be6adb32445d9f3660e : SUCCESS in 1h 52m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e11d1dd3870848e6b3fa92918853d435 : SUCCESS in 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/88e28f981a0540e1b6e81a1ac19d459d : SUCCESS in 1h 22m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f2efbb28be3477baf9b207add02d282 : SUCCESS in 1h 02m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/782ad8a17073495aa6ba4ac59f921209 : SUCCESS in 1h 03m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/37b28a1de70e427ea07d13da5829e920 : SUCCESS in 1h 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2c3cbb43f236954934e70340f66d8f09e4c7a74","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-10-13 14:36:11.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e3010cec6ecce73de1760216ada52cc479ed4df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 17:17:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/26acad8afae54d45948ae6810ddc01dd : SUCCESS in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a838e0d389bc42ceb43cee3da449ff54 : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef1e5c7a3d97472fa146d787bd918581 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb048078258b422eb1ef9c69ea0c4e6f : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38224201139d4cfd970a982063e7581b : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c848ad91c17d4ba8888877e9b08c12b2 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ebfd2d312a04e04b4467fd729196b31 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b9167eb41843809e59b95f365ecbe6 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed3b6df7219d474b9d56da57dd5bab61 : SUCCESS in 5m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7006f7ce6c1e48b78970905681e58906 : SUCCESS in 3m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/feb18487f10a4edfb45119601c9c4d45 : SUCCESS in 5m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/db38c393cfc8413db6782d409d550c10 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70a2f7bc9f57436d97a9b6964fd6434f : SUCCESS in 4m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b40a652ccf044b41938b0c3ac6d181fc : SUCCESS in 50m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7f2434dd3a8b4f2cac30945933e692d1 : SUCCESS in 37m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d02b0e3bc7ef4f51a49d802194991513 : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f59d59d680b0423e8de3ea769f7c6611 : FAILURE in 1h 29m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/47a966f3edf046ce81af529673df4f5b : FAILURE in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad5bc76c63724e928e0eb1179fec8c73 : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbbe70c332704bbfb4201f6203debbef : TIMED_OUT in 2h 33m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84041aff10d24e2496db4ed86f643a4a : SUCCESS in 1h 04m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fec1f6bbe23a4161913f2f2641e1eccb : FAILURE in 28m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fa3f3878209419d9754a54c2fb49945 : FAILURE in 1h 07m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/92bfc145ff194ccfba6c16a370510ccd : FAILURE in 1h 04m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a86768a9d7b2443eb7e3abd2c564797a : FAILURE in 1h 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15ce3c2ae56bd2f23dc2465901354bc36bd939e5","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-10-14 09:32:51.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0b9db945444f3eb95d00b15ec5f5733e238f7cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 11:53:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebf3583726044b7fb9422ff152554bb4 : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df777bb8db0d4f8f89b64ce572632e93 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9976857b11b44716a32bc1b8c042dc63 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b8feb81e934d14b8ea9cb61dea0af6 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ff78d66fe33421b8b362c4c38f4197a : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1e336069e79445895492cc864d94e96 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54c4de7902254d85837fd1f6478fa9b2 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725e1dd40e5b48cf841858d74736c315 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b0425c87c7e42aeb57cf72ea439ea39 : SUCCESS in 5m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c4cbebc8d9df4914ae52e3dc045475f7 : SUCCESS in 4m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eb84f79aba6d4a0b99550ca7512a945e : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7e00ed9fcd2e494990c3b882a215aca1 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8c9b21c4a1b24bfeb2c149b7f4385644 : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85a6adecc7384bf98a25e55f105440c4 : SUCCESS in 49m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/82bd075fa03943d783e243edaccd52e6 : SUCCESS in 44m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fae45c2d7d2e4e07a535c13273a486a8 : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1810e4ceff694f4a843f349e845f74c0 : SUCCESS in 1h 35m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5ec5458a98464449abda5893b74bcdb7 : SUCCESS in 2h 16m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc7390b4e4ff49f8ad3a1380dc47cdb4 : SUCCESS in 1h 21m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2615a1d066841b597efbefdab38e6a5 : SUCCESS in 1h 26m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0cc8331391374649b8b689438bec0ffc : SUCCESS in 48m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9693b52330514f32bd5f98301834dff1 : SUCCESS in 1h 38m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d79f9e310a1e4a80b2ade168113716c2 : FAILURE in 1h 43m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b0fc27f464e64448a9e81e6a7907388e : SUCCESS in 1h 07m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d986f601b987489c9f65bdfda5d94153 : FAILURE in 1h 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c05cc3f85c3b284c12e63da1cfcc30821e6b8da2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 11:54:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"114a576d0d8839aae938bd199ed8a32dd23d1123","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 13:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9dd138793cf583c705507afb90d22bb37fab3a49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 13:43:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9772f7d160954a31bf39ad4104a90638 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd16cfac2b89463180000284d543c1cd : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/990e3b7e7a3c4c649faf65e4e5dd95bd : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24a64d25beda472fb0fb1e9882aa4762 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff056485a70741abb4d761a359ac2c05 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbc2d8086b304432af8816934391f493 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95bb6b4dacea456d8eff41788d32075f : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8881d51f9d1c46289244ab598640b693 : SUCCESS in 6m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dbae83e0dd5342d2a2109379237e288f : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a10584398471410686382b7aec658e47 : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0171c1ebc51e4ce1b73071ed57853299 : SUCCESS in 42m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/428838bbc17e40cba25006a6ba6d7712 : SUCCESS in 47m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52e0c4e3427e490a9bf70a2d96a81b4e : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d82ea608c3c4b67a53bf2640dd13763 : SUCCESS in 1h 43m 18s","accounts_in_message":[],"_revision_number":2},{"id":"eb77c659c94567b91567b71ff2374258a3e15cb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 13:45:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d42062eb1b284e708742e481944b035c : SUCCESS in 1m 14s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6551106cc37b42e49fe7d95ea1f62e6d : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2303defb82041e496c81c7a5de03667 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2},{"id":"2bfa85062af43a1e293b0018b0dfb15fc489389c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-18 20:55:45.000000000","message":"Created a revert of this change as I4941df3894148482c597d30f3a8db70659fd1b7a","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b8e6de1b8230e5d31ca36624e7fdfba25e3e077","revisions":{"4fe93511820410d9c78588d3bc0e8384f834ff03":{"kind":"REWORK","_number":1,"created":"2019-10-10 20:17:02.000000000","uploader":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"ref":"refs/changes/78/687978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/687978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/687978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/687978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/687978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/687978/1"}}},"commit":{"parents":[{"commit":"e6cd43d6cc2710266a1990c952201c5714a03393","subject":"Remove duplicate keys in sample config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6cd43d6cc2710266a1990c952201c5714a03393"}]}],"author":{"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","date":"2019-10-10 20:12:51.000000000","tz":-240},"committer":{"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","date":"2019-10-10 20:17:00.000000000","tz":-240},"subject":"Clean up requirements.txt from tox.ini","message":"Clean up requirements.txt from tox.ini\n\nSince the base testenv tox target sets `usedevelop\u003dTrue`, tox will\ninstall the application (Octavia) into the virtualenv as well. Since\ninstalling Octavia will install everything in requirements.txt, we don\u0027t\nneed to specify it again in tox.ini.\n\nChange-Id: I31bdb2956ae37d1116069c2b37656ce2ee3c2dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4fe93511820410d9c78588d3bc0e8384f834ff03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4fe93511820410d9c78588d3bc0e8384f834ff03"}]},"branch":"refs/heads/master"},"7b8e6de1b8230e5d31ca36624e7fdfba25e3e077":{"kind":"REWORK","_number":2,"created":"2019-10-11 01:41:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/78/687978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/687978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/687978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/687978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/687978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/687978/2"}}},"commit":{"parents":[{"commit":"e6cd43d6cc2710266a1990c952201c5714a03393","subject":"Remove duplicate keys in sample config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6cd43d6cc2710266a1990c952201c5714a03393"}]}],"author":{"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","date":"2019-10-10 20:12:51.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-11 01:40:58.000000000","tz":-420},"subject":"Clean up requirements.txt from tox.ini","message":"Clean up requirements.txt from tox.ini\n\nSince the base testenv tox target sets `usedevelop\u003dTrue`, tox will\ninstall the application (Octavia) into the virtualenv as well. Since\ninstalling Octavia will install everything in requirements.txt, we don\u0027t\nneed to specify it again in tox.ini.\n\nChange-Id: I31bdb2956ae37d1116069c2b37656ce2ee3c2dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b8e6de1b8230e5d31ca36624e7fdfba25e3e077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b8e6de1b8230e5d31ca36624e7fdfba25e3e077"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
