)]}'
{"id":"openstack%2Foctavia~688180","triplet_id":"openstack%2Foctavia~master~I824c3ed541cfd42e3c8f02be5da694f42f40d7c2","project":"openstack/octavia","branch":"master","topic":"switch-to-ussuri-jobs","hashtags":[],"change_id":"I824c3ed541cfd42e3c8f02be5da694f42f40d7c2","subject":"Switch to openstack-python3-ussuri-jobs","status":"MERGED","created":"2019-10-11 17:28:51.000000000","updated":"2019-10-12 12:43:43.000000000","submitted":"2019-10-12 12:40:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688180-1570884043707-47151255","meta_rev_id":"6028183fecb515c3db15ec7113063744df074e56","_number":688180,"virtual_id_number":688180,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-12 12:40:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-12 09:54:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-12 08:59:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-10-11 17:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-10-12 09:54:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 17:39:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-12 08:59:50.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-12 09:54:17.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-12 12:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c120b8787daebf54615eb62095a6cee507cdb25c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39b55e810ce9cb4fb0b2e862b6d7a34ec9928f79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 17:39:32.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"bfd6c080ef1d4b50cdfae08a234abadbcb85315a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 19:43:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55f9f6281c774eb384855f26f394e234 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52f24f07f0b3489cabd705096e3e682d : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a126e1599f32488fab22047c5d38ad51 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e86b38bc68314187abef6d815854a415 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c198ad07f620481dbd5ba359b5f8538b : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a51dad92a9442dc8e060da69e11fba7 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f310aaf680e54bf4858217455dcc15a3 : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4c378e5806d24d95985139c27ee5810d : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb579b494f9c4876931ca14202bfb249 : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/46895828c8b8487db8eb21aee5e14148 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65ce6ac434a54a64b4beda6b1fb932cf : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2534fe962a1f4a1b871417d1e7e3eff3 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4de3c10d3ac54665a3f9b850391013ca : SUCCESS in 42m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/185d61241d224d069d035f8bf5b34bd9 : SUCCESS in 1h 43m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/67f90500c7cd451492c30098790faf9f : SUCCESS in 1h 59m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eb4ab73d5a3b475eb3e5157d4146cdae : FAILURE in 2h 08m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f84e921e7a145cfb70c61c18196c648 : SUCCESS in 1h 56m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d250b3e1b774c9c8ca019032e4ab878 : SUCCESS in 1h 07m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bfc6e3c0eb984e47b15fd3f01475536c : SUCCESS in 1h 23m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29854122175d45eea2530eacd38e6d39 : SUCCESS in 1h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f4d4843f0fd4bc0aefb8a0b9892594d : SUCCESS in 1h 42m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e655f5bec49b4eaa8c75697d6a27a1cd : SUCCESS in 1h 08m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ea477d97b7334c6c9179bcffd555dcd9 : FAILURE in 59m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d83f115c70474da3945b5405181c1dc1 : SUCCESS in 1h 25m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83b0f4f428fa8806d2d38a0af8fc2657c6ba4ea4","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-10-12 08:59:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d17bf07d99398b599f77046e2692dbdbbc2fad5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 09:54:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89abbb82f186bf51c58e5f51b345c101d9833b4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 09:54:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c20eade8898d9c918c795995c4bbf8cb05d85e3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 12:40:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe1df596c38212ad83da538aab398a578c370071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 12:40:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7098c0a6b0ef4837a82e23c11a0fe3f1 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cdf7e9fd74440fa869f6435c430d73e : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50db61c10eb04879a5129ca4fef159a0 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a5b824f66324a648a1de9292ae31d4f : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65d5c96f99dc4b38a55fcd8dc7a0988a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26edc31e78ae4f3a98af369fb6515cd3 : SUCCESS in 9m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a68c7b8a55904705b6afb96755fdd226 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9938887f5692437f9bef503ac0017804 : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d112fc92b5f44854b81f3fdad5e949e4 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8e721632ef4f452281fda77b4fb74f30 : SUCCESS in 45m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88d3a771e7c144d880d57ba644204722 : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e6738220c27145549129b5eb9ef19a5e : SUCCESS in 1h 34m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7b891f14ae740068d24bd7aba360c19 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"05fff9ca1523e7abfd825e97aaf744348ec45105","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 12:43:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a6f69ef5854d8aa55a117e437d5dc3 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5934280e03710bb5ddb3b19527ec44a9d2cf831b","revisions":{"5934280e03710bb5ddb3b19527ec44a9d2cf831b":{"kind":"REWORK","_number":1,"created":"2019-10-11 17:28:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/80/688180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/688180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/688180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/688180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/688180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/688180/1"}}},"commit":{"parents":[{"commit":"e6cd43d6cc2710266a1990c952201c5714a03393","subject":"Remove duplicate keys in sample config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6cd43d6cc2710266a1990c952201c5714a03393"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 17:27:50.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 17:27:50.000000000","tz":-240},"subject":"Switch to openstack-python3-ussuri-jobs","message":"Switch to openstack-python3-ussuri-jobs\n\nOctavia was using train jobs template: openstack-python3-train-jobs\nbut now we are in Ussuri cycle so let\u0027s switch to new template.\nModeled after neutron change https://review.opendev.org/#/c/688104/\n\nChange-Id: I824c3ed541cfd42e3c8f02be5da694f42f40d7c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5934280e03710bb5ddb3b19527ec44a9d2cf831b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5934280e03710bb5ddb3b19527ec44a9d2cf831b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
