)]}'
{"id":"openstack%2Foctavia~688548","triplet_id":"openstack%2Foctavia~master~I161a522abad4a2aa521ea46cb1065c5b05a2cd2e","project":"openstack/octavia","branch":"master","topic":"tox-cleanup","hashtags":[],"change_id":"I161a522abad4a2aa521ea46cb1065c5b05a2cd2e","subject":"Fix batch member update error on empty change list","status":"MERGED","created":"2019-10-14 19:47:13.000000000","updated":"2020-03-04 16:49:16.000000000","submitted":"2019-10-28 13:34:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688548-1572269653581-cd6aae0a","meta_rev_id":"a14cdffb81db56314d0fd217e702d97957517d27","_number":688548,"virtual_id_number":688548,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 13:34:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-28 11:33:18.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-10-21 23:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-10-23 17:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-10-28 11:33:18.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 23:56:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-23 17:09:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-28 11:33:18.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-10-28 13:34:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fc71f8d4b6f231f09b8700f317a9c7646270eb3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-14 19:47:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47fcbe2022e2c87fc85e5492b4f30d058b9e593c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-14 20:00:29.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"292f7260d14b36eeac6c3d63bdcb399edbd76a38","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-14 20:03:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70f2dc5ea1154783a935d58ac41adedd99a8d914","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-14 20:45:10.000000000","message":"Patch Set 2: Code-Review-1\n\nBlah, yeah, bummer.\nThis also needs to update the driver guide: https://docs.openstack.org/octavia/latest/contributor/guides/providers.html#batch-update","accounts_in_message":[],"_revision_number":2},{"id":"d86e8b73aeddf0009a66f98f91959c3faf776be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:39:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef3c32491ae4a729149409476c1f953 : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/653b9fd48a6a4c089d3c7f72ee5a0801 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beaf1382d3694cb69af433549b8a4ce0 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c4b484881b54dc8ac1444a1338a0983 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d12d7e4afcc44c08829b9ce1c1df97cf : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e0848a077294ee9958fe151e063055b : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da2c86755c34d7e98c38147b70c8a18 : SUCCESS in 10m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/973663c8a8ee4d4fa16ed43e58a8f47c : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cc8b1d58c2244388cefebd173eb2731 : SUCCESS in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/044038e0b8594d4fb9172c9a6ab10040 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c95815e67f9246b3812850111eae62e4 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c6d032930824a32963fc4f08a232273 : SUCCESS in 42m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/38458a036f744b558befe4a10be3de56 : SUCCESS in 52m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/253b42e168ce4b978fd0087a3eb32d5d : SUCCESS in 1h 49m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/66b2fdb4c08146ad849b2a8db653ed23 : SUCCESS in 1h 43m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ee2110fcc3c4f149f1aa21762d7a6cf : SUCCESS in 2h 29m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a8d9cf73dba4b87bfa90e2e8bc12d58 : SUCCESS in 1h 56m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d8b8b187ff641cb939090d851100591 : SUCCESS in 1h 55m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74ae60a779e14f6aa58f4bf3d7e0e078 : SUCCESS in 1h 17m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf97ce20d422408c902d0e7d8b30d763 : SUCCESS in 50m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/33822440436e4e8080a5f189896c0ad9 : SUCCESS in 1h 43m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc3cbf836e46477bbfe47519796a6251 : SUCCESS in 1h 05m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/206dfdd0b28548bc9283f13eaaefaa82 : SUCCESS in 1h 20m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2c4bddbea7624fc9906cc7385f8dd3c5 : FAILURE in 1h 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a5566b94adb082c84a6b860407f78f22cfe428f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 23:09:20.000000000","message":"Patch Set 2:\n\nGood catch.","accounts_in_message":[],"_revision_number":2},{"id":"5e79095dc97719e778ccd0cbe8a81b3b621dc689","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 23:10:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0079f435574cf076c817e476e5bc5beca3788547","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-18 23:51:12.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f8cdb76540a9137a6a4ccd721cd769ffb9ba0c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 01:35:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c82225527764991a09d7871ced301c6 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1795fd99d4c748b78027cd0fd3427fcc : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e92e022d68f4953acc0e1a1178859c1 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f82df85df55649dd96a3d743ba09f1d7 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22b4c612220e40f78a8005a174b4b077 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9576bb7f66664992a563a66ac181987b : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2229fe382445f69b0ff6b4ebdc17b1 : SUCCESS in 10m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a1d8ffc9d4924ab8986ee23aa79e6d11 : SUCCESS in 4m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/053160aa0442458c99b7a5e5911bd3c5 : SUCCESS in 6m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f7b7273e7dab429f8cbbb8208d3dde3a : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66b2825b43814a3ca4797e557c6d3e48 : SUCCESS in 5m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d673ac6d9004b9ba0ce740465d5acea : SUCCESS in 42m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3b75868d840485fb92feaeb502fdcc1 : SUCCESS in 40m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b88773429bc24252a1cb1438d70bcd94 : SUCCESS in 1h 34m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3e3fd95c040d46e6a47524889fb809ec : SUCCESS in 1h 23m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c27f2bb1dd294bc19572def42108a585 : SUCCESS in 2h 19m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a861ada74bcc4e0cb2114879ac24bd17 : SUCCESS in 1h 10m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6363f30b19ca49d7abbf9c69182a121c : SUCCESS in 1h 55m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dfbc521f8dea4bbe94f3442f6b75b9cc : SUCCESS in 1h 06m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/823e50a623a140868ade1ef473817ea1 : SUCCESS in 42m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4d3d310a440541f998682922e6ee6cc9 : SUCCESS in 1h 35m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46f5bfb4f0da47aa92c582c0ab82e308 : FAILURE in 53m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5f44506d7243467e85a21fb36e3f700b : SUCCESS in 1h 07m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4aa42edc7e7d474aaa2d3119801d6c78 : FAILURE in 1h 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45f74768574473a1eddadd581efa9817c5f473ad","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-21 22:51:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"703703c7cc1a522077a699f750403e38e115cdb2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-21 23:56:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"04f8404c11f68d7c84a5cbf986c0b32118872217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 01:18:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/519776f21059434fa5857067b8762f38 : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28459d6dedac4d82a9c306f94e85d7fa : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de28e3f6d324d3fb8d6d65895042853 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/827795cb34f64e8c8e8066b76f6bcfa1 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d651cfe382b1416bb99d26875ac47cc3 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a20530275fe4bdca85989158fb25bf3 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2144d9862121438c9845ff0b5910c754 : SUCCESS in 10m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff9ddc552b3f48028aa1e0dbafb0c51b : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/beb240af3041400a85ccb82d8effedee : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c26ac60fd8eb47c29a158dce4b878854 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f49cb9cb02e454c81f1f6d9adfd9f60 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f8339dbcbee488db5a4064f6f274630 : SUCCESS in 45m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5ecca5f9814340db9e536ac68e109eba : SUCCESS in 45m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78fcfcf6bd83499fb133c362aa924aba : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e651367ac2434e4eaee7693d13674104 : SUCCESS in 1h 45m 16s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/976f4dcd1af24baf99d435117efa997d : FAILURE in 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bcf9b2190e8d4ddda6f5feeb9e0d6118 : SUCCESS in 2h 00m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/02d55169afbc46a2b4cb5741ca3e63a1 : SUCCESS in 2h 19m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57531ee9ed264133a473fc51101f683c : SUCCESS in 1h 17m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/acea0190609c4fef8b6a44732c92f9fe : SUCCESS in 54m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/24139c86f8594916a130470470ef4f9f : SUCCESS in 1h 36m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c71210f299a4a8081efb3d8ec045a83 : SUCCESS in 1h 01m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f37a093ac90047f1890bbc5f91bb6eef : SUCCESS in 1h 09m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6b39f73d937c405d865f26cf7ee0d8ac : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da9f76058d48f802436b53bb2db91f290a4764c3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-23 17:09:56.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"f175b9d02a09a65f916b0bef597af2812b4308af","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-10-28 11:33:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2c0b6c8e57240342c6b3f71044c75ba20720aa73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:34:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1c60990b9123fe84cb6e01d58b89f7ec25a35df7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 13:34:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b962ea3c0441d5fded44dd14fb07a5e00ad0bfe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 13:34:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301a4d62d79b48359e90cd3b41b53687 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f54b27aa5040d880ed10075cd38f31 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/253f41c3028544d095332fe3d0c0140c : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/430b753780b547af8125347ed25758d2 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdbb7d38372c4f6e9af1c08e19801f06 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b785acd539d49f895653a7d704ed2df : SUCCESS in 11m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4a6cb994239d4ff7b59ac3397893204b : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44b7294976274329b45c02063b5b0416 : SUCCESS in 6m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df735ca37dd642eebd5e4d7ffb260e9a : SUCCESS in 35m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f7cebafd8eaa421f9714412ca08f43a8 : SUCCESS in 53m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e43a4ae472024a4da7a3d1527b669b1d : SUCCESS in 1h 58m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8e5c78051015429b8da8de0a31b9f983 : SUCCESS in 1h 51m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d4bb6a5c61dc456ea57afbeace8e4be6 : SUCCESS in 1h 33m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c187823e2c3e454f69f2f888672d93cfe339261c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 13:35:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632af9dc1a5d49499654dfd91a356e3d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4},{"id":"9530a74e80bcb95044562d44248c0ae033a9b7c3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-04 16:49:16.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 495749dbd5f85e9fd30b0697095e671829440728","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"01dfb8ffaf6f22e48915e05f70553870c39ea96a","revisions":{"0299a380aac5728eef23b27554b06d72d7f950af":{"kind":"REWORK","_number":1,"created":"2019-10-14 19:47:13.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/48/688548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/688548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/688548/1"}}},"commit":{"parents":[{"commit":"3177aade971dbab78e3d02a9888b9003c8d51773","subject":"Merge \"Clean up requirements.txt from tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3177aade971dbab78e3d02a9888b9003c8d51773"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 19:43:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 19:43:32.000000000","tz":-420},"subject":"Fix batch member update error on empty change list","message":"Fix batch member update error on empty change list\n\nIf the list of changes was empty, the worker would fail to fetch the\npool because it was retrieved implicitly from one of the changed\nmembers. Pass it explicitly instead, and also short-circuit on NOOPs.\n\nStory: 2006719\nTask: 37090\n\nDepends-On: https://review.opendev.org/#/c/688546/\nChange-Id: I161a522abad4a2aa521ea46cb1065c5b05a2cd2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0299a380aac5728eef23b27554b06d72d7f950af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0299a380aac5728eef23b27554b06d72d7f950af"}]},"branch":"refs/heads/master"},"40b590d68e2c373cdf84f5d6eaf1bcb72c02f826":{"kind":"REWORK","_number":2,"created":"2019-10-14 20:03:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/48/688548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/688548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/688548/2"}}},"commit":{"parents":[{"commit":"3177aade971dbab78e3d02a9888b9003c8d51773","subject":"Merge \"Clean up requirements.txt from tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3177aade971dbab78e3d02a9888b9003c8d51773"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 19:43:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 20:02:43.000000000","tz":-420},"subject":"Fix batch member update error on empty change list","message":"Fix batch member update error on empty change list\n\nIf the list of changes was empty, the worker would fail to fetch the\npool because it was retrieved implicitly from one of the changed\nmembers. Pass it explicitly instead, and also short-circuit on NOOPs.\n\nStory: 2006719\nTask: 37090\n\nDepends-On: https://review.opendev.org/#/c/688546/\nChange-Id: I161a522abad4a2aa521ea46cb1065c5b05a2cd2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40b590d68e2c373cdf84f5d6eaf1bcb72c02f826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40b590d68e2c373cdf84f5d6eaf1bcb72c02f826"}]},"branch":"refs/heads/master"},"b18b287194f6e7ba7b8a4f32cefb3834be710091":{"kind":"REWORK","_number":3,"created":"2019-10-18 23:10:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/48/688548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/688548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/688548/3"}}},"commit":{"parents":[{"commit":"fce2cd4f4996824c5512270138cb8f237457acec","subject":"Revert \"Clean up requirements.txt from tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fce2cd4f4996824c5512270138cb8f237457acec"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 19:43:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-18 23:09:06.000000000","tz":-420},"subject":"Fix batch member update error on empty change list","message":"Fix batch member update error on empty change list\n\nIf the list of changes was empty, the worker would fail to fetch the\npool because it was retrieved implicitly from one of the changed\nmembers. Pass it explicitly instead, and also short-circuit on NOOPs.\n\nStory: 2006719\nTask: 37090\n\nDepends-On: https://review.opendev.org/#/c/688546/\nChange-Id: I161a522abad4a2aa521ea46cb1065c5b05a2cd2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b18b287194f6e7ba7b8a4f32cefb3834be710091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b18b287194f6e7ba7b8a4f32cefb3834be710091"}]},"branch":"refs/heads/master"},"01dfb8ffaf6f22e48915e05f70553870c39ea96a":{"kind":"REWORK","_number":4,"created":"2019-10-21 22:51:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/48/688548/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/688548/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/688548/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/688548/4"}}},"commit":{"parents":[{"commit":"fce2cd4f4996824c5512270138cb8f237457acec","subject":"Revert \"Clean up requirements.txt from tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fce2cd4f4996824c5512270138cb8f237457acec"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-14 19:43:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-21 22:51:27.000000000","tz":0},"subject":"Fix batch member update error on empty change list","message":"Fix batch member update error on empty change list\n\nIf the list of changes was empty, the worker would fail to fetch the\npool because it was retrieved implicitly from one of the changed\nmembers. Pass it explicitly instead, and also short-circuit on NOOPs.\n\nStory: 2006719\nTask: 37090\n\nDepends-On: https://review.opendev.org/#/c/688546/\nChange-Id: I161a522abad4a2aa521ea46cb1065c5b05a2cd2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01dfb8ffaf6f22e48915e05f70553870c39ea96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01dfb8ffaf6f22e48915e05f70553870c39ea96a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
