)]}'
{"id":"openstack%2Foctavia~688776","triplet_id":"openstack%2Foctavia~master~Iaada1b5d519bfc57528aa15bae8c0ee2b55f0567","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iaada1b5d519bfc57528aa15bae8c0ee2b55f0567","subject":"Add client authentication to the LB cookbook","status":"MERGED","created":"2019-10-15 18:11:11.000000000","updated":"2019-10-16 16:49:55.000000000","submitted":"2019-10-16 16:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688776-1571244422466-9e252c24","meta_rev_id":"c1a40d0c7fa6d87d0c38420f7a20d6b2ec04a243","_number":688776,"virtual_id_number":688776,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 16:47:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2019-10-16 16:30:29.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-10-16 15:56:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-15 20:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":1,"date":"2019-10-16 15:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":1,"date":"2019-10-16 16:30:29.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 20:52:18.000000000","updated_by":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"reviewer":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"state":"REVIEWER"},{"updated":"2019-10-16 15:29:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-16 15:56:37.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-16 16:30:29.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-10-16 16:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e81a801e9096a573cd5c6613437cdd129c6b03ea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-15 18:11:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bd06dc696cd3034ac0d2fbee1b4fd34ffa8600d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-15 18:12:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0acc600f890234d02f436aea039d64ded4340aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:22:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6116bc78ea41cc93970f112ada293c : FAILURE in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3034bd6a49246579e19ab533d6db67d : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":2},{"id":"02efe5edc0145c50394bcd5bdaf0be0603a24d4c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-15 18:24:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8f0203887dd551adf1e9912095ae1dfe719432c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:39:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41370da77b5e48b0b29d8e5450959ad2 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1509cbcd0444f1294d83424f7e82c38 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a9cc9e01cd85a75ac36c19ddd24effeffa2a7954","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-15 19:20:46.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b6521021b355661e5daba1243cc7b13768802b74","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-15 20:30:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9310cf9d6a709a5abeb4a239a9952e9b2f3a1505","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-15 20:46:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b046d76a0ce59c944ffe5bf7d1ad7374ee34f3f6","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-10-15 20:52:18.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"87192e4b1a08eadd8aa69377558371fea2b57f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:56:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d89b73de934d0385a49b79fa23139e : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a615c3cc060548b9a898525a0ebcef71 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":4},{"id":"6106d78896260fe5740a1fc4315af6735c2c25fe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 15:29:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3fb685218198a68500b856bc5ba5dd8590844409","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-16 15:56:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b3212e0c88c679a67cd5551ab64292f310bb9f21","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-10-16 16:30:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"894d804850b03b3f245ef71da82da14a88fb547a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:30:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"77724456afa7e252c6689562b49c3cd24220db29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:47:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c29bd1071a249e483fa11ee98a13a84 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06b9a75abe64220b99a076ad4daf2e3 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":4},{"id":"bc215f7de819be54df83debd4519b3217a5b818a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:47:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c1a40d0c7fa6d87d0c38420f7a20d6b2ec04a243","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:49:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7f1eecb5bf4ede8b563145460cafc0 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4b4638e67d0a324839fabf1524a8f0a1e0c1dc31","revisions":{"65cbd58b16f108737bf4534dcef4a420a4fede20":{"kind":"REWORK","_number":1,"created":"2019-10-15 18:11:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/688776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/688776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/688776/1"}}},"commit":{"parents":[{"commit":"25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf","subject":"Merge \"Use bandit block in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:09:42.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:11:10.000000000","tz":-420},"subject":"Add client authentication to the LB cookbook","message":"Add client authentication to the LB cookbook\n\nThis patch adds a cookbook section for creating client authentication\nenabled listeners.\n\nIt also removes two references to granting access to secrets in\nbarbican that are no longer required.\n\nChange-Id: Iaada1b5d519bfc57528aa15bae8c0ee2b55f0567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65cbd58b16f108737bf4534dcef4a420a4fede20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65cbd58b16f108737bf4534dcef4a420a4fede20"}]},"branch":"refs/heads/master"},"f88315ccdbc9cbcbd1934664d39b8c246f72c7f2":{"kind":"REWORK","_number":2,"created":"2019-10-15 18:12:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/688776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/688776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/688776/2"}}},"commit":{"parents":[{"commit":"25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf","subject":"Merge \"Use bandit block in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:09:42.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:12:40.000000000","tz":-420},"subject":"Add client authentication to the LB cookbook","message":"Add client authentication to the LB cookbook\n\nThis patch adds a cookbook section for creating client authentication\nenabled listeners.\n\nIt also removes two references to granting access to secrets in\nbarbican that are no longer required.\n\nChange-Id: Iaada1b5d519bfc57528aa15bae8c0ee2b55f0567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f88315ccdbc9cbcbd1934664d39b8c246f72c7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f88315ccdbc9cbcbd1934664d39b8c246f72c7f2"}]},"branch":"refs/heads/master"},"6606ff7610260b9f7a1fd345a48d372931ed8dc8":{"kind":"REWORK","_number":3,"created":"2019-10-15 18:24:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/688776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/688776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/688776/3"}}},"commit":{"parents":[{"commit":"25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf","subject":"Merge \"Use bandit block in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:09:42.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:24:35.000000000","tz":-420},"subject":"Add client authentication to the LB cookbook","message":"Add client authentication to the LB cookbook\n\nThis patch adds a cookbook section for creating client authentication\nenabled listeners.\n\nIt also removes two references to granting access to secrets in\nbarbican that are no longer required.\n\nChange-Id: Iaada1b5d519bfc57528aa15bae8c0ee2b55f0567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6606ff7610260b9f7a1fd345a48d372931ed8dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6606ff7610260b9f7a1fd345a48d372931ed8dc8"}]},"branch":"refs/heads/master"},"4b4638e67d0a324839fabf1524a8f0a1e0c1dc31":{"kind":"REWORK","_number":4,"created":"2019-10-15 20:46:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/688776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/688776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/688776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/688776/4"}}},"commit":{"parents":[{"commit":"25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf","subject":"Merge \"Use bandit block in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25c31fe6d22d0feb7fdd2bc1fdf9ba0f707058cf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 18:09:42.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-15 20:45:55.000000000","tz":-420},"subject":"Add client authentication to the LB cookbook","message":"Add client authentication to the LB cookbook\n\nThis patch adds a cookbook section for creating client authentication\nenabled listeners.\n\nIt also removes two references to granting access to secrets in\nbarbican that are no longer required.\n\nChange-Id: Iaada1b5d519bfc57528aa15bae8c0ee2b55f0567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b4638e67d0a324839fabf1524a8f0a1e0c1dc31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b4638e67d0a324839fabf1524a8f0a1e0c1dc31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
