)]}'
{"id":"openstack%2Foctavia~689286","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I8a0cdf7ee912d39d9ac24244728094a1d1e15c63","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I8a0cdf7ee912d39d9ac24244728094a1d1e15c63","subject":"Do not set log format when \u0027no log\u0027 is set","status":"MERGED","created":"2019-10-18 01:15:21.000000000","updated":"2019-10-29 01:03:30.000000000","submitted":"2019-10-29 01:01:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689286-1572310918658-fa70c53b","meta_rev_id":"85e2ca2e67089127e78b5ee59830d5e035dc9484","_number":689286,"virtual_id_number":689286,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 01:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-18 21:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-25 13:14:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-10-28 23:17:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2019-10-25 13:14:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-10-18 01:15:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 01:15:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-25 13:14:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-29 01:01:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e987e1f1c72e58dbb59efa2a1f9ebe894a1b4ff2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 01:15:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b3cf4467dd8afb185ebd7ab9c26a1f3c8fb9934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55dca895c88b4dafb7a01c1ca0bb458a : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a763da9cebf44578ee5da5ea0c9534a : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ecbe8785784d179a34bd4e6de3058a : SUCCESS in 8m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2996ec112c8d417285bb21cadab76a46 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caeec6544b4a4dd2b0d31af457d97392 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35211de022ee4b849af7c26442937197 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d424db457d67478491701c288ba60659 : SUCCESS in 16m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ffa4b690b8734e8da227418eecc25418 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3ce8fdae3bcd4a1ca94bede26dd286df : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ea82f13d2a624f9e87626f361db111ab : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf2886f4f75f47b69ecabc820d4aeed2 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2608f9bb4e64119916766a117433e90 : SUCCESS in 40m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c4d5a0e613514fbd89e1408729c92ce7 : SUCCESS in 41m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f8d8441662b4137acbd19cf6ef11533 : SUCCESS in 1h 39m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2c559b480b4d4628a1000e319491e5b0 : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/407be74604f44b29a5b77932092708f9 : SUCCESS in 2h 07m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47c6985e340c4cd09085ab61d6b4bc04 : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed4c48cbbf204202a12fd1c0c6ffc294 : FAILURE in 1h 44m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57be7fb56548406486c02601949cc7ae : SUCCESS in 1h 09m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb6ac14bdefd499991c6fe8a6e0b12de : SUCCESS in 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/71448bded9be4e87b350f12c969f0605 : SUCCESS in 1h 27m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a5daeaf822244e5b0845265226138b9 : SUCCESS in 59m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dbba0c091a6e45a18ef5b262b982988a : SUCCESS in 1h 06m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46df86917d564694abf8432b97ee6b56 : FAILURE in 1h 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"540a2da40f419ac7d88f613fdac0508960f5acf4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 21:21:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4785277a21be1fbbfbfc4606f943395883b489c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-25 13:14:38.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2\n\nBC-2 just to signal no need to backport farther back.","accounts_in_message":[],"_revision_number":1},{"id":"b85979930e7019e74a688034622ff94312b29f46","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-28 23:17:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88e60c0cadb26c8cfe701f86a517d6bd691ec067","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 23:17:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7aa71fee5015c0a9484ba9b83e6efbad5ed639a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:01:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18f929d1c9184b50aa8aa394fa6442c5 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fddd31d1a334f67b1955685d58261cf : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e36cafa65a7946b3847d3d7b8edf72a1 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad0025924d4842deb676c19287954592 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac561c7f884346ceb12740195e48d948 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026bd10667f044fbbc1080be5b33e380 : SUCCESS in 11m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bdce21ca7c4e44a385d7e4fc58ce7f39 : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fbfa9e10fca44642bf186ee5350a7135 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ec0f9395a4541559ffbb9e8ea78698b : SUCCESS in 37m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e15d9ff858464cfc85050d8ec5077b48 : SUCCESS in 37m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5490fd0d0044bf7a0929c8faee33a74 : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f285dd8eab6646d5b5a0ed6a077a1717 : SUCCESS in 1h 42m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/150810d991b2431dbd438949c868b757 : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"005c8b24f04e92ff26faf7b6d788464b1716468d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:01:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"85e2ca2e67089127e78b5ee59830d5e035dc9484","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:03:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd03502d42e04d8ea8497879eaaaeb67 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d64576e3facebe0811677d7cf9a53a2310c555bc","revisions":{"d64576e3facebe0811677d7cf9a53a2310c555bc":{"kind":"REWORK","_number":1,"created":"2019-10-18 01:15:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/86/689286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/689286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/689286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/689286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/689286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/689286/1"}}},"commit":{"parents":[{"commit":"6f54167b05cab1ef2721d63ddee99f92d61ea14f","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f54167b05cab1ef2721d63ddee99f92d61ea14f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 15:04:25.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-18 01:15:21.000000000","tz":0},"subject":"Do not set log format when \u0027no log\u0027 is set","message":"Do not set log format when \u0027no log\u0027 is set\n\nRe-factored code to not insert a log-format line in haproxy\nconfig when \u0027no log\u0027 is set in the template. Adjusted tests\nas necessary.\n\nStory: 2006181\nTask: 35706\n\nChange-Id: I8a0cdf7ee912d39d9ac24244728094a1d1e15c63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d64576e3facebe0811677d7cf9a53a2310c555bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d64576e3facebe0811677d7cf9a53a2310c555bc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
