)]}'
{"id":"openstack%2Foctavia~689588","triplet_id":"openstack%2Foctavia~master~I4941df3894148482c597d30f3a8db70659fd1b7a","project":"openstack/octavia","branch":"master","topic":"tox-cleanup","hashtags":[],"change_id":"I4941df3894148482c597d30f3a8db70659fd1b7a","subject":"Revert \"Clean up requirements.txt from tox.ini\"","status":"MERGED","created":"2019-10-18 20:55:45.000000000","updated":"2019-10-19 01:04:05.000000000","submitted":"2019-10-19 01:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689588-1571446967705-f62c7b6a","meta_rev_id":"d13bbfb5f03d709a9098c94fca519a7b8314e232","_number":689588,"virtual_id_number":689588,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-19 01:02:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-18 22:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-18 21:05:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-10-18 22:24:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2019-10-18 21:05:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-10-18 20:55:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"date":"2019-10-18 20:55:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"date":"2019-10-18 20:55:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 20:55:45.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2019-10-18 20:55:45.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-10-18 20:55:45.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-18 21:05:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-18 22:24:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-19 01:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa7a585c349309951bb0d6826a08ca52f96dfe80","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-18 20:55:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6eaa8bd3946215c2208d19adb7a75a2d477b190","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-18 21:05:24.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4a64e9f9fecbb7a42d5d1b8f4a537d6b113844c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 22:24:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nYep, I see what\u0027s happening now. Confirmed locally that it does the develop install in a different stage, so it doesn\u0027t follow U-C if we don\u0027t pre-install the deps using U-C explicitly.","accounts_in_message":[],"_revision_number":1},{"id":"12d29139d0a34a44924a1b74c72929fc70b441bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:11:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3696ad75a3d427ea3a9b7d5f04fb7ed : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86486dc2c3574f9d99246909083f7297 : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9c7908a120467ba8bfd5d1e474dabe : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/112a22784bde4d0ebcc438657dc2a833 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/935cf97a3b714a9f9f9d643e2990ad48 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80f08a97f1234c16be5dad420af1e44d : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01ee629deb73403ea387db08f0729cdb : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82ea3e6691048d1bbe792c318b19aea : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11f08ea4c2774987935b6952c9461e29 : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30e97a1160ce4b9c8adb828dc6ece07e : SUCCESS in 4m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/54d4b6d8b3894fbc9ffe265ae7bb35dc : SUCCESS in 6m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e9e631176284dd2bdbe83896f65c3c8 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2975c80e8ccd4c389409ad59c28a723e : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ab9c776e026475bb385cc91e765aee3 : SUCCESS in 47m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/024f33cf658647779cce6ea89c1940ec : SUCCESS in 46m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2598c28c3d24b7a9e224222f4469a27 : SUCCESS in 1h 47m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b9b8e9737b134383ae4be421e349d2ba : SUCCESS in 1h 34m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/39fec6e5de994e42b16655faebbe2fa4 : SUCCESS in 2h 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e30587cd02fa4f4697d0016a4554233f : SUCCESS in 1h 55m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/15e0eb2013be4b98b4f1d4220d20db41 : SUCCESS in 1h 55m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8130461c6ce481d9c6847ffb2f58c20 : SUCCESS in 51m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/736f331682b349529f78cd81fc718fe7 : SUCCESS in 1h 29m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b99dd400d6ae40738e4d4f7fe2163645 : SUCCESS in 1h 10m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d56e9df2540045f5b6f68f3d024d6c91 : SUCCESS in 1h 04m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/38b6a7fd572c4bdf84e8257701a56fc9 : SUCCESS in 1h 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a76a382ed4ffd1833af560d82e2fba0b24bac062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:11:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fefab423e2cc388ccb3841e4fb52fcdca5707058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 01:02:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b865cba6e098ce5fb8b2e7cf3a5e146f975aa69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 01:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/56fffc1befc04d07aa5b04bf5ff92936 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc104e03de934bd7b8bb2a8892507dc6 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f19eb27fcdf48e3a9f4f4d006fb66ce : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e634d50a9e884af39d296d0fcf12899c : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95e86ab849934b9c9e548411cd8f6154 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/760daa7888c849a4b2a5b029987ce7a6 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4dca1f8e5a4eafa94d6cd1d278e32f : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cc2c761822f4099aaf6644147ec6cfd : SUCCESS in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2bf16e8c07cb47c196cff646ca6d5a72 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8cee88f7b84c4ec9aecf0112cad41715 : SUCCESS in 5m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb3eb9f05b1c4071932a7d890c4d566f : SUCCESS in 45m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/73d271d3d12c4b65ac5111592d2314c4 : SUCCESS in 40m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03993afce44843268987a0764f640476 : SUCCESS in 1h 29m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6fe53b362d8349218e3d93b3927717bd : SUCCESS in 1h 49m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3c03c1ac4eb7cfca5e89da31bfa3246b12403470","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 01:04:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8521ab4724aa4d3b8321ce91f0bb2deb : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce5dc53061a4270bedad771fb23178f : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f918bc3c49954b0782d9999b564593a6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fce2cd4f4996824c5512270138cb8f237457acec","revisions":{"fce2cd4f4996824c5512270138cb8f237457acec":{"kind":"REWORK","_number":1,"created":"2019-10-18 20:55:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/88/689588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/689588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/689588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/689588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/689588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/689588/1"}}},"commit":{"parents":[{"commit":"7b8e6de1b8230e5d31ca36624e7fdfba25e3e077","subject":"Clean up requirements.txt from tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b8e6de1b8230e5d31ca36624e7fdfba25e3e077"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-18 20:55:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-18 20:55:45.000000000","tz":0},"subject":"Revert \"Clean up requirements.txt from tox.ini\"","message":"Revert \"Clean up requirements.txt from tox.ini\"\n\nIt appears that these are required to have the upper constraints properly applied. We need to revert this change.\n\nThis reverts commit 7b8e6de1b8230e5d31ca36624e7fdfba25e3e077.\n\nChange-Id: I4941df3894148482c597d30f3a8db70659fd1b7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fce2cd4f4996824c5512270138cb8f237457acec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fce2cd4f4996824c5512270138cb8f237457acec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
