)]}'
{"id":"openstack%2Foctavia~689974","triplet_id":"openstack%2Foctavia~master~Iedb0e9f5414a9fd82e189c71e565ac0771a1482d","project":"openstack/octavia","branch":"master","topic":"missing_ws","hashtags":[],"change_id":"Iedb0e9f5414a9fd82e189c71e565ac0771a1482d","subject":"[Trivial]Add missing ws between words","status":"MERGED","created":"2019-10-22 06:32:31.000000000","updated":"2019-10-23 05:35:50.000000000","submitted":"2019-10-23 05:33:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689974-1571808798744-fe29816b","meta_rev_id":"ebf9f8ae72576eecd60ed9418b617fd5170b9bcd","_number":689974,"virtual_id_number":689974,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 05:33:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-10-23 03:40:37.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-10-22 14:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-10-22 22:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-10-23 03:40:37.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 14:53:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-22 22:33:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-23 03:40:37.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-10-23 05:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc811217b827766ee383e0f20f6aca58376f1fa4","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-10-22 06:32:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2866c56b0b53b323798b9185526043b33c4e53e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:07:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/638c5430c40c4321b60b57237199432b : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f8d31ec48bf44c1b7230c86a6ef2dd9 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89af3ad800e74c2692a590011f7f4b3d : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1f0989b80aa4f0f88fba4c678f8fa82 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae6003878e4541e09441c6c988805564 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5adeafdadca84230aaeb50587b1a229f : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93121e619db74b29a32308888b3c54a6 : SUCCESS in 11m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d702d4b7bbc84f81986187abe76b3e99 : SUCCESS in 9m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68bb5f45fa894f84a02bbbb4871f3372 : SUCCESS in 11m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f23bed3c2c5447c9b4ce4833be17dea : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8402840d43884d15a2692ccb58fb5bb0 : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3be114c69a8417abe793166e84f53ae : SUCCESS in 57m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2761799b8a4c43ad817446bc35ba1719 : SUCCESS in 1h 04m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b82919bd54de423e84a48833d7b83c2b : SUCCESS in 1h 53m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8ba47edb30214d38ae3b240d374d2c28 : SUCCESS in 1h 47m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7bc654bb79454579b735187fd0f93327 : SUCCESS in 2h 28m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c60e67f6a954375a489eb9cde50b32c : SUCCESS in 2h 25m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1880724c4625452eb8442166d4e33a29 : FAILURE in 1h 29m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/adcbbf8c113c4faaa5898efd3348c99f : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c38d6204f65e4609b217b1bf37f6508a : SUCCESS in 1h 13m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e655d2d2ca544dc98d9fcaa501dc7ba3 : SUCCESS in 1h 58m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a65677af4194750b02dba411ec338c1 : SUCCESS in 1h 13m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6beb3a6dab55434d8b8f3abe74b4c091 : SUCCESS in 1h 35m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/16262505cc4648e6804d32e647fb3ee0 : FAILURE in 1h 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4781578e21bf498f9e160e0de80e4428f2843806","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-22 14:53:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed88e5659aefda1b3dd5c7200138e19dd7c6ae44","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-22 22:33:12.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":1},{"id":"dfb4743d704bd7808668e5e4c83440347a885d3f","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-10-23 03:40:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd97ce512134c4df62ff60fe9672ee77509bbab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:40:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab2eb88e24c56bd7b252c00f4f0803ee405edccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 05:33:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e92a0b007cf6f0889c3db2acca4556ff5a25a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 05:33:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/384e7cb963c94093b057a55069b42eec : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab818c9812b74cc68354a03ba1b2f5bc : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/333a325a7a9741f4a247fbc339e1bfe1 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bae217d081444f4b9058f671d0813ff8 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23ae4eb8b67b4f4fad2dd52b57ffaf4b : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b18be5a91440fa80414104afb7685e : SUCCESS in 8m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/26e7d906cf354d5ab1f19453f961401b : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08a52c4c86e44158b2df7aa3cafc54fe : SUCCESS in 4m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ecd71b4dd45f45ebaa97f4b5994beb78 : SUCCESS in 35m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a7feb458fa5c424ea795e31b234bb0c9 : SUCCESS in 39m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01a4f103ae1c40e0b1919e82f1d22c0f : SUCCESS in 1h 24m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2697e1f0bba943f6b41c492ff08a7498 : SUCCESS in 1h 46m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e375e6af1f34477b09a9f89eacc54da : SUCCESS in 1h 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"764aa373cf208b9244e61d9fc13061b284053675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 05:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0868bc0b6bcd47529b2006afcf684beb : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28694f02335a45606f8a5a2b2e8603ea7cef0743","revisions":{"28694f02335a45606f8a5a2b2e8603ea7cef0743":{"kind":"REWORK","_number":1,"created":"2019-10-22 06:32:31.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/74/689974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/689974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/689974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/689974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/689974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/689974/1"}}},"commit":{"parents":[{"commit":"ae9895adccab4886187ddb31cbb9ff084db589a9","subject":"Merge \"Fix log offload file permissions in CentOS devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae9895adccab4886187ddb31cbb9ff084db589a9"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-10-22 03:51:43.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-10-22 03:51:43.000000000","tz":480},"subject":"[Trivial]Add missing ws between words","message":"[Trivial]Add missing ws between words\n\nThis is to add missing ws between words in help messages.\n\nChange-Id: Iedb0e9f5414a9fd82e189c71e565ac0771a1482d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28694f02335a45606f8a5a2b2e8603ea7cef0743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28694f02335a45606f8a5a2b2e8603ea7cef0743"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
