)]}'
{"id":"openstack%2Foctavia~690498","triplet_id":"openstack%2Foctavia~stable%2Frocky~I0403afe369fa67d52c9bb68c432fdf2c86edcdec","project":"openstack/octavia","branch":"stable/rocky","topic":"bp/Add_amphora_metadata","hashtags":[],"change_id":"I0403afe369fa67d52c9bb68c432fdf2c86edcdec","subject":"Support to create amphora with customized metadata","status":"ABANDONED","created":"2019-10-23 08:06:13.000000000","updated":"2019-10-28 03:42:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"638831a2a3683d2f5839fe7b6ecc188e52fdcd54","_number":690498,"virtual_id_number":690498,"owner":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-23 08:13:06.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-23 10:04:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-24 14:20:33.000000000","_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"disliked":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":-2,"date":"2019-10-23 16:28:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2019-10-24 09:29:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":-2,"date":"2019-10-23 16:22:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 08:13:06.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-10-23 10:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-23 16:22:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-23 16:28:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-24 09:29:37.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"}],"messages":[{"id":"a6dcdfe96b99c362abfad535e7d0c11d312e4c9f","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-23 08:06:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0aafdeae130e5c7b8c7371cf91a62dccad0673a2","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-23 08:13:06.000000000","message":"Patch Set 1:\n\n\u003e Uploaded patch set 1.\n\nWhat is your purpose?","accounts_in_message":[],"_revision_number":1},{"id":"aaa23e8024c2b99779e7f99f5a0fb38b29a64b2c","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-23 09:55:48.000000000","message":"Patch Set 1:\n\n\u003e What is your purpose?\n\nIf you want to use service discovery to filter amphora VM, you need to add more metadata for filtering target VM. Beside that, adding name of load balancer, project id,... to metadata helps discovery get extra information for monitoring. :)","accounts_in_message":[],"_revision_number":1},{"id":"e8cde1304bd2d739b3bd1a48412c89354777f251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 10:04:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08c331d1d3f746b69e0d2ce4330c3f0e : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f3bea6158d7461b8f37322142bf5aaf : FAILURE in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15b9327b6a4c4551a7521e5aa615c363 : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d47bf7f477c347929966a0160eccd908 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b794a8dfb38453ab8ad7af64f55d047 : FAILURE in 4m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/24672c2aaa5d47ca8c2b775bed1a6731 : SUCCESS in 1h 23m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9b93955b9ff84976b539e0e187225c6c : SUCCESS in 1h 27m 49s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdedd6fb14214377a8dcd94ba107454d : FAILURE in 1h 09m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/16937ea3b01a4af1adfec99c41ff6ec3 : SUCCESS in 1h 31m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ad84023173a040a595c2134fa7abfd40 : SUCCESS in 6m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2ea5d5ea77e448e582cd4372e7d33927 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f44661b10384ca1a47f7ebfeb62043c : SUCCESS in 6m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b9bd6bfe3294a41919bfc557ce5fa3f : SUCCESS in 1h 37m 03s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c9f78c9e5b8840de96a979f93856b9c0 : SUCCESS in 1h 05m 59s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/9901be9d525e4e628c7720369bde15c5 : FAILURE in 1h 35m 25s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a3ba9df5da5247ec8ae98a0b81f84655 : FAILURE in 1h 11m 02s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/defa53d792ae48e083469d9fc179da2d : SUCCESS in 1h 12m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4492c3fc80d48a4a5aadb3d932ec9c9 : SUCCESS in 45m 23s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e72f3df063ea4baf841cdea425620209 : SUCCESS in 46m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eddb6ba8d15e4f4f91d40c7979e9664f : SUCCESS in 1h 50m 34s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3694bfb3aa134ea6a1346ab75bd6be8c : SUCCESS in 1h 53m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/041ed0abb06c40518c6071fbd388e0cf : SUCCESS in 1h 22m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/248439ba52914980b746ea8324a4f725 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e75f70b51fba920f3e3b53c08ec8aabdd1745bc0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-23 16:22:02.000000000","message":"Patch Set 1: Code-Review-2\n\nHi Dat Le, thank you for your patch. Unfortunately we cannot add this to stable/rocky. Octavia follows the OpenStack stable release policies which do not allow feature additions to stable branches. Please see this guide: https://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes\nIf you would like this added to Octavia, it will need to be submitted to the \"master\" branch which will become the \"Ussuri\" release.\nTypically patches need to be applied to the master branch first and then backported to stabled branches if it meets the stable policy criteria.","accounts_in_message":[],"_revision_number":1},{"id":"cb561a1ffea1281cf208af7d5dfe3e283431cf51","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-23 16:28:50.000000000","message":"Patch Set 1: Code-Review-2\n\nI\u0027d also encourage opening a story on Storyboard (https://storyboard.openstack.org/#!/project/openstack/octavia) to present the problem, use case and propose approaches to implement it.","accounts_in_message":[],"_revision_number":1},{"id":"c09d7d03f1a2d4d86ba5a15abff46fde3e3f244f","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-24 09:29:37.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for your patch.\nI agree with Johnson, you should put this patch to master and then this can be back-ported using cherry pick option.","accounts_in_message":[],"_revision_number":1},{"id":"2737b4e2d25b97790dc421ca731d3cbf49c8dd7b","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-24 14:20:33.000000000","message":"Patch Set 1:\n\nYou \u0027re right. I moved this change to master branch. You can review it at https://review.opendev.org/#/c/690925/\nThanks all.","accounts_in_message":[],"_revision_number":1},{"id":"61665a9a8741404c5f9d65d165b88de1f9f8f2f9","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-28 03:42:19.000000000","message":"Abandoned\n\nContribute to master branch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41bda2d75dee36cadc29c682c75f3b1c2f3263cf","revisions":{"41bda2d75dee36cadc29c682c75f3b1c2f3263cf":{"kind":"REWORK","_number":1,"created":"2019-10-23 08:06:13.000000000","uploader":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"ref":"refs/changes/98/690498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/690498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/690498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/690498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/690498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/690498/1"}}},"commit":{"parents":[{"commit":"43d3e1bde3e1257b7e3622eb222d4ab6a47114cf","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d3e1bde3e1257b7e3622eb222d4ab6a47114cf"}]}],"author":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2019-10-23 08:05:20.000000000","tz":420},"committer":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2019-10-23 08:05:20.000000000","tz":420},"subject":"Support to create amphora with customized metadata","message":"Support to create amphora with customized metadata\n\nChange-Id: I0403afe369fa67d52c9bb68c432fdf2c86edcdec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41bda2d75dee36cadc29c682c75f3b1c2f3263cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41bda2d75dee36cadc29c682c75f3b1c2f3263cf"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
