)]}'
{"id":"openstack%2Foctavia~690884","triplet_id":"openstack%2Foctavia~master~I79979198e6cb4ee8c6ad49a7b9bd8e4093cf1934","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I79979198e6cb4ee8c6ad49a7b9bd8e4093cf1934","subject":"Fix typo in doc agent.py-\u003eagent","status":"MERGED","created":"2019-10-24 08:57:31.000000000","updated":"2019-11-11 17:44:59.000000000","submitted":"2019-11-11 17:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690884-1573494209208-9a54f4b4","meta_rev_id":"c4232c51a5b958d45a589f77324d4349cf60d563","_number":690884,"virtual_id_number":690884,"owner":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-11 17:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-11 15:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-10-24 08:59:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-10-24 19:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-11-04 17:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-28 01:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-11 15:44:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 08:59:10.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-10-24 19:26:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-28 01:17:28.000000000","updated_by":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"reviewer":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"state":"REVIEWER"},{"updated":"2019-11-04 17:08:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-11 15:44:06.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-11 17:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52279afc3576d6d7b93748d50464618df5df63ca","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-24 08:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6358c9c6b62d34cb68189a0173bd0c696d7ba6b3","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-24 08:59:10.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"084ef79ea918473ccd1adf3a4f4052eb2f3403e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 11:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6672c801bc1647e8a5a18d47ac91d8d7 : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83e1bd188e1742cab49fe9b331621bd9 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4abd5e22d44d94b1da78b772ab386e : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/062cb95bf3dd4d66ada07acc2ef415ad : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98b138050627407baea0ee120c5dad95 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3a1fd7fd63b410d93bcd28f3bb10a0b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff5d065cfc14078abefad12ff8b99f9 : SUCCESS in 11m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6d695fb2379a479c8b69d24f6d23e0cb : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fef6c9c7604a43b08195f0231cce4364 : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/26278b256e2f4fa59d18b1e5e8cd2572 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4701f25d25de4c10b59bada2688bc5d4 : SUCCESS in 6m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/972608db797c42aba775c3e5062caebb : SUCCESS in 47m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/760ad2facc2b401c8616061a4068cf9a : SUCCESS in 42m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b07ccb76a775491b8317a319d43b31fc : SUCCESS in 1h 47m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2c5c6273e37e4576a891b8c3e0d87006 : SUCCESS in 2h 18m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/43dffeedb1a7417ea4ba7746fbbe2d72 : SUCCESS in 2h 01m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d72ca768873a45b49fff5a22d6e98113 : FAILURE in 1h 25m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f591c454a8cd431099e666baf35a9b54 : SUCCESS in 1h 54m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f8a32eed00f4e009804d16adcc3b070 : SUCCESS in 1h 30m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98645beb265e492ca7370dc477d54229 : SUCCESS in 51m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d92253c96a1e444a845c978b46a33e5a : SUCCESS in 1h 24m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15552525ac5e46ada19e168b77616968 : SUCCESS in 1h 15m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cd72daf21b8847da8fdfcad47f782106 : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/23fb19774e0c491ea0343623e057a78a : FAILURE in 1h 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00eb814e7191dd503c0b6030fd8312c61129932b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-24 19:26:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4b861bc46c0f873acd85f64b590768f2f81aac4","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-28 01:17:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"618259e110f53c14f470591adc584bfe415cb0d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-04 17:08:01.000000000","message":"Patch Set 1: Code-Review+2\n\nSure","accounts_in_message":[],"_revision_number":1},{"id":"dbe524f7fb2138e12da6483e45c97db773983079","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-11 15:44:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nk","accounts_in_message":[],"_revision_number":1},{"id":"e7000147d092636b6d409efa4823f542d25448a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:44:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"149378aad608298ff2af55301081addbd266e9e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 17:43:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bff76849b80642afbf90d0afe3270537 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159cd9a28d764f5895ece4b036238f65 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e2fea88db3443a8a12f8a074d5cbad6 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f37803c583104d3a8280ff75a2cf1916 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae94e3e92ec043b8938c253fb52040d6 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278865432917467ab015e9b8b22778f1 : SUCCESS in 10m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/663dd12b94ee472f95b1680dde9b6ad7 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07f289003cef496eaef860357cfd6c15 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4b3adf07eab4839bff5d52d81b163f9 : SUCCESS in 46m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/42d8c64807e64e109dc2da578a11d911 : SUCCESS in 47m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59d59934a621412eb6a1edd365466e07 : SUCCESS in 1h 28m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ed322cd3e3b34524842725660ff899db : SUCCESS in 1h 57m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78c035284d04447687aab22184349f3b : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d575614c0f0849ed0cba8791ae6dda469650f881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 17:43:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c4232c51a5b958d45a589f77324d4349cf60d563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 17:44:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2dfd9a21e424122882e9c1041936e5f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"935567b84c05be6bd75df372a9de51d43267ab9a","revisions":{"935567b84c05be6bd75df372a9de51d43267ab9a":{"kind":"REWORK","_number":1,"created":"2019-10-24 08:57:31.000000000","uploader":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"ref":"refs/changes/84/690884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/690884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/690884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/690884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/690884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/690884/1"}}},"commit":{"parents":[{"commit":"6b58e66868cd806fff4f777b776985c2e94cb75d","subject":"Merge \"Add Gentoo platform detection in diskimage-create/diskimage-create.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b58e66868cd806fff4f777b776985c2e94cb75d"}]}],"author":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-10-24 08:51:59.000000000","tz":330},"committer":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-10-24 08:57:27.000000000","tz":330},"subject":"Fix typo in doc agent.py-\u003eagent","message":"Fix typo in doc agent.py-\u003eagent\n\nThis patch fix the wrong description of agent_server_cert\n\nChange-Id: I79979198e6cb4ee8c6ad49a7b9bd8e4093cf1934\nStory: 2006612\nTask: 36797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/935567b84c05be6bd75df372a9de51d43267ab9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/935567b84c05be6bd75df372a9de51d43267ab9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
