)]}'
{"id":"openstack%2Foctavia~690925","triplet_id":"openstack%2Foctavia~master~I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be","project":"openstack/octavia","branch":"master","topic":"2006763-Add_amphora_metadata","hashtags":[],"change_id":"I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be","subject":"Support to create amphora with customized metadata","status":"NEW","created":"2019-10-24 11:24:57.000000000","updated":"2020-12-01 09:08:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9fc0f461b25c4f114011be70d7ad4495af59222b","_number":690925,"virtual_id_number":690925,"owner":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-19 09:23:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":-1,"date":"2020-12-01 07:01:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 03:53:57.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-10-28 04:39:04.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-11-11 15:39:45.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-04-09 23:14:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-19 09:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 23:15:59.000000000","updated_by":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-30 23:15:59.000000000","updated_by":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"c633afa12c5fef3fb979bd6fc11581f721620918","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-24 11:24:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"128e5bf3c2fff5e5f753613ef280d2c84f8447e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:55:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da3957cd560b48ac8224d71ffed558ee : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45f62b6c79894026b46bb165e8785a78 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1d6e450e0f449799034f25cbf45c10 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfd224a02a00486aa756fc6d48ffaf25 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b000253aed184b6b840c22d32acfac69 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f25eab8fd69450aaa3a1dbdd086db5d : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eefce0d6be24429f8c2110b013fa5e38 : SUCCESS in 11m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3fad8a348ae494f8d05e3d9e17de461 : SUCCESS in 4m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/84ddb83e8c474eabbe3757f6d46079ef : SUCCESS in 7m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ca6a60caddfa4ac0a3acbbc848886055 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a154922d0564a019ff265090d0d2186 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a648e6346f645b190e4c29241025b09 : FAILURE in 37m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/079b752da619429e93fedfeb6b6f4a87 : FAILURE in 42m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c44801cb5f7d4ebb9460f28d8aa35529 : FAILURE in 1h 48m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1bb6887628f4484ab088b73288b61b5d : SUCCESS in 1h 54m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/783a713ec4254e7c95dc4240ec288f9f : SUCCESS in 2h 23m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/042e91f7dd5b4a86a92a297fe5704a8d : FAILURE in 1h 47m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbde95a49e5944349bbf7b99d3daf576 : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e09a478f71234d4e99696befb6708aa9 : SUCCESS in 1h 18m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41fdac8e4d25464a872cd7e53db80501 : SUCCESS in 52m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/008ea689f4a44fd2ad26f647cb20bc0d : SUCCESS in 1h 21m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15b2918168084009b95c9a2e6484986d : FAILURE in 1h 05m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/33300382d9374c9f994547371b8f0c6c : SUCCESS in 1h 25m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c77f8529ef974e9c8b715d5428531268 : SUCCESS in 1h 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"063eb72086353e12fe81b4358666492f79ffdb4b","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-28 02:58:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cbb453914c2b51c37bf56dd4512d87d0735e0da","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 03:53:57.000000000","message":"Patch Set 2:\n\nCouldn\u0027t meta be provided as part of the flavor?","accounts_in_message":[],"_revision_number":2},{"id":"b91a2312467f2bb6c35531d0e52f49a2dc6e221f","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-28 04:38:06.000000000","message":"Patch Set 2:\n\n\u003e Couldn\u0027t meta be provided as part of the flavor?\n\nYes, however, we can not get dynamic info (project id, name of load-balancer,...) if only use flavor.","accounts_in_message":[],"_revision_number":2},{"id":"cfab40c9b72aab17533376e988e22dba19bdae84","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-28 04:39:04.000000000","message":"Patch Set 2:\n\n\u003e \u003e Couldn\u0027t meta be provided as part of the flavor?\n \u003e \n \u003e Yes, however, we can not get dynamic info (project id, name of\n \u003e load-balancer,...) if only use flavor.\n\n+1 :))","accounts_in_message":[],"_revision_number":2},{"id":"14218641b592073f74eb95e65825f8c69b24e1c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 04:55:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92e0497eaf5e47a7b2403cab943a0c73 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8550ca27437447da2ecb92b8c645791 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff28bcbea274c1cb04b93408dc4d112 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1b7b1d8af3c4a37992a2b7c9e3c2631 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/973ade8d8d28446482fe811a27b8e21d : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1815c472860a44009e3dc11270d906bb : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e24f39fa0464c56adad0cd7dd76622b : SUCCESS in 8m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/33c091445e784f8292fb3dd9433290cb : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/67f8721a38a6472486e56bb841173568 : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b140e1d8b16846c69649a75ad6023b0a : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eef7d00b3d734c9e8a298dc459cad42a : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1367ada972a4da5931c48f5f838cca9 : SUCCESS in 40m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3b3446a6216c437ab877a2f7f00048a6 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef3d70ead35a46709e1c25ea8505aa33 : FAILURE in 1h 54m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a5ac2bcc5c5e4d6fa60c3bda6e8ee67c : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c153f32d721d4edf8d9bf694ebde659c : SUCCESS in 1h 47m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/95545a03f4f544cfb5c69a8c3b7893c9 : FAILURE in 1h 44m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6c281ee86f14755b06746a54f6df545 : SUCCESS in 1h 41m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27434015f295435ba2f3a3d3536fd7d1 : SUCCESS in 1h 26m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b4015e5679c459998ffa70f998eb47f : SUCCESS in 46m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6efa7651535a46779080b91db8df857c : SUCCESS in 1h 38m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58e9a3de7bcf408592938ffa4008acb5 : SUCCESS in 1h 00m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/64d5053ab23644d787f918dcd6e1daac : FAILURE in 1h 10m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f65ee6f40a0a4a12ac1854a9372d40c4 : SUCCESS in 1h 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45ee28fade59f61258fe18802d2ea09aa86b4691","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2019-10-28 06:47:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ecf2435e76729b61be5f87ab73825d7bb63e8dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 09:24:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47b7ee351b104d90b9186576b1a82c80 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b6094b3d2534f498d26c748a46acdeb : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4dfcbc360e14b08b5f4d63fa6a36bb4 : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/981fd0fedfe04a54885b48285f24f3e3 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432fb1d4e780407aae55404548fcb66b : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fa226f0493e47439c96f3b85623d611 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e8c05cc88f544989703c1619f1db954 : SUCCESS in 14m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3ad5b71852354f2583690f1c9249b1f3 : SUCCESS in 8m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/580839448bd245d4b1b9763fc97c2058 : SUCCESS in 12m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/27ac2be806f34861b64006a434e18b8c : SUCCESS in 9m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/072288f4986f41078629f3a54118fe1e : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cdc8ad0b62314c82a72e6efc6b325c24 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/91fbb13b3cb046bdab12633bf7a8add0 : SUCCESS in 42m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad974eab3bbb4cf8910577353e374b46 : SUCCESS in 1h 50m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0b740b8a51f54ce390c9ae98ed019032 : SUCCESS in 1h 49m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d4a097077cfa40fc91581b8544d27339 : FAILURE in 2h 29m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edab7cbd4ebc4328b5b635c603808e38 : SUCCESS in 1h 53m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/65410d2ac05f4f92bcbd9894f1ba5cd2 : SUCCESS in 1h 47m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a8a8e2f79f7484ba9520793d3afaf4c : SUCCESS in 1h 21m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24cfcfcafb3640ceae5fa170c5312b5d : SUCCESS in 55m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6820507ef245403988e0441789b84883 : SUCCESS in 1h 29m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/737ebbaa87e74693bfdc79534a7175ac : SUCCESS in 1h 07m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/12ca7273adc44033b65edc08139bc3a6 : SUCCESS in 1h 09m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/885f75e15daa4da79887d19aa05d431d : FAILURE in 1h 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf82a1a82f7939da0408a1014342f336ff4c87a4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-11 15:39:45.000000000","message":"Patch Set 2: Code-Review-1\n\nSo, I guess it could be cool to have dynamic metadata, but I don\u0027t see how this accomplishes that? It\u0027s still static (whatever is provided in config) from what I can tell. There\u0027s nothing to allow attaching the project_id / loadbalancer_id?\n\nI would generally agree that without allowing for dynamic data, this should simply be added as part of a flavor (which I was considering doing soon).","accounts_in_message":[],"_revision_number":2},{"id":"f355105b2f2b7ba9119395939d9b64861eac1afd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-09 23:14:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"54901676a4fb94f420c4b76d333e548781d3cd95","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-04 07:06:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4c686549622c53ae4dc915b4177802719b4a252","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-04 07:09:54.000000000","message":"Patch Set 3:\n\n\u003e (4 comments)\n\nSorry, I missed some code in refactoring. So, I fixed in patch 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe636c559eb5c17d528ec6c86ec4d70f43093e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 09:42:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f3fb40c019d4403bcc2792140c1b333 : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5534b80a2b464b2f966727bb7b859627 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3dd383323e438e9ca1d773ae59295d : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30e8acbb6f7c4a7189f0a9e0848ff507 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/398807205dae43c6b26e408fdafea21c : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e240abbc7908484da2e9b304a5d05c4c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1998a19da17f4a97bdafd0d8cd003d37 : SUCCESS in 19m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/008f7d6d832e47339b229b5986cb0613 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7d9f56938e134b09b549b036ed653e96 : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/01a04b0fe4c54e4a9f2e106faf34a953 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e16497486b0d43b7996af188d3ce9dd0 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff1cdf6869c446b3979bc37aef0ba88a : SUCCESS in 1h 30m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8ce8b70e02f442e8e74f7e25ec018c2 : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ca749139e664a639295f11c29095496 : SUCCESS in 47m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2c1ad3ceb24a4691856f4df071c0c1a4 : SUCCESS in 43m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf678721c4c949f7af9bb4f6c0d26555 : SUCCESS in 45m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/499228d199de42989c665816de285b86 : FAILURE in 55m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1c0542a59ee4e85b9b6828b7bb4f211 : SUCCESS in 1h 40m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e7e1829bfb494004bfd23ae58cf3aaa7 : TIMED_OUT in 2h 32m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/15224f3582e943adb454b613734ae724 : FAILURE in 2h 31m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/689b008e79ff4f8aa6d4602725dd7ca0 : SUCCESS in 2h 01m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c9545171ce3748ddb89789b708153309 : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dc1f3fae2f924a3cb98b8470d04119e1 : FAILURE in 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b179c51783bf1ae2dfbd9153c13c75a5dce288e7","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-04 10:24:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fc875cce6453f8983fc692973299b80d55d1bfdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 12:35:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/887f38c139724783a958aa109254bfbd : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71d523912f144f76b6071f427b6e52e1 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d170c0f15de4708924b782f3037a89a : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1fd2ef96a83451c85772cc4b665b4bb : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1cba54431774f789d8910cd3e2da3c1 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eafd685752f46f8a4a69318bde90ddc : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7911491b3e94aaa8f1adc9c075dce9f : SUCCESS in 17m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0858682650e34d1b9f1025b08489ae7b : SUCCESS in 4m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8352131faad14a1e8939aef286c0b37d : SUCCESS in 10m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c7a2e83601c24a558f8cec04b3ef8e47 : SUCCESS in 4m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1dce3381d9240cab153cf44dda292b2 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59ded76f17be462fb11aa91662343d14 : SUCCESS in 1h 43m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/275f457064784fe285848fdcecd40903 : SUCCESS in 1h 38m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/656af9f3160040e89e8e41c9750c4d4b : SUCCESS in 45m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e7386737d0b482b8e7f2564f0b0973a : SUCCESS in 40m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78c082c984b541f3ad418b45a96ddb6c : SUCCESS in 37m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/605aaeaa05fd4e20a5a52709598c19a6 : FAILURE in 56m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/198bb631c4f34118824a01bde80a91bf : SUCCESS in 1h 57m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30fe7e40c4044f83a17982653dea4b7c : FAILURE in 2h 05m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2166e5ee8a4f4238af398e865a06b1d0 : SUCCESS in 2h 02m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f8f8933385164482993d1809483e14d0 : SUCCESS in 1h 53m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e122354395b74b598e37f97d22afb245 : FAILURE in 1h 41m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7c35654a4d934099ac617211490eeff0 : FAILURE in 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f5c4496bc38528babfaa0ea940990a1d4ce41bd0","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-05 02:37:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d1e5bbf48e1835878520c4f60d919389e73efced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 04:52:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38404d29d3fa49cdbb1104fdb4b7f942 : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e72517e538c4730845363387d898c70 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e098beb3515480d930b72dcd4fe2a4c : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0565abc77340481fa8e0156166e954eb : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a78e879215f949148078f5ac6a1dd65f : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4038418f10c34d4684153d6133443e05 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8efbaaf22448fd9c303bd95505e0e4 : SUCCESS in 18m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d24c40d4477430aa65b553505c12efe : SUCCESS in 4m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a1dc87521022491bb5c66185547b11c5 : SUCCESS in 6m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3c244abcfcf45e786fe6aba891bfb35 : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9464eb77b8d74e42ae40433999cae496 : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d684f113ca98462a968444549187777a : SUCCESS in 1h 41m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15486c73b7714cf18308a6ff7732b367 : FAILURE in 1h 37m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14796fbe36c74879a82cea6b49ca02f1 : SUCCESS in 45m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1342e0896ba84a92851c2e97684116bb : SUCCESS in 43m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50ba75d886564b23a259bfd8b3f1c57f : SUCCESS in 55m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1dfcb01910194c11846c2940586b2848 : FAILURE in 49m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd1d1c1faf4449e59654a372d67e559a : SUCCESS in 1h 40m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87626604596d419ca2ad29cf7fef5ff9 : FAILURE in 2h 03m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea41257541e456f9f3acaa68e6c06bf : SUCCESS in 2h 00m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ad2f2c8a22f44be8fd00c4cfa4ac367 : SUCCESS in 1h 52m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5b3fdd96ab8b4e94a3f9b9c0a36a022d : SUCCESS in 2h 11m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4ec68d5c28464c138de0714ba728bd72 : FAILURE in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"84dd3c514b408670f79a8916be3188d4fd903ac9","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-08 15:42:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b90699d4ae2047db01cda9d1563aec1026516036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 17:58:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7bf2b6456aa4a94adaa4b0cc0f8bf68 : SUCCESS in 11m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e09f448973e4402aa25eefd3f3525e27 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3672b6f8d53e4b77b51d80dc2d4fba0b : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5245c69b394446c9ae5b9b418f0ed6b : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26040dfeb38d4f2faa0b49cb827e4288 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfd943656e224857a37bd8a99bcaa8d5 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1279da53aeb643689ac8a9422379de86 : SUCCESS in 15m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff172dd1b63e4754a48a3c02596326a2 : SUCCESS in 11m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a52fe4938740456293eb3d2cf301bad4 : SUCCESS in 7m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5d8d222098e0491c85c4f836f4be2be7 : SUCCESS in 3m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80f717ed25974f45a0c8758025853ecd : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c1a4f397c9345ae969cc23968e157dc : SUCCESS in 1h 04m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e384c212b212404092adadaaea6822b8 : SUCCESS in 1h 32m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6bc89ede04084328938ba26a8e39f136 : SUCCESS in 33m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9e679f2575564521a77bd3dd66bacb9d : SUCCESS in 32m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5760b35227344dbf937cdce78c4ad40b : SUCCESS in 35m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/238b16707b9243a3905b3fd8780b6657 : FAILURE in 39m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6efbaeaf47a24cbeb98f4cbd660a7901 : SUCCESS in 1h 25m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9228e63521504b538fa2358c2b709afe : FAILURE in 2h 15m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/07e97fc9975b40f7b209be979c70d7f7 : SUCCESS in 1h 39m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6d57a8ce45014047bdc2ef90024f1d05 : SUCCESS in 1h 41m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bbf032db008a4c70a65dbcae9ec72f4b : SUCCESS in 1h 21m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b984caa54ed34af1bae547764b540cc9 : FAILURE in 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d6e79e6f0a9b5b03281c00e458c0f6fcbf34533","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-19 04:08:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"90ff8a77a925bd8a56ed4cf6efc255466e4661df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 06:45:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42bfeba4478145a4ad6db89d435b3187 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d157ccf62e9448d2b2f7c7fdcdfcfc39 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202f6310c24744cca58e38d00bb39f41 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b5e02f1dc3740ef835fe3688e6d367f : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1cb9f0967a400681bb0471f0969d98 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a00a76a4e2b4ec4915f701fa1c4a2b2 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac21ccbb939d47589833f3b9a8f525c8 : SUCCESS in 18m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb70a634b0ef4254a5d82e837511c828 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/433c1a1c91e4489bbabc4808df49213c : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3aa88f5f52394cb096ae6614ea7b8450 : SUCCESS in 4m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e50b92e907d145e78ea2257236e027a3 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3753e565ab16441e91a5dcbe7e864033 : SUCCESS in 1h 17m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b101a6dd80844d3caa8bf6ef690a164d : FAILURE in 2h 10m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b57ed84927a44e78caf0bff53260486 : SUCCESS in 39m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f6fc244632934bb9b72826861fde3622 : SUCCESS in 38m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6163d961c0d46638e15decc9ab8b5c4 : SUCCESS in 42m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/044a870e421b4541a0fbc0948c5e6e6c : SUCCESS in 57m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b5ea596325147dc8f9d996d612a37d1 : SUCCESS in 1h 54m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b073e988dea41f992736a1a4cf9dd49 : FAILURE in 2h 28m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1193aabf4414f6587996666b5359085 : SUCCESS in 1h 34m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efbac0ccdcde4288b9bca7c46a8c817b : FAILURE in 1h 50m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d000a8998a0041d48f94b83ddd8b6fa5 : FAILURE in 2h 14m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cb10da9d3d864a00b1eb15fa05b25b12 : FAILURE in 34m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"142dfd8940b40978ebd4f52d34990e9ec2f4822a","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-19 06:48:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a3c32db682f6a90000f252245d4bef50318357b1","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-11-19 06:49:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"87e6d040d68a71d3f24c5d556fad6c9d9d1ee7f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 09:23:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a04831cc75e4d6f9502892325c862be : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e00c403d293b459087b713c8ba1e5938 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b8aaa5dc736403c9ead6a8b65799f5d : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e61399c8444185b2fe7149c45ba718 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6735c27dc986481a8a39dd0cac05aca0 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/094ba4c468f741b0815508ee53c008a2 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3623b55113cf40b78d628c23789d9746 : SUCCESS in 14m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9964cff137084036a201c84c2e36b569 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/985c738a1fd94bc5baafa85e77326a34 : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1999a82db034477a8ce1c4423072d110 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65a5f3fdd0c54bd5b28bec8f54373cfc : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2ded0086af24dc7b48a4591060b8c2a : SUCCESS in 1h 32m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/763a640c9a144569b246c88b7adc708f : SUCCESS in 2h 22m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac4a2f48e3324e23af976f8ea3cf8726 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3d75e2264382412195fdc1f4b247545e : SUCCESS in 59m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/845bde32ab794d4b8502c11122c70788 : SUCCESS in 47m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49583b7d04f040f3a78a4eeab8ff6366 : SUCCESS in 1h 04m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d199a0ff94b443e597b56090761bf5cf : SUCCESS in 1h 44m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72400e39c95b4f0eb2ee67334ccd3f3e : FAILURE in 1h 50m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1468c98d05541e88b4a6ed91edf5a2a : SUCCESS in 2h 01m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a094c997d22840acb489ee3ce4058987 : SUCCESS in 1h 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cbebe80bf4304ced8179d65d25915cbf : SUCCESS in 2h 23m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/60b6f58bec514cae97d27bfdaad5ae73 : FAILURE in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"87db8ead96c1bea5c6dacd7389f2e6568975f051","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-01 07:01:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nTwo general concerns here:\n- I would say two changes here - one adding support for VM metadata and the other support of additional fields, so I would split them in two smaller, maybe one based on another.\n- I\u0027m not sure about having loadbalancer name, ID is more unique.","accounts_in_message":[],"_revision_number":5},{"id":"9fc0f461b25c4f114011be70d7ad4495af59222b","author":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"date":"2020-12-01 09:08:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3fe107f01fa21fbd0d561c47e270196b2df38571","revisions":{"e16ad4c379c52e4346c2d7439743552507e6f0b7":{"kind":"REWORK","_number":1,"created":"2019-10-24 11:24:57.000000000","uploader":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"ref":"refs/changes/25/690925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/690925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/690925/1"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2019-10-24 11:23:47.000000000","tz":420},"committer":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2019-10-24 11:23:47.000000000","tz":420},"subject":"Support to create amphora with customized metadata","message":"Support to create amphora with customized metadata\n\nhttps://storyboard.openstack.org/#!/story/2006763\n\nChange-Id: I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e16ad4c379c52e4346c2d7439743552507e6f0b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e16ad4c379c52e4346c2d7439743552507e6f0b7"}]},"branch":"refs/heads/master"},"0b95bd42fee53c30bccda3ad6af173d9bc0ef965":{"kind":"REWORK","_number":2,"created":"2019-10-28 02:58:17.000000000","uploader":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"ref":"refs/changes/25/690925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/690925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/690925/2"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2019-10-24 11:23:47.000000000","tz":420},"committer":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2019-10-28 02:55:20.000000000","tz":420},"subject":"Support to create amphora with customized metadata","message":"Support to create amphora with customized metadata\n\nRefer: https://storyboard.openstack.org/#!/story/2006763\n\nChange-Id: I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b95bd42fee53c30bccda3ad6af173d9bc0ef965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b95bd42fee53c30bccda3ad6af173d9bc0ef965"}]},"branch":"refs/heads/master"},"80465c7aa928acef7781b3ec91a6ed932237c3bd":{"kind":"REWORK","_number":3,"created":"2020-11-04 07:06:23.000000000","uploader":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"ref":"refs/changes/25/690925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/690925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/690925/3"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2020-11-04 07:03:19.000000000","tz":420},"committer":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2020-11-04 07:03:19.000000000","tz":420},"subject":"Support to create amphora with customized metadata","message":"Support to create amphora with customized metadata\n\nRefer: https://storyboard.openstack.org/#!/story/2006763\n\nChange-Id: I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80465c7aa928acef7781b3ec91a6ed932237c3bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80465c7aa928acef7781b3ec91a6ed932237c3bd"}]},"branch":"refs/heads/master"},"1b6774bf5c953867f8eaab7b0e0d21c688a69964":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-19 04:08:09.000000000","uploader":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"ref":"refs/changes/25/690925/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/690925/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/690925/4"}}},"commit":{"parents":[{"commit":"bc8f281071d37a34e83544c22b24bbfc71514aa1","subject":"Merge \"Handle listener not found during loadbalancer status update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc8f281071d37a34e83544c22b24bbfc71514aa1"}]}],"author":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2020-11-04 07:03:19.000000000","tz":420},"committer":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2020-11-19 04:08:09.000000000","tz":0},"subject":"Support to create amphora with customized metadata","message":"Support to create amphora with customized metadata\n\nRefer: https://storyboard.openstack.org/#!/story/2006763\n\nChange-Id: I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b6774bf5c953867f8eaab7b0e0d21c688a69964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b6774bf5c953867f8eaab7b0e0d21c688a69964"}]},"branch":"refs/heads/master"},"3fe107f01fa21fbd0d561c47e270196b2df38571":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-11-19 06:49:15.000000000","uploader":{"_account_id":28910,"name":"DatLQ","email":"datlq3@viettel.com.vn","username":"datlq"},"ref":"refs/changes/25/690925/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/690925/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/690925/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/690925/5"}}},"commit":{"parents":[{"commit":"3831c66315ef679a6c8a55b7d2f0a4857955239b","subject":"Merge \"Fix load balancers with failed amphora failover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3831c66315ef679a6c8a55b7d2f0a4857955239b"}]}],"author":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2020-11-04 07:03:19.000000000","tz":420},"committer":{"name":"Dat Le","email":"datlq3@viettel.com.vn","date":"2020-11-19 06:49:15.000000000","tz":0},"subject":"Support to create amphora with customized metadata","message":"Support to create amphora with customized metadata\n\nRefer: https://storyboard.openstack.org/#!/story/2006763\n\nChange-Id: I5340eda9378b9df410f5f2ea0f9c3e8fc4d672be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fe107f01fa21fbd0d561c47e270196b2df38571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fe107f01fa21fbd0d561c47e270196b2df38571"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
