)]}'
{"id":"openstack%2Foctavia~690984","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d","subject":"Fix issues with unavailable secrets","status":"MERGED","created":"2019-10-24 14:30:56.000000000","updated":"2019-11-11 16:49:57.000000000","submitted":"2019-11-11 16:47:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690984-1573490859115-91d4b73c","meta_rev_id":"3858567b7c5b970b97ddb166dd36a185d5db21c5","_number":690984,"virtual_id_number":690984,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-11 16:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-11 14:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-25 15:29:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-29 19:22:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-11 14:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 15:29:49.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-29 19:22:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-11 14:59:21.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-11 16:47:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"094f3dc671cf537d1e5169a8b1b0770a0e5aa944","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-24 14:30:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f1690e2565f71934c54b6870774f774eccb1e98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:19:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4af8ebc69b74e1d9ab287c8ade5da67 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35fa50e038be415ab23880b488dfe3e0 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4986d396ebfa469988bdbf7db94c0641 : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa0d3fd94c4d4fba927eda79bdc24a83 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29bf25a164d949f3afb0ec95b304639d : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/affdbad7c4d64da88790f8100b648b7f : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89110b97f989435eb5c45b41f811478d : SUCCESS in 11m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/838da4906d8943a49251b7d2f8ac762e : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f908d78814f0478e822e65139907fc89 : SUCCESS in 8m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7e23ae0cf238487bbfdc97687b7f8368 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f27ef8a1e4504fcdba64619b11a64101 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/539c2e6e8cb747b28d8ef15d0e5aefba : SUCCESS in 55m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/25fd6b5c07854d7e96c367803fbb5ebc : SUCCESS in 52m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9bc2b31344b463b9f5abaa737ffe2d1 : FAILURE in 1h 44m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/73ed85acc26f4ffe827c5444268939d0 : SUCCESS in 1h 41m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fbcc7e780ae54c4f936770f508f889bc : SUCCESS in 1h 58m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05816cd400b549d8a1cfa63bde87784b : SUCCESS in 1h 31m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7451c4aa9ffc4f7b98d1ec96e5fe4066 : SUCCESS in 1h 27m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eb4ec34642954a23af5dc52ed8ea566b : SUCCESS in 1h 24m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/067f6bbf9e32428d86f0f3cb96ff4944 : SUCCESS in 1h 12m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d777b347862344cea6af0b959b3cc6a7 : FAILURE in 1h 40m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f1482aa91b54173809e44a2be802f38 : FAILURE in 1h 22m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/af36a2330df44aafa57b664a91983909 : FAILURE in 1h 14m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b850ecc3d0fe4e2595ac32fd22d1e5bf : FAILURE in 1h 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90112aae9094a73efa611932a995f0ec3a23e371","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-25 13:12:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d142fc154bedeaa39d1c26cbb30a5126dd6709fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00b1d0c2743a4a21bf62e180f9d57830 : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88ec1f5b58734d73a72eaad14c97394a : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b945d71cc84343feac400a1ab1c60697 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/033809006de949709b25bbbbc78b1015 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4106ea37ef1a427eb83268c54e7c2406 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/000d6c71011148d3b6c9f5d043a38e8f : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bede759d0aeb4ba396312c385b483f6e : SUCCESS in 10m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b670ee00ce2445518263282f95e50724 : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a3738f5e31434481a3ffabcadfb364d2 : SUCCESS in 6m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e42ceaf23ea24e7fae32b16e563145f9 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e0ee42203da5457390d2a882d578f521 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aaf79e85fdda480ab6510b6a17869c70 : SUCCESS in 50m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6a1d37c52d8e4c1fab061193c252067d : SUCCESS in 42m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2ee7af1231343a186eb8b9effec8dc2 : SUCCESS in 1h 44m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8b7c964508a943f9b269cf0b0e85cf6e : SUCCESS in 1h 40m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/caaa9107e7d643b1a31264df7f4d1d36 : SUCCESS in 2h 14m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8cd1426f91641e8a0ada377c8e6c8f0 : SUCCESS in 1h 50m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c283c2f6d28d45f6929cb38d1d703bd9 : SUCCESS in 1h 53m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e74ab6f9922d416cb120f84d5c6b6111 : SUCCESS in 1h 27m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86e59ade5eaf43aa95e6faca97c51239 : SUCCESS in 50m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8ea74261aa384a91a1abf2f290b3a8b9 : SUCCESS in 1h 36m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69d8c21b53634584b5520d26cf347a9c : SUCCESS in 1h 03m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7c1dfa9f9eff4ccbb2b28629449d9797 : SUCCESS in 1h 12m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9a289a4e2a3e45c2b392d0bde67f659d : FAILURE in 1h 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30065902ac62738396731cf3783c41b6000a7451","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-25 15:29:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b8ed5991f64e15376d6557528c29aac73d9c1af","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-29 19:22:20.000000000","message":"Patch Set 1: Code-Review+1\n\nWe\u0027ll have to talk at the PTG if you want me to help merge octavia stable changes as the privs cross all repos, not just the neutron* and networking* ones.","accounts_in_message":[],"_revision_number":1},{"id":"02df3ad1c9e95d957f2384b34577dacb0b709674","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-11 14:59:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9b88b2ec7f01215f65f3937317e14e013ba598c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:59:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbfc97a90624d7f331f60bfe387bf28119d8f87f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:47:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01134aecc25049d881f45c945a331ece : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73fbaf1ed80f44cf8a99d44c96790bb8 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57978b15430344ab868dad451214f3a8 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9795b3a0424d2e998633481447a927 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33ca95ea6b9140969048e38da6ee7ee5 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b9df8f9bce84b0b887fb9214b2def41 : SUCCESS in 10m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0039d800508f4ba4aa5476e27813a3cb : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49ef10aab30e4ceb8c272c8150d4b919 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31f4d1a085194cd8a605f6de167d6665 : SUCCESS in 51m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/58b98566057d4415ae9bd0f6339fc1c1 : SUCCESS in 47m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/475baf6ec319447582da092cd29bbfac : SUCCESS in 1h 32m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0af0a6b609544e998b30338fe95163b4 : SUCCESS in 1h 41m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cbd9e8f64e974defa5fef6927620c001 : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"84cee1f37460f903a560e89cfdc979742f7eb8e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:47:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3858567b7c5b970b97ddb166dd36a185d5db21c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:49:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1725292ad6104f318938e1278b5e7f30 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fd3f6ef64ffa4e843a8e41b33c7d6171cf7e49c","revisions":{"2fd3f6ef64ffa4e843a8e41b33c7d6171cf7e49c":{"kind":"REWORK","_number":1,"created":"2019-10-24 14:30:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/690984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/690984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/690984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/690984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/690984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/690984/1"}}},"commit":{"parents":[{"commit":"6f54167b05cab1ef2721d63ddee99f92d61ea14f","subject":"Fix urgent amphora two-way auth security bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f54167b05cab1ef2721d63ddee99f92d61ea14f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-11 23:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-24 14:30:56.000000000","tz":0},"subject":"Fix issues with unavailable secrets","message":"Fix issues with unavailable secrets\n\nPreviously, if a secret became unavailable in barbican, the user saw\na generic error raised from the barbican client and was unable to delete\nload balancer resources. This patch corrects both of those issues.\n\nChange-Id: I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d\nStory: 2006676\nTask: 37012\nTask: 37077\n(cherry picked from commit 7fcef817eca3ad177460ee6976633f3dbd078cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fd3f6ef64ffa4e843a8e41b33c7d6171cf7e49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fd3f6ef64ffa4e843a8e41b33c7d6171cf7e49c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
