)]}'
{"id":"openstack%2Foctavia~691469","triplet_id":"openstack%2Foctavia~master~I8cdc7eac65be4b3fa1fe444c769a51ce60469263","project":"openstack/octavia","branch":"master","topic":"fix-URL","hashtags":[],"change_id":"I8cdc7eac65be4b3fa1fe444c769a51ce60469263","subject":"Fix wrong urls in Quota API","status":"NEW","created":"2019-10-26 08:06:33.000000000","updated":"2019-11-11 09:30:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b7fd66fdd87989052dac5c87b9585226e30d177","_number":691469,"virtual_id_number":691469,"owner":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-26 08:21:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-11 09:30:01.000000000","_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2019-10-28 14:19:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-26 08:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 14:19:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9bc384378835803b35e72aa834a9df42ec93e2ad","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-26 08:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f88abb4fedd0e6dc1a836bc1f04e3b73291a6580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 08:21:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/684668eb1ea34f209f6368257f86c350 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7c50974f384aa1b0fe1a011a2e05e0 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12166cceed314231b5b2fe2546d5098c : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9fb9db04be4579da3b71f66f8a4b97de829daca0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-28 14:19:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think there is some confusion as there is no project_id in the \"get quota defaults\" method.","accounts_in_message":[],"_revision_number":1},{"id":"8cb5dd4ab2d3d71fb04d277e06a978a3dc734fe7","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-29 00:52:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4edf5e031c49320ab759b0cc99eafcfb5930582b","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-10-31 12:56:22.000000000","message":"Patch Set 1:\n\nPTAL","accounts_in_message":[],"_revision_number":1},{"id":"a02478a3ce7a0272a78cdaf788a3143080449c3d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-04 17:06:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"760228f7ef3a52f300d01429041365b114a9a995","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-11-11 08:34:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e85b011c0d1dc53b629f514e6931ebe7adb22ac","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2019-11-11 09:30:01.000000000","message":"Patch Set 1:\n\nI think there is some inconsistence between doc and code in the \"get default quota\" method.One defines the default quota in deploment scope, while the other one defines it in project scope. Either looks good.\n\nIf we think the pricinple of defining in deployment scope used by doc is what we should follow, then we should modify code. And vice versa,\nthat\u0027s what my change is doing.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e5a058fc2918487daa207803e4cb070883f2073","revisions":{"0e5a058fc2918487daa207803e4cb070883f2073":{"kind":"REWORK","_number":1,"created":"2019-10-26 08:06:33.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/69/691469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/691469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/691469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/691469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/691469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/691469/1"}}},"commit":{"parents":[{"commit":"93a7116b547e7861b2bb93a362cc0f7e2787b6db","subject":"Merge \"Fix some plug.py unit tests that broke on OSX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93a7116b547e7861b2bb93a362cc0f7e2787b6db"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2019-10-26 07:45:11.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2019-10-26 07:45:11.000000000","tz":480},"subject":"Fix wrong urls in Quota API","message":"Fix wrong urls in Quota API\n\nIn the Octavia v2 API implemention, the url of getting project\ndefault quota is \u0027quotas/{project_id}/defaults\u0027, which is wrong\nabout \u0027quotas/defaults\u0027 in API doc, So we fix it\n\nChange-Id: I8cdc7eac65be4b3fa1fe444c769a51ce60469263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e5a058fc2918487daa207803e4cb070883f2073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e5a058fc2918487daa207803e4cb070883f2073"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
