)]}'
{"id":"openstack%2Foctavia~691693","triplet_id":"openstack%2Foctavia~stable%2Fstein~I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d","subject":"Fix issues with unavailable secrets","status":"MERGED","created":"2019-10-28 15:28:13.000000000","updated":"2019-11-11 16:49:57.000000000","submitted":"2019-11-11 16:47:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691693-1573490865591-9a54f4b4","meta_rev_id":"e17a127dcd6a0d49e142b585a74e94b1e477166e","_number":691693,"virtual_id_number":691693,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-11 16:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-11 14:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-08 02:08:19.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-11 14:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 02:08:19.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-11 14:59:52.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-11 16:47:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd68013ef680f60822fec8d09e28f8404f8b52e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-28 15:28:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba94695a33bff9c09527a02e6f7c068a4427f8c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-28 15:36:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9c297748eb5b5ac29102af96b1917cc348548aa5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-28 15:48:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67e9814a24a4ba5be539abe8e8db73650475fa25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:22:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d020ffc961094d61990d92ec65b1336f : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8ffc78aac364a1a80c033ae0e819b98 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd342958ba2452888edbfa7f80291dc : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd38070f23a345f9b2f5144ea4a9d266 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be266693860a440089a4844612a8437d : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6fd32a700c2459f9243da006abd76a9 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffafa02e71c14d2abac27112d3155f08 : SUCCESS in 9m 55s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/900346a3921a494dbf0f4d301daa21e3 : SUCCESS in 1h 25m 07s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/84ea15be5b664eb5b5ad5b8ad7aa6986 : SUCCESS in 1h 30m 36s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e51188701d6c46dcbad7f967c08d0856 : FAILURE in 1h 44m 00s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c728a867e90046ec824d9b0ec59401e8 : FAILURE in 40m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c33fa7616b44ccab209c15a84232861 : SUCCESS in 6m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c9a01a354f1a424db7dacd2c42a54183 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8db6d50acf724fc280e329372fa21000 : SUCCESS in 7m 44s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a51906422d340979b4af64bc10cded4 : SUCCESS in 1h 15m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d7814dd675ef440292b32db73d060de6 : SUCCESS in 1h 05m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/923bea6d876f4a198a54370c1e6b21c1 : SUCCESS in 49m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0c3175877ada4111841beb005d50dbcf : SUCCESS in 44m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5934b778289941eb9fba4873b27ecf9b : SUCCESS in 1h 43m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7df3fe9dd79c4be8b1ad243fe276a6b3 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4ee7e68aab0a4696a177682ac94e2b93 : SUCCESS in 1h 42m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2e864b4a615842c9828d3dabeb6c4845 : FAILURE in 2h 25m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a875a1f99a7c4087a27a41ce7da5d1d4 : FAILURE in 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac9518c51f2a4d26be305aa8381a00fe : SUCCESS in 1h 09m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac4891445b6f4e178a2a40811af8eb37 : SUCCESS in 1h 21m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2f60c2d26afd4ebebd89725f6b072857 : SUCCESS in 1h 03m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/413f5f85621744e8b6a75601ddc3a66c : SUCCESS in 49m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b8c9a60afba17fa125b648c8fe7a8fd47a5b582","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-08 02:08:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"91c3dbfd451c90a9b3b9610dd33b09e047a5316e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-11 14:59:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e47393de41292326b5fce88e27072200061ed35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:00:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01452f45037f38daab7a49990399d43261c50050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:47:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a5a883fc1984dcdc03f5ba7ef603c17af0c39ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:47:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17bb79f9dbbf4ac0996faefaa1a0dd07 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2452f0b97742b4a8b4966007e35c75 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/420f505f9e8b45419664996dd0091a0d : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3a8daaee7c4247ebadbccfc665227ab1 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8346d9cf01a40228e1d8a0af98d3f5b : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3348b2480a4ff0982d6a32b4cdbc43 : SUCCESS in 6m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/772460e2a25d434e8339cfadbf71a0d8 : SUCCESS in 6m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/28ba3b490f8743699ec26d67b15d3cf6 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e137e21ef774c2fbaecf1c829895b3b : SUCCESS in 6m 02s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23533801791841afa573112166b1650b : SUCCESS in 1h 06m 57s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e73d64adf2114c468f499077ddeb25b1 : SUCCESS in 1h 02m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1a31ed1dfc4e4d56a91a3a82671991d8 : SUCCESS in 1h 13m 23s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ab498b503fce4abe8115eedda6f70b68 : SUCCESS in 1h 20m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f4c40d6bb7c4ab98cb4731669d4de6f : SUCCESS in 43m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a0da405ebac24dd6a97002e477b20d75 : SUCCESS in 40m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c615608e47544df8f6536a095406fff : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/299c71406e064b5893099d57f59e6864 : SUCCESS in 1h 27m 31s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/49da7600161140ef813d826cd4d876df : SUCCESS in 1h 37m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a374d901465c45a1acbcd09520d2ae2f : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e17a127dcd6a0d49e142b585a74e94b1e477166e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:49:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd7b9c2ee394fd28d53c428569a138f : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bfa553925c22b637a11b4cbf567919d0edcb9203","revisions":{"8abc249ff309ea24aa17653e08fd2788b9236129":{"kind":"REWORK","_number":1,"created":"2019-10-28 15:28:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/691693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/691693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/691693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/691693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/691693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/691693/1"}}},"commit":{"parents":[{"commit":"e51752e43305cc2d7446aa4af89378878708e73e","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e51752e43305cc2d7446aa4af89378878708e73e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-11 23:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-28 15:28:12.000000000","tz":-420},"subject":"Fix issues with unavailable secrets","message":"Fix issues with unavailable secrets\n\nPreviously, if a secret became unavailable in barbican, the user saw\na generic error raised from the barbican client and was unable to delete\nload balancer resources. This patch corrects both of those issues.\n\nChange-Id: I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d\nStory: 2006676\nTask: 37012\nTask: 37077\n(cherry picked from commit 7fcef817eca3ad177460ee6976633f3dbd078cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8abc249ff309ea24aa17653e08fd2788b9236129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8abc249ff309ea24aa17653e08fd2788b9236129"}]},"branch":"refs/heads/stable/stein"},"bfa553925c22b637a11b4cbf567919d0edcb9203":{"kind":"REWORK","_number":2,"created":"2019-10-28 15:48:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/691693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/691693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/691693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/691693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/691693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/691693/2"}}},"commit":{"parents":[{"commit":"e51752e43305cc2d7446aa4af89378878708e73e","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e51752e43305cc2d7446aa4af89378878708e73e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-11 23:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-28 15:47:54.000000000","tz":-420},"subject":"Fix issues with unavailable secrets","message":"Fix issues with unavailable secrets\n\nPreviously, if a secret became unavailable in barbican, the user saw\na generic error raised from the barbican client and was unable to delete\nload balancer resources. This patch corrects both of those issues.\n\nChange-Id: I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d\nStory: 2006676\nTask: 37012\nTask: 37077\n(cherry picked from commit 7fcef817eca3ad177460ee6976633f3dbd078cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa553925c22b637a11b4cbf567919d0edcb9203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa553925c22b637a11b4cbf567919d0edcb9203"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
