)]}'
{"id":"openstack%2Foctavia~691748","triplet_id":"openstack%2Foctavia~stable%2Frocky~I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d","subject":"Fix issues with unavailable secrets","status":"MERGED","created":"2019-10-28 20:22:32.000000000","updated":"2019-11-23 00:08:51.000000000","submitted":"2019-11-23 00:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691748-1574467658077-29596582","meta_rev_id":"e0461e6dfcf681a54cfff5887529c312908a9cd7","_number":691748,"virtual_id_number":691748,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-23 00:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2019-11-15 15:17:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2019-11-18 15:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-22 22:28:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-11-22 22:28:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 15:17:41.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2019-11-18 15:47:30.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-22 22:28:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-23 00:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"927f35f5be99a174d8da81ae782aac3f619552e0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-28 20:22:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2184c95613087ec574c66570a7c9155de3e5b458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 22:20:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6054b0d9a0c249a4aac7f3b3c165e97d : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d4ec3a908f429c8bc9b7509cc602d4 : SUCCESS in 7m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67549445a0514e89aac6521806d5d15e : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61b26e2016af4d6194f3a407890a01a0 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f306ecb504324e808d459f60f05a406d : SUCCESS in 6m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/da13bb5376504ed3bc15006a2b33e2ce : SUCCESS in 1h 11m 16s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/00509f18a52b4b86ad0122d372eea6a2 : SUCCESS in 1h 24m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fb2674fd65d424cac047b5b5129931d : SUCCESS in 1h 34m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7e15edbaaaf54eddb2e1162d9a160ca4 : SUCCESS in 1h 23m 32s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6d8f2931809349c481c30a387842a888 : SUCCESS in 5m 01s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4d66ea0b89094b6aa6731882f457abad : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5578ace651a141eba085824e56e21946 : SUCCESS in 7m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6d0d76f527a41d9a654fef23e17c495 : SUCCESS in 1h 06m 04s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3708b617f17247bc89a1ed593d4f42a1 : SUCCESS in 1h 00m 38s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/999af5044d524447ab6d6f14ee9e252c : FAILURE in 1h 27m 17s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/2bb99c810a774d54b5cc3f03eb9d0cb4 : FAILURE in 1h 29m 00s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/dc42f6ae0b3c415e97d93e92d7f2bcfa : FAILURE in 32m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9641f9a6d9164a24b39120f78d137103 : SUCCESS in 45m 11s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/3ca0a4b772a6440782949265a3c94e8c : SUCCESS in 43m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ee0aeaff50741c1ac1ddb901ea6d82a : SUCCESS in 1h 51m 54s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b762853190dc422d8cef180049125da9 : SUCCESS in 1h 28m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1e683459d854a87ac9d3f095131dda8 : SUCCESS in 1h 16m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56dfb0ca49f44c76839f7691ebfeae0b : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"316e04cf64f1d2ba8daa67246069ac7e6ed2942c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-15 15:17:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b059a8f1f1d95dd8fdb19003262832b440c017ab","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 15:47:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e74a267c262b50ca052a46e5c65b2f7b763b8582","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 22:28:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58e1f3d98663e426e08170781968d6df9d677bbc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 22:28:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"944e2b5dfec8f367b4723617b042cc49d122a3fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 22:28:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8011edd74b13046954a54e8ba7fa57780ab653d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 00:07:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60bc0ed38ea544258807499379629107 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b31e396d1c0b46daae0ce79eaddcdbcb : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00208b92cf0d4cf0b59fd5fddd8922bf : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a419b1f4d4a4704890e78bf40f2e75f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2034d3207b44041a3cde4a177b00e4c : SUCCESS in 6m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/64dca7688e13465cbcb2a9bada37923b : SUCCESS in 6m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/edf34812fb8c4926a242df7cc898232f : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6b6ef1a8ee3438993cb1d8f286f3957 : SUCCESS in 5m 21s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69639f3ba247425aa639aae99361ea03 : SUCCESS in 58m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/27c692e9f64c49289b5ac7518eb5f116 : SUCCESS in 1h 04m 13s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b93cab53c3114cd19137ce7367153d44 : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/068a251cfeb94c28ba647158be98eac9 : SUCCESS in 44m 18s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/60aff9e9c2ae40c481bb7401794f2b6d : SUCCESS in 47m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0b64bbca334403c8270d8596c0b990c : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/043493076a064ccf8d13ebaaa1a6d0b7 : SUCCESS in 1h 34m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63b921eca118420fa514784bb03b1f69 : SUCCESS in 1h 06m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e105b5dce05342ffa61cf51c0074efa7 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d74414062ab7cfda02037dd1d82de7956c103ec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 00:07:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0461e6dfcf681a54cfff5887529c312908a9cd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 00:08:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5194db1836940068b2aa84017715f9a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f68460ddd31f9b09d59fff876f103324078473a6","revisions":{"f68460ddd31f9b09d59fff876f103324078473a6":{"kind":"REWORK","_number":1,"created":"2019-10-28 20:22:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/48/691748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/691748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/691748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/691748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/691748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/691748/1"}}},"commit":{"parents":[{"commit":"43d3e1bde3e1257b7e3622eb222d4ab6a47114cf","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d3e1bde3e1257b7e3622eb222d4ab6a47114cf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-11 23:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-28 20:22:31.000000000","tz":-420},"subject":"Fix issues with unavailable secrets","message":"Fix issues with unavailable secrets\n\nPreviously, if a secret became unavailable in barbican, the user saw\na generic error raised from the barbican client and was unable to delete\nload balancer resources. This patch corrects both of those issues.\n\nChange-Id: I97bd1b2a97a44d0a0566ae56167fa4f2e503ee2d\nStory: 2006676\nTask: 37012\nTask: 37077\n(cherry picked from commit 7fcef817eca3ad177460ee6976633f3dbd078cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68460ddd31f9b09d59fff876f103324078473a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68460ddd31f9b09d59fff876f103324078473a6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
