)]}'
{"id":"openstack%2Foctavia~691987","triplet_id":"openstack%2Foctavia~master~I96908e6cbdb523f95298aff463a749d15e98e1ad","project":"openstack/octavia","branch":"master","topic":"fix-client-auth","hashtags":[],"change_id":"I96908e6cbdb523f95298aff463a749d15e98e1ad","subject":"Fix update API when barbican secret is missing","status":"MERGED","created":"2019-10-29 21:18:15.000000000","updated":"2019-12-11 16:54:48.000000000","submitted":"2019-11-28 09:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691987-1574934360053-c4d0231e","meta_rev_id":"d47ba82d84e201f54ccc1e741009ddbda6ffd5b0","_number":691987,"virtual_id_number":691987,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 09:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-27 18:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-27 22:45:04.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-11-27 22:45:04.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-27 22:45:04.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 02:01:20.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-13 06:18:34.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-13 08:34:35.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-19 01:51:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-26 09:56:17.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-11-27 18:32:58.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-27 22:45:04.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-28 09:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcc1c067bac850eb4760044ab736a675289304cb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-29 21:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eafbc876e0f1f812d9eb21358a5baf775b29b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 00:00:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37a43d9e70264bfc8a88a6a7f982abf7 : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62eb61ea5663405cba77c5c5be2b12bb : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a6709e8cc348b3b079fd55fd353a66 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02b1039a72f84e999d65f05c5c768a13 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21270ed476474a04b99ce2dd00875876 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/288649f4ae7943578e301545ec13b253 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e92d34499bb4ac3b8bbb2a77ede2119 : SUCCESS in 9m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e27cfb7fce7643fdaf1bdf934fce4a8f : SUCCESS in 4m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/47d7114f1dc34251b0c5c38a0923db7e : SUCCESS in 6m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/37ca16f76b1248ae8cbd1489f47a095a : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7966340209d44249608177fd2f5d55c : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/41cddb91cbcc438ea79a9e99ed7db712 : SUCCESS in 43m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e3bc6d1f98dd425c9100f5cd640ad89c : SUCCESS in 39m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ae5db5137054f3db76e9f42901022f9 : SUCCESS in 1h 38m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7531bdcae8a74fcd8c653e23cb604948 : FAILURE in 1h 35m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/82780be0273e4e3e9df6f6f2eb562a6e : SUCCESS in 2h 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5d18ba9815c4202b67fc04967f3eb7e : FAILURE in 1h 44m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf02bf6b7de648b9981b735e8eb3dae5 : SUCCESS in 1h 28m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95597bb5382f4895bb936d94b82f6aec : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a41d3821a0a4de798a660e0b4b5aa94 : SUCCESS in 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0890d03c1b3f43398b56ef7fd668374f : SUCCESS in 1h 49m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/979ac6c359fe49c39f9a1ec8e071975f : SUCCESS in 1h 04m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cee775d6bc1241a2b995dea958f934d3 : SUCCESS in 1h 22m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9169c709dec947998099efa19e0cc09c : SUCCESS in 1h 33m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20b9f0f4cd67c0881e90d2812d9410292a0f8f5a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-31 18:46:50.000000000","message":"Patch Set 1:\n\nrecheck\nNetwork troubles","accounts_in_message":[],"_revision_number":1},{"id":"7063fb9a3f7e98d212ebb19bab16ce55b3407834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 21:54:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18f6c1227f2e4e608faf30f945b66ef0 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72fc4b75cea7422d9856b451f5a9a988 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7376147f7f8344cf8f2aedc3f334acf7 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de9ba0599d5347fca691c092b6b508ad : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb45f94b3e0c48ec8724cbdfc37e05d0 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/949ca72c03d7432bbd94cd6289b0e67c : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f91c6d444f8448c8c7cd53561c20171 : SUCCESS in 9m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e8febba693194fd985970215d273b9a4 : SUCCESS in 4m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/15f7afa5dbdb41809162896c7483dd2a : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/10e1063773724fb5978109809b7b95b1 : SUCCESS in 10m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2009869fa9484df0835e33e64a138abb : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee135febd85743f69db7dfe514d244b5 : SUCCESS in 48m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/29f0ed81ae6148eca38d135dd0e563d3 : SUCCESS in 46m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7f767ab21e944a8bd917a35655f1685 : SUCCESS in 1h 19m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/264223080c4b41b589ee86bcaf2b734c : SUCCESS in 2h 27m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5414d62c8415409fab7c621855c5e1a4 : TIMED_OUT in 2h 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1253bf5dfe940e6877f71517ffe2907 : SUCCESS in 1h 19m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/80f9297485e04cc98ec62e43f5be0f20 : SUCCESS in 1h 13m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7110617726b49d7ab2890d1afc2d807 : SUCCESS in 1h 13m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97cad62ed69b4bed83e06a513b0164e5 : SUCCESS in 44m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef1361227faa4ec4ad603c4692a7499b : SUCCESS in 2h 11m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31689c710eca4a449f633f5a8a1403ff : SUCCESS in 1h 05m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4a889435901545be8e37f05b49b51676 : SUCCESS in 1h 09m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0940d4839ec7434bb5057a480a868ad6 : FAILURE in 1h 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9df629f4bb9bc4d998dc7955bf2b5188f6dc35ee","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 19:52:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6d9633b3142646911ffbce5bb5ac48c7786b1133","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 20:07:29.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"99ad554d3a3493d6ab84d7121a5fcb97fa6b01c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 22:30:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c146d879b3c4a849345208a6d657fdf : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8157581cdfa946818c0d443028fc34c4 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6501f79ecc594130b1a1fe48fe601d7b : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/644ee8ae9c6c449296a5f144632658d9 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bbea1d47a14408cab53c9ce19379b03 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a066f38e1fc445ebb4701b7598f6feef : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec837294e924534a42d08b7f00d179a : SUCCESS in 10m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c3e6aab1597b4f988d6c5b90397bbaff : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ef7c0eda3379404ebb67c3132a45d51a : SUCCESS in 8m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/831bb9dc509d4051bae6f21828a7d56e : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17471fc1fa664933881ea18163231b83 : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db67f40cfa27439b89fd2714a59ae396 : SUCCESS in 48m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/15c013ef3e4b4645bd19eccbd2f12c9b : SUCCESS in 50m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/204e012bb40748e5966616e2b8103477 : FAILURE in 43m 51s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/31735575e172484c8b388169a6a33d81 : FAILURE in 1h 23m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/879aabcdfd0946cc8a3f9c320732fc8e : TIMED_OUT in 2h 35m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0caf533877c8436897003a16cc588219 : FAILURE in 32m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf29ac357e344610b43ce4ed525255c4 : TIMED_OUT in 2h 33m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/547e99d59d694dbeb27e76e3ca45dc76 : FAILURE in 52m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/011a048584db4b6f936714489627ae5d : SUCCESS in 59m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cc6ca658c674487bba3e25ee40f337ed : SUCCESS in 1h 40m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/819acf8032c6400caa757c3e2f95dda9 : SUCCESS in 1h 14m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ef91abef1a5b4ac394eeb066991ff0b0 : SUCCESS in 1h 12m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/24d4cfe1fbc74e7b81d3f8429054ad71 : FAILURE in 1h 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc49eee69f86946cddb69686c5e82ae8c6c99d1d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 22:31:32.000000000","message":"Patch Set 2:\n\nrecheck\nInfra issue:\nkeystoneauth1.exceptions.connection.ConnectFailure: Unable to establish connection to https://192.168.48.135:9696/v2.0/security-group-rules?security_group_id\u003da28312f4-886f-47cb-9032-60d7c9fc4a0d: (\u0027Connection aborted.\u0027, OSError(\"(104, \u0027ECONNRESET\u0027)\",))","accounts_in_message":[],"_revision_number":2},{"id":"276c56201bb8df264e257a9a0675a683d9880cee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 00:26:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6a3bd1f61f84c839416de986ad6ed9f : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55d44f57834649db86d1f5dafc0bde1b : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89c4a86f9c143d19e8d8976d6301cec : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d93fb2a2f3d94a9b824dcfd475a25a5e : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3dee176647a4f64847214db0eacf11b : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e24fa5e59554fe4a7a5343d9fbdfebc : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f312598fbb0d4608a24ceb8eee9e1e2e : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1b82b452e6404fafb09ff69a9bdc536b : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ea878e9801804a44bb3381c63424eff1 : SUCCESS in 9m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c06c4ae58a5f4b88a00537c724bef978 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7c7ad39db3a41278dbf0e35c378f444 : SUCCESS in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7f8b18886c548fbbd30e0fbb0a94141 : SUCCESS in 53m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/badfb56f714e47fd9eac9a05b2b7e77f : SUCCESS in 45m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/108caf9994634dbdb9fc7adafdb19dbe : SUCCESS in 1h 32m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/eb2e0f3f4dd94ca199433e6458ab194d : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e89122af508a4e8294192dae165c5294 : FAILURE in 1h 54m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aabcf42f97c444fcb6f3934b92f3308f : FAILURE in 40m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f91e22dfae3541e09551006b60e262f8 : SUCCESS in 1h 15m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/31c0de1860ba4ff8b429374612e50ca4 : FAILURE in 1h 24m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02854833f1274393b24268a1107f4cbf : SUCCESS in 48m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3a817abffb06480183c9d6652ad85aa2 : SUCCESS in 1h 34m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ced29a8955fa4aba9938b029ca764e7a : SUCCESS in 1h 00m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b384b7f0bf7b48b38d9f6c37d73711e1 : SUCCESS in 1h 14m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/36822b8f007f40d38241bcf710b7e164 : SUCCESS in 1h 17m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d9f2c847570deba8791c86472bbf04529225764","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-09 00:28:40.000000000","message":"Patch Set 2:\n\nrecheck\nnova failure:\nsetUpClass (tempest.api.compute.servers.test_server_addresses.ServerAddressesTestJSON) [0.000000s] ... FAILED","accounts_in_message":[],"_revision_number":2},{"id":"4465596d050140527d244550830acf4dcb723a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 03:02:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb569f71dc814d9faa49f70cefcc1541 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d403b75cca7049909de40cfd0655b9fd : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9374f42c6cb441fdaba4da3886c6b42d : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f3625a73a734a35b9149782fb6293d0 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/defdd4bf65124c11891e5b2bc1e7507e : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04a47f88ec5841f5b7d0c623b185a6bc : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314c0c91207543a4a03495785b1a826a : SUCCESS in 9m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a15d30a13b134ef7b01989f319bc1951 : SUCCESS in 4m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8e180774283b407f88197fcddcc82aea : SUCCESS in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7a713bdba8ff4e74a77322f4e96bce79 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/612d1e7ade2f42488e03fefd0f708835 : SUCCESS in 5m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae133b78e5f14c3882847a108f4c9d19 : SUCCESS in 50m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0005064b15264a539fb8bf422a3715e0 : SUCCESS in 46m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e861e8658c7a4d639c297017fbc1943a : SUCCESS in 1h 24m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1eaf20ba2993448e8d361384092ec94a : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4c5a0b838ec54194ba489e1dae74be0a : POST_FAILURE in 2h 27m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2becc9d316b34b83902a25a82a69b902 : FAILURE in 32m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f590839484e54d198dbb19ff2a940837 : SUCCESS in 1h 11m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4719d3c28403426a80c1d8a2f6a54991 : SUCCESS in 1h 15m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7defcd4923854accac7c22b5075ba647 : SUCCESS in 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a1c0a88379b4e4b800f780d5d24ec83 : SUCCESS in 1h 33m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91b284233a034354a050fbc3de9e7a82 : SUCCESS in 1h 03m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/80a8869eb93845dea246b31617666a08 : SUCCESS in 1h 03m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ae16e5c459684e3caa0b58fa9e505f9d : SUCCESS in 1h 08m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3fe51bf516bd10f0a12725a090429c1aaff5c32","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-11 16:31:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c39afcd1e52acbd2b65a951d67a26a9406aa5916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 19:11:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4f09ac80d344a66a9e92e0780f21f96 : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d9f2500d37c478dbaedd8d23a4793de : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a63fc65e76f434eb9b0e58c646ffc14 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/873792e206084a0e86131ccaf31838f2 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12eee864595c4c0bb2b5914189abd5e3 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/127f9fa70fad4e118df237471291d17a : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad98c78709949618a16896b014507e0 : SUCCESS in 10m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/008459093659473587ef46bf9da168a7 : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5a245619c1b74b7ea8e8d4ffcff45fe7 : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69a43dd13e424162a3a4d9fc62e5e53e : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1a93bbbc9b848098795c96cc5964bf0 : SUCCESS in 5m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fa59aceb063c49da98d7ed2aeea9e883 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/db7dc9196a9d440e9e6998119ae71d32 : SUCCESS in 45m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edfa2962a8234bf8afe6c7110d4d6b35 : SUCCESS in 1h 44m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/691bc8ce21b94e0cbe5f809bbda3a2aa : FAILURE in 1h 40m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a3524b75e58140d08d3a0c11f0d65881 : TIMED_OUT in 2h 35m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2cc0d86467794ae6b6199a1702bf7580 : FAILURE in 31m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/091156a252d54e40bd3df6ce92479be7 : SUCCESS in 1h 58m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/442763548bbe4788a61893500998609f : SUCCESS in 1h 19m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5873520e60a846258f4a16897566347d : SUCCESS in 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f23d62a4f7274b4eb4ec6bd31f941cf4 : SUCCESS in 1h 37m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a12f219938194fcebfe11bb3238e2b7e : SUCCESS in 1h 12m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f63ad612f1fd4524a93b2dcc852bbcf5 : SUCCESS in 1h 06m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/24a7291811a745208bc1821ccc09d51a : SUCCESS in 1h 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9202c3f72b64348bbbce286b444a7c25d80466ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-11 21:33:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"29bfe5a86ce5c219b0523be9a9374d674073dc5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 00:11:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/684c47ef990b44d6b5b8a6dc2ceac9d2 : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ace1dcbe0554f8f94f92ff337fbad26 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb23d41ccb847cfb7ff5c849680ce56 : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa1c42875c834c45ad348a204413bbcf : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4421ecf9719a404da7ffd40a05ee7533 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6195eb1705b24562b8f2db8509c8d46e : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b85934e5b94ae2b34a20b0e100acc0 : SUCCESS in 10m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd0e850e732b4ddd8304fea18deca6ec : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4a174bdd9a154ce1ac7be95176ae5603 : SUCCESS in 7m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/30e94fa2e7e647aeaccdad309846121d : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94c09970de4d4fe397ed6543539108f9 : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8432ff0d8bdc46aa951a6760c981d030 : SUCCESS in 45m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0d0967739f824815ac60852761065376 : SUCCESS in 45m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9fc09b8489847c19b01fdd344c03425 : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/71438395b0e74cc0844f39cd00eca5b3 : SUCCESS in 1h 58m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/64b3219ac3484265b6f1ad675f541c04 : FAILURE in 2h 18m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/39ceee948e86461194c13e3d046a1f1a : FAILURE in 33m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/05b56b58fe3d4b8db706d51496b1f2fa : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f08cffb58704218a57ce60d4df4991e : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8dbd345b79064f579cb75d67ae598201 : SUCCESS in 51m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3b93c43b928c4ad7981ba2af8850c36e : SUCCESS in 1h 45m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63a96f131b4f4b508f715c810f15cb43 : SUCCESS in 1h 05m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a7bdb9db209a4726aaec775ffa07babb : SUCCESS in 1h 21m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/65bd24fffd54498d9f7ea10ef687defa : SUCCESS in 1h 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0911ca38f105ba48195b95c82b91d536d0e6a25a","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-11-12 02:01:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2412a8a2e71b3d5a7cf6189604572bfd10eacbb4","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 06:18:34.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"be35d20fdc85dfefdf05759922f9351f47b52b0d","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 08:34:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ba86068840f36b0c36da0dbc62508dea25f6c237","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 01:51:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dafb6b9ecc7e8e04a7f0ca9e6cc11532175975da","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 09:56:17.000000000","message":"Patch Set 3:\n\n(1 comment)\n\none nit comment, otherwise +1","accounts_in_message":[],"_revision_number":3},{"id":"1b5d5cb846475ae16024a1c971731dbf67442c42","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 18:32:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"952c66638c05f323624cbf4189e91a96abc663d5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 18:32:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2eabf9ace11ada35d0ebd3cee5b2a918635dee32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 21:51:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2bf477ddf28459b81e4d60bba6c1e35 : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4a118bc1efa4472a28e9bd615aafcf3 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923ad44216064e3e93794e685356162a : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44453aec2c524419a5087cfa41a1c3ec : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46404df9832d42269f64b0eb253b8808 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb0ee7e70deb442682f92518dc1fcc01 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df1aa72bf544b2783baa3a61d0745c4 : SUCCESS in 9m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ae6f07b09f64d9695af5fde47219742 : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ebd761ef49c54f7d8f79718449a268ea : SUCCESS in 9m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b19c638076b0481991a633baff8d93ae : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a25e874f16704a1698ff8ce9861ec0e7 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed4c706d15c74eac8da02e33b2ff6cb1 : SUCCESS in 42m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6b5486fdb1ea4290a124d818d26aec7c : SUCCESS in 40m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec66a47332aa47a2aa82f114f8f8955e : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a408b8142518459d82ee98bb4db9a166 : SUCCESS in 1h 53m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7ee2e3cbec7e40fe93e6b73e4b8a8e6b : TIMED_OUT in 2h 36m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/632c61e0c6af4bedbc6eb92a45fbca0d : FAILURE in 34m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/96ec705e45c94d01acad77c3afad1ce2 : SUCCESS in 1h 40m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/21035ee5d39247f385df903b2e959f2b : FAILURE in 1h 04m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bdf5b53a0de4b448b6ae146fea01598 : SUCCESS in 59m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ccfb083a27d440e58484858a33658afd : SUCCESS in 1h 34m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7543b26f7364163b0c422f865f42022 : SUCCESS in 1h 08m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/52c06bc7bee24d9bb32a02f5fdaa9c84 : FAILURE in 1h 16m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/130b5c4da9344593850fa8bba867664d : SUCCESS in 1h 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b78fd1ca47c543babb9b71f1c8c8b5efed2eb02f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-27 22:45:04.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4ec3b69a1d9199c349a1286a5362ca39728ff9f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 22:45:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6fab643f7ed9cf8c94f2d65464e6f764f855e2e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:33:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88f398f527174a42ac820edb34e4a5e7 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2a5ab9a1ed4900bfc11aafdffc0e6c : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a866bc17db364f99a2410afc41ac30bf : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25a731a6f8374c16af82e801bab82a3a : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30e400e835de485d80351c4e6dc7b5e3 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9626d2582344efdb4a29bec27110047 : SUCCESS in 11m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/71680ff7fe9f45a997bc332f28664230 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4909d3db5b2e4640b4b5151bf400ce00 : SUCCESS in 5m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c9d520e249c43ebaa044c6e24b37471 : SUCCESS in 47m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9735af687f884fc78150ac76be056811 : SUCCESS in 43m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d68547f6bdd44d1aabd13998eeff286c : SUCCESS in 1h 33m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ed8c2600b32b4d33918be4fa4b490f08 : FAILURE in 1h 46m 24s","accounts_in_message":[],"_revision_number":4},{"id":"ec16304782ca537330133ba02f48bf5ce351ae6e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-28 05:14:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1d021cc4625f0868c885054b3b754170a5e53b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 07:07:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5be250a4b14a416fb522c0642c16a10c : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81c3262c72064c31b5e7432e9f3558ac : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83afce143e141c69a8659c061883a90 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2987cd16b5104a62b7ae75cb9cee2bfd : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c107b8740fee45f4a0ce4479bc92dbc0 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1763d8394f044655a0a6e6fd28a67c94 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339c389dc1a04a90a9be130ff9b9d996 : SUCCESS in 10m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2c66c30504943b0903bb14aadb889c9 : SUCCESS in 4m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/244d8c1248b04862a9b041159c75d75e : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0025a408897f478881e25a96d54b9550 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3284cf0682f45bbbcefcde19de6740a : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/212ce5fb0925414b8f9dfa9ad36682b1 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e027f867fdbe403c9ee087a5f44b433c : SUCCESS in 46m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f359be8c69394d03a5d270e0c6669400 : SUCCESS in 1h 32m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d5309f3be82b4cb8a9ff7ca6cb7115f4 : SUCCESS in 1h 42m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c4ea1db76f604a1582e6ccfe7569014d : FAILURE in 1h 51m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c936f910bb8a4180adc3ef1d5aa81752 : FAILURE in 32m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/573e3ac7a1ba4b98bf968bcc8f2ea091 : SUCCESS in 1h 46m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b84f0ff03d13436ab789215c0bb78446 : FAILURE in 59m 22s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3949f444cfd4481a3c7adc4e2f7def8 : SUCCESS in 54m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4155cf4594434040823629c7a76ddb99 : SUCCESS in 1h 26m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd59e86fd9b34aef8228b810769ea546 : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6ea735a0d84f45e0a1b2985abfa3a4c7 : SUCCESS in 1h 05m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/53dc819dbe9c4bc0862b47b5ee731c30 : SUCCESS in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"38bcf286760005abcbb66fc40b8d5bc9dbe0819a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 07:07:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fab6f19915bba3a8debe5d9e875cdebf6687cee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:45:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b41ad0768f3b4e81aefb6a64b67bfdb4 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35695cffa9af4696835156fe088502f1 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5708245ab0484c78a866c5024f8c7adb : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d63b93591764c78a694d05b5b85185a : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/673ffe5ff868432280ef6f550454cc3e : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c032aa2f4f64dc5a30c4ff35ccd4a29 : SUCCESS in 11m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ed6760b3e5ad4b91b9829ca76c954a40 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0273baf68fb4450a976152e69b5468e9 : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ea7723e5f4046618b0a812503b26004 : SUCCESS in 1h 08m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/02dd52644df643d9ad7c5e6491962142 : SUCCESS in 41m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78d37ea8a2434c32b6af21a6d50c1c8e : SUCCESS in 2h 34m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c4cb855582634fcda86c7f9e59e063a5 : SUCCESS in 1h 58m 49s","accounts_in_message":[],"_revision_number":4},{"id":"f12d8c80b5f4eb7c5c7c754c38291674508ab341","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:46:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3b30059a376b0c38a1077422f30da389c177ed24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:47:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ad623184224211a1a7b9f7f82c71bb : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4},{"id":"1f23198603fc25811ff437a913341716934274b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 16:54:48.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 343e3d8f38dff0310d314a0ca8d7d44d53d506ed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5af97a248b2c20cd12d743b124fec9848b1d965c","revisions":{"36260b3df7eb547cc289473050bcd527274fb915":{"kind":"REWORK","_number":1,"created":"2019-10-29 21:18:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/691987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/691987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/691987/1"}}},"commit":{"parents":[{"commit":"43577a6c047d9e76c690c001f81c45020d554c27","subject":"Merge \"Fix batch member update error on empty change list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43577a6c047d9e76c690c001f81c45020d554c27"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:15:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:18:14.000000000","tz":-420},"subject":"Fix update API when barbican secret is missing","message":"Fix update API when barbican secret is missing\n\nAPI update calls were blocked if the load balancer had a TLS\nterminated listener that the secret has been removed from barbican.\nThis patch corrects this problem allowing users to update the\ncertificate reference.\n\nChange-Id: I96908e6cbdb523f95298aff463a749d15e98e1ad\nStory: 2006676\nTask: 37322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}]},"branch":"refs/heads/master"},"02f54bc562f157b02324682c604791fed889f22b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-08 19:52:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/691987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/691987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/691987/2"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:15:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:49:37.000000000","tz":-480},"subject":"Fix update API when barbican secret is missing","message":"Fix update API when barbican secret is missing\n\nAPI update calls were blocked if the load balancer had a TLS\nterminated listener that the secret has been removed from barbican.\nThis patch corrects this problem allowing users to update the\ncertificate reference.\n\nChange-Id: I96908e6cbdb523f95298aff463a749d15e98e1ad\nStory: 2006676\nTask: 37322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02f54bc562f157b02324682c604791fed889f22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02f54bc562f157b02324682c604791fed889f22b"}]},"branch":"refs/heads/master"},"4217859ad95fd2090a2488d7944a9be1cd057326":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-11 16:31:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/691987/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/691987/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/691987/3"}}},"commit":{"parents":[{"commit":"3c05ce1297492ad1346ace3377f700c062e628da","subject":"Fix listeners with SNI certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05ce1297492ad1346ace3377f700c062e628da"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:15:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-11 16:31:47.000000000","tz":0},"subject":"Fix update API when barbican secret is missing","message":"Fix update API when barbican secret is missing\n\nAPI update calls were blocked if the load balancer had a TLS\nterminated listener that the secret has been removed from barbican.\nThis patch corrects this problem allowing users to update the\ncertificate reference.\n\nChange-Id: I96908e6cbdb523f95298aff463a749d15e98e1ad\nStory: 2006676\nTask: 37322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4217859ad95fd2090a2488d7944a9be1cd057326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4217859ad95fd2090a2488d7944a9be1cd057326"}]},"branch":"refs/heads/master"},"5af97a248b2c20cd12d743b124fec9848b1d965c":{"kind":"REWORK","_number":4,"created":"2019-11-27 18:32:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/87/691987/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/691987/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/691987/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/691987/4"}}},"commit":{"parents":[{"commit":"3c05ce1297492ad1346ace3377f700c062e628da","subject":"Fix listeners with SNI certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05ce1297492ad1346ace3377f700c062e628da"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:15:13.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-27 18:32:42.000000000","tz":0},"subject":"Fix update API when barbican secret is missing","message":"Fix update API when barbican secret is missing\n\nAPI update calls were blocked if the load balancer had a TLS\nterminated listener that the secret has been removed from barbican.\nThis patch corrects this problem allowing users to update the\ncertificate reference.\n\nChange-Id: I96908e6cbdb523f95298aff463a749d15e98e1ad\nStory: 2006676\nTask: 37322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5af97a248b2c20cd12d743b124fec9848b1d965c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5af97a248b2c20cd12d743b124fec9848b1d965c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
