)]}'
{"id":"openstack%2Foctavia~692208","triplet_id":"openstack%2Foctavia~master~I300839fe7cf88763e1e0b8c484029662beb64f0a","project":"openstack/octavia","branch":"master","topic":"fix-missing-cert","hashtags":[],"change_id":"I300839fe7cf88763e1e0b8c484029662beb64f0a","subject":"Fix multi-listener LB with missing certificate","status":"MERGED","created":"2019-10-30 18:48:49.000000000","updated":"2019-12-09 23:08:42.000000000","submitted":"2019-12-09 23:05:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692208-1575932737162-758459c9","meta_rev_id":"f3b963cf65940ee7864a1e200976a24fe46a41e6","_number":692208,"virtual_id_number":692208,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 23:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-12-09 20:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-09 18:12:35.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-12-09 18:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-12-09 20:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-09 18:12:35.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 06:18:48.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-26 10:13:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-12-09 18:12:35.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-09 18:20:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-09 20:59:36.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-09 23:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c766cce193b814b57499d718d6536a40b26efa18","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-30 18:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a5f0002da6441602e217c8eb73f763448d38bea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-30 18:49:08.000000000","message":"Patch Set 1: Workflow-1\n\nNot finished, just saving my place.","accounts_in_message":[],"_revision_number":1},{"id":"9dbf6b3525c527bf9c96e7f18d80766479534f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 21:27:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44dcfae04191460187b5f025fad63739 : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bdae91640444d89a5d0b415f9e16adc : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb7c5bbee2f432795506aa6d5bc03bc : FAILURE in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22f37d83410f450c83635955d1f3ad06 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44437ca43ce448f8b86f9d74ad8fb710 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbe97fc226424d168dd45f7a24c0137d : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536429bcb256406a94a7d2df830944e6 : SUCCESS in 9m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/49897b17f941468995278b383a11a6d7 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/356d641e78264d4c832cb1aebbbe05bf : SUCCESS in 7m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a7930f69c619443597e0442d9a058ee4 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/883ff2d7db764a84acdc6ff16a6922ea : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e15cdeaacbf746ae9686cd9020de5527 : SUCCESS in 58m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dc8d1d79d59e4eb4b4e0d7e8b904e5ed : SUCCESS in 50m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17907671caa64c87a1a2545c64eaccbd : SUCCESS in 1h 29m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6037cc61ce12447da00fcdd2cac65f7b : SUCCESS in 1h 37m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bc25706ad55249bca9f36f86a5cc6934 : FAILURE in 2h 11m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66fbb88c672f44f383b3b9f31dc20de6 : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/667e4a0131164603a7f592ad3f9f4691 : SUCCESS in 1h 22m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/295c8bc4f08b43f3923828b9875c3596 : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d24440213c044e8c9d0627bd84647f23 : SUCCESS in 51m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f0f13f53ec234ca7bc7455bb2c689676 : SUCCESS in 1h 59m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f94ef02bb87842418fe2c000d83e3fb3 : FAILURE in 1h 34m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fb173a59b281473a8639ac77df11caa9 : SUCCESS in 1h 01m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3ca094b16f9c464c896be95fbd6880f3 : SUCCESS in 1h 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3300cbf7a87b57e30c2731e6be570af2815ca1f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-01 22:58:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"539adc2be8bb20400904d8542da321ed07cb91cb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-01 22:58:15.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"36e53364ed5fce0ca6ed1f879ea1c1ac5ee5bfa1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-02 00:21:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0daa24f2809c99af6e0bf09858b23f23497f9c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 02:58:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09c3d68fbf6344cfa6affbb14012d758 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad6302ca16d84458bfec690f88c04094 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b8c38ee3ce43f5add30cd67dcd9d6c : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a03cb7b25ed45c0b6836718bd93df66 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549f2402f54f4ae996f02429bf866745 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32674f5699ef4f23b65ca41f1a760b05 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8bb080ec42144e68739b510ba38392e : SUCCESS in 9m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/411821f5a2cb4fbfb6396b58c6e5f293 : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/75ceb73a29c347bb8e87d863f86f5633 : SUCCESS in 6m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/88e612c8a4e3428390e561ece1eba8c4 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f74905e31fe442ea105882bda420b0f : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ab3df2ea3a343b7bc4419091659c593 : FAILURE in 41m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aac18e92364e4010a599536582b93c27 : FAILURE in 44m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7340df1e5e204b70bacfce472d993f2b : FAILURE in 1h 15m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/10e18c6c0cee44c488f7dc3e6648b30d : FAILURE in 1h 33m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1823f6e3d9674aa498032f7f51aef52e : SUCCESS in 2h 03m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3cc4ab197af4e40b44109a78a430d46 : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/24cbab106bb9480bb7823e9c17410d3d : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cbc7be97137b4816ba847484c6c54031 : SUCCESS in 1h 25m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9398b21753954c7bb796c1c40528676b : SUCCESS in 57m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be5b188c5d8144ff8d6067a3ce1dc270 : FAILURE in 1h 46m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c679313312d541cdbcd47c5a9ef40e47 : SUCCESS in 1h 22m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/96776c5d34ca4fdd974cf9f01fe3848b : SUCCESS in 1h 04m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a24eac13698943449837d53da8ae3a69 : FAILURE in 1h 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa40e2e5c00ab92f9fbfb6ee8188a4f3561b08fe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-06 21:11:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1c8b5aaf4d46fb7f00f3be6c5122abba011333d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 23:06:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f02cdc67fadc45628dd016aa5d607df8 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/013603bb9c1047eebfebefea7206d0a8 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81656f7343c472f89affaaa8a0e1af2 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f69aa6872bc440618b83cb7c881e8c25 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cd53217a06548e4aa55cbb95d8149bb : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81ff82d00304477f97ed01665fc925ed : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3c93ea28f24d1cae3981ab069b5bb7 : SUCCESS in 9m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/05c01bcedd1a44658a559bc4f3f90adf : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fdb8ef6201d04270898421aa4e19a339 : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1966f6d1b0bb4dbaa7720c413229dd8a : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e28d1fa62d274b3cbcdf2f2a5d95d5c4 : SUCCESS in 6m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44bac33a512f4086b21422de77074095 : SUCCESS in 39m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4ece998037c749a598e8a3f04a9a08c2 : SUCCESS in 39m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95aeaf81843848d9a253dee5950a6750 : FAILURE in 1h 21m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1ca2cdc099914ebf8232b4d8cb25dde3 : SUCCESS in 1h 36m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7a82b393059545299e216151acb16b35 : FAILURE in 1h 53m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7126404f228a4deb9c35ab774e452112 : FAILURE in 34m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5dcc9fbeb8b49909aeddcbf871e5ba2 : FAILURE in 1h 39m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82096942ef974ab5bc23859994caf9a2 : SUCCESS in 1h 24m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a9d64e2fcb124cd0b3d6ca43b4d8ce63 : SUCCESS in 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/60bba55c89ab4c88888e867b7299a496 : FAILURE in 1h 36m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8839c079693e43509d2d86212beb2796 : SUCCESS in 1h 06m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/786a82d6772c42338f5b99dd26596469 : FAILURE in 1h 27m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ed5633fe12494519840c51106199be85 : FAILURE in 1h 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a3d56161f405879d73c7febd2e7a0ed3ce80e0ce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 00:03:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af4c72f3d88ee5183290cdcfef079993a044f213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 01:50:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c8fddf3e3048e49f10ed13fceaab51 : FAILURE in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49db4122373449a787c5a0376d8f8c2d : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6cce5ddeb245ea983c13852906997b : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc20a739be9b4715999f177def43dae2 : FAILURE in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5eaf9cbb1e14c8f8cdbc2ea8c895db8 : FAILURE in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b1c1cb8d17f40d9bba784892434422e : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb66a0abd47490b9e23e50b74a8a767 : SUCCESS in 9m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7866f237d5764878b67c12423b7574eb : FAILURE in 4m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e16610933ec74766ac71158e22d88ad5 : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/90a542392f584cd68281b73861addfc5 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fef948c25ac04532badb75f4771f1556 : SUCCESS in 5m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2c13046d47a4ab69dfa10b3e527bd15 : SUCCESS in 39m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a4c92f774ae74cf4ad0a8a07be061550 : SUCCESS in 42m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8e10504233f4e9cb8dbed41f5056bf1 : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0780df562760425fa39479dea6c02025 : SUCCESS in 1h 29m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ad2c76fd84d64c54b433d71439cf8776 : SUCCESS in 1h 47m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e69110ccd494f5594ed53805f554f61 : FAILURE in 31m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/007a56b801314c648ce9df976c86f8b3 : SUCCESS in 1h 44m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7094fa20f6c24792b7e2886c5ed76a1d : SUCCESS in 1h 12m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b5b5fb2d0be4c3687204f802fb30200 : SUCCESS in 46m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d4d2d749f6cc4a82b0bde1d7cafadd3c : SUCCESS in 1h 25m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc4e1d77d409467c8bba786e0e42d54a : SUCCESS in 59m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6e0488cab57640a0ab833be7fe785d6f : SUCCESS in 1h 01m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/162d6c6422d54abca654be618e67705b : SUCCESS in 1h 19m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"381189bd9146423c1c54636af4a58cd9763bbb8c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 15:44:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"133ebc990d82b0a467d772ef4a708e77951fcced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 18:06:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c569842663214c4f98aefd095a4c768e : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc02e1af53174824b2a62b4b13d1dadb : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157b6c39bb98498f9c47a8c3a0d1af5e : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51a7b6ef6d6844a2a2f64dd93ca2d489 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9025748fed8a4700a66bcc4f958f477c : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1af97f8af2334d6c8e78c3b73225172f : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8439d49016ac418e96ee924da045e27f : SUCCESS in 11m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0a305a43dcd444598c7f285dde3aa326 : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e5b356d2e2754e8989d43dfbdcf71e05 : SUCCESS in 8m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d7cd5d89a8794e47a32bb5b490ca2833 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a04be49f2f74fd99c72d3dc971f92fe : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1f6398543ec4127a68e03da0d7e6490 : SUCCESS in 50m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e905a63a6395404a9f20b20cf243ff7f : SUCCESS in 47m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce0caa61fdb245a0ba03e73c023bec82 : SUCCESS in 1h 49m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fd89d155a1b744e1ba272a7dcd3a46e7 : SUCCESS in 1h 43m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/20acc6b703b44d1799458cfce76be666 : SUCCESS in 2h 18m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c6e6a12c4c24a308d1e027ec8e3640f : FAILURE in 42m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a2fe54596c0f4574bf57691c4ff9b3d3 : SUCCESS in 2h 15m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86ecd10075e1404d94b1a9d6382b2148 : SUCCESS in 1h 19m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2df9f3a80a3541258f2bcca75c29fd19 : SUCCESS in 56m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f82c5baff48c42719bce3a4c81971310 : SUCCESS in 1h 49m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5874563b023423fab07d9f13382a481 : SUCCESS in 1h 04m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8c515ffafeed41e08f4435075ce5158d : SUCCESS in 1h 28m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/247f6103d70d4298ad72eac3eafa4ced : FAILURE in 1h 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"22154401f6020038e46b6a8eb8862118e42308e7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 19:52:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1cd2516d3957bbb7c5185c6e8d8115285da51fb4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 19:54:58.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1296e7aac0bf23f5cb8719a51df4d8bef4282539","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 20:07:45.000000000","message":"Patch Set 8: Backport-Candidate+2","accounts_in_message":[],"_revision_number":8},{"id":"7f71d468f65b855ba10d98fcab6e698cff76b989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 21:59:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fd5a520698643d2ba92bd63d1bafecd : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/058ed58ac7ba4611b7842c3d6bea44d2 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccaf5a6a2ef84217ba007995a69bb8df : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1dcdae0ee0c4dff8f33ea374b55104e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06d53cb405dc4da7aa617887faca1708 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c4b99c29aa840ed9b9d09c463b272f2 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62c7e955ecb43b28616b4312c94c2ce : SUCCESS in 10m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4add8cef44a34434ab433ccca2027f74 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4c84486ad1634ad695d462b76e8128e5 : SUCCESS in 7m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2c6ba58622874ea39f4f5495c192150e : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5926c401d7e8451ba5e923d3559eb493 : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a720073acfe416e82563f0143fc858a : SUCCESS in 43m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/eae133bbba924da7a69cf82dd8b07a5b : SUCCESS in 47m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59e2989cc5d84d7499ab5a56b3326dec : SUCCESS in 1h 36m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/752294114d5140f19c7ad7a24f69367f : SUCCESS in 1h 38m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/221287b93dbf49f189dc6489302d2fc2 : SUCCESS in 2h 02m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/115d09278ff84d0384e317486b2ca98b : FAILURE in 39m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/42a5abd0c07d4a33a869559a322cae61 : FAILURE in 1h 47m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43117336b8bf4260abccc367d254880f : FAILURE in 1h 23m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1369bfc8fd1d46cbbff709e81045e447 : SUCCESS in 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd3f01a4620b477a86804e54868f0a3f : SUCCESS in 1h 27m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee9398a1c8df424b8d3cde3ec563792e : SUCCESS in 1h 13m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/efcdb1c3570d4bf8a254feb7ab161cf4 : SUCCESS in 59m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7fabf24925c94f29820d45b710481499 : SUCCESS in 1h 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8fde11f4cc1ccdf9cf29c4383a57eb25f22c3ae7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 22:07:43.000000000","message":"Patch Set 8:\n\nrecheck\nnova failure: tempest.api.compute.servers.test_attach_interfaces.AttachInterfacesUnderV243Test.test_add_remove_fixed_ip [71.026210s] ... FAILED","accounts_in_message":[],"_revision_number":8},{"id":"ae48b349888b39d69e0171659dcc012ba7e368ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 00:16:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f86a3369fd44277912ae6165edd1bc5 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d564d0d31699411c8125d574495b2502 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2bfb1ab61e94215b2afb77cbe50ede0 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/576c88bdf31f423098b09912a7ffb108 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd5ec6f50eb54562b7ceee936095464d : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb293e9662b840a68b27a1e001b5461f : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e373855ff1a249ffa82f86436571ad09 : SUCCESS in 9m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c02ccb5518c5473f96377a00d7879d0c : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c3baf856823f430ebe40ef2cc309763a : SUCCESS in 5m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/24896ee5f94944c4841a012441072393 : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cfd5ca1e08f44a690cb982e497dbfd0 : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82b8fd158175450f9574c56baee8484e : SUCCESS in 38m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c75603fa9fdd4da2acc987a22e7ddf58 : SUCCESS in 48m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6db39c7c50fb480cbd94b3318f512c1d : SUCCESS in 1h 41m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c13e8eee7d45429a898438a82e0be0da : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8d14d7b0f0164b158bf1882cd6defc7e : SUCCESS in 2h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/532c46a04f594e458910e68b885ef23b : FAILURE in 37m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec40fd1434f742d693eba3026e58cdc3 : SUCCESS in 2h 00m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86fa2143ec8940959b8cf54f0fc59767 : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eca0a71632904d7986b37a462f5abba0 : SUCCESS in 56m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3a7c97af869b468f80804ca904c38950 : SUCCESS in 1h 45m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/372ef791a9f6499a8900dcf208572eca : SUCCESS in 1h 15m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4938b297b997448d854f5f674041750e : SUCCESS in 1h 18m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ff05bd5c636b443bb370bebd83c3ef69 : FAILURE in 1h 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"176ff0f0ab80eb6dd6f49bacdb6782cb2b345a8c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-11 16:46:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6f91d188e0d4b1affdaaed56b7b731d54d0fc541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 18:39:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d05a275a254bdda7fa3ad2d8ea260d : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/953259a889334fc7a13d4d598412bd97 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02fa0f1ff44f4d5589720e7ecba67410 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe665c23d1d4a24a57cf0d4e8a33039 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4631d34f37aa4531a6d456ba7cfc0ad2 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a48dbbf121a4a6097df6768b367b1c7 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f25df996984eee8ffd1fe3dae3bb70 : SUCCESS in 10m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/70c944e373e14c43b66f995fcad7d3c1 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e2d226d9de8c401198066c834f641ff4 : SUCCESS in 7m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4f79c30e73834128ab414e5abfa6fe82 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccd026d3b8b5475f932c62c42cacc256 : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4a66c2c66e94d16aa554269e47b329a : SUCCESS in 42m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e003151e2eb747f5bf66045061aa5f1f : SUCCESS in 45m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bbddcca9b694c12ab512be2f6c36269 : SUCCESS in 1h 26m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8d421d2e29b4433f80119b6d8c4a0466 : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b06d05f22dd94d31b94d1e672b811fb0 : SUCCESS in 1h 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6c2a3ad4e804e5c86ea1586f84e7579 : FAILURE in 34m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1a014b189a04380a12a5f72381e3363 : FAILURE in 1h 09m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d7e021f35744dadb8b3e26a289f6d84 : FAILURE in 1h 20m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a931ece228c245d6bd9a844c81eb2304 : SUCCESS in 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc044f7f04494c36974c5b4604e6553b : SUCCESS in 1h 41m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00ccce14a2f243558620abbd23670fde : SUCCESS in 1h 04m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/95f5e402bcac4021989adab66abfef1e : SUCCESS in 1h 00m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b9370bfe86045848baf55bd9e5d322b : SUCCESS in 1h 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ba62ba7469d5e46cd2db03c59afd379498f9549c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-11 21:32:09.000000000","message":"Patch Set 9:\n\nrecheck\nnova failure:\ntempest.api.compute.servers.test_attach_interfaces.AttachInterfacesUnderV243Test.test_add_remove_fixed_ip [69.753469s] ... FAILED","accounts_in_message":[],"_revision_number":9},{"id":"3ef445435c355444b1ed88377d8d744bcaa0c24d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 23:19:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d02713d2c19c41cbb3059e831c2bd649 : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b88ec0620bd44ab09980b4f339abfdd4 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91c8325edc54691ac9c3a268706585a : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d934271a512542409f3cbda0e2fac442 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd3ade694844812906b566d8c6399df : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ca7f8a3319b4b8ea7f959a35696b18c : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42c8e1ff4314b0f96192a5f9b49aae6 : SUCCESS in 10m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc782a17a17e4bee9ec971b09ccf8553 : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5b78c5703a9649bc8f2219f92c2f927a : SUCCESS in 7m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f5962dab44cb46f78f1c55ab8b88a285 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b6590faa63b40598e93d5fbb0b1fc59 : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3a1aac6fe3340da9eb31751207ae008 : SUCCESS in 54m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/16bb605feba64d20b75912410beb134e : SUCCESS in 44m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2f05f1196f145e2aff20093f80d60b1 : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/75c7dd91d3f4431c8bd3b1c2101b22aa : SUCCESS in 1h 33m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a57b67ae3fdc44c6a4bd1ed78ead24be : SUCCESS in 1h 42m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59f5428e658f4d228cf5cc1d4dbdb0c7 : FAILURE in 32m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0989355146004e2cb3bbd0ff4a203839 : SUCCESS in 1h 45m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa4866e5e41643a8a40952bc262b003c : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64e4d44ac5684eaa9a753e81757a606c : SUCCESS in 51m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f59baaee6bab45bd996ede8ddafe6c3d : SUCCESS in 1h 23m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/715bd5d386d04a0db6cbd384d1749a49 : SUCCESS in 1h 09m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/91b67ffaa27a45008f469adcbd91eb3d : SUCCESS in 1h 04m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/77f60b223b4d4305ad28b0c08d6edf97 : SUCCESS in 1h 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"88eeccfc2275fd3734927711ce17fb8ff265fa14","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 06:18:48.000000000","message":"Patch Set 9: Backport-Candidate+1","accounts_in_message":[],"_revision_number":9},{"id":"00f786f6311d1b197a87fd272f6a297d1d12ee8c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 10:13:42.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"b1e8f02876db62df09b485931daa9f20c461d7f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 18:39:41.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"14086a5cd1dfa4dcc0602f98d1236a2e96fe8feb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-03 22:34:19.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c302c858012b4316fa1c2c3e26f7e137e488c6d2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 01:40:08.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"dc3032cd432515c4ac0b61b8914e6e21931f13cf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 01:46:42.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"1550dc52ba177790ce1a27ea6b036a1fd8edd19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:33:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc8761f4553c431ab641aac4ec652589 : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdd11ef9fd6a4ccfa366d3d95398eb4e : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f457ddae9ed04fb0aab2566abd076f53 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/054dc9356ba44a439e562b9562c28e2a : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfb9c3275a1a4916beb7b7f297e766de : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4e4528faac4ca9871412d740a3d2b6 : SUCCESS in 10m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1c31f37fe00b4984ae146493f198df73 : SUCCESS in 4m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c00cec1efa214ff2a58ff17cd8f2768a : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc034c02e2cc4a1ca11d67aa214eba9c : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/440a4d8452e6420987c9374453e4cbaa : SUCCESS in 43m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94c062f9828246d0afb6a5f24d727e81 : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72121b624d5e4c628365d208356194ed : FAILURE in 37m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff1f95d6db504a12b1f2bab3db4135ae : FAILURE in 1h 04m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4372c3c4ac38487997a08070fa161fa3 : SUCCESS in 1h 02m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fdba2255e3344dd1989cf0c7dd6fa9df : SUCCESS in 1h 43m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71be7518cc4f48e69cb61641ae06e9f6 : SUCCESS in 1h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6af72c7ff05d9d521a732ad8dd8e8513c5ea964b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 19:18:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d05366fb27d1c2ccacfbbffc51d981170363fd2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 20:58:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9da5bfcb07594b7f91d11fd1dbf70a4f : FAILURE in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b040d59c39d49a49a8b04571852a97c : FAILURE in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c279c2db2148399cbcede786191a41 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43ac771d49134f43b2593e65becd231a : FAILURE in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5521f8b65984711a7db6a45a064c083 : FAILURE in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3ef443025b44e3a9669aca7b41eb85 : SUCCESS in 10m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/846d931b17b84ce59ab67452269ce04f : FAILURE in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c7d644dca7ec44ea970df6e9738e1166 : FAILURE in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a3050cc0821483c8e6cd01e092882b9 : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ffad133a9a554fa48754d576685e5b5c : SUCCESS in 49m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e63df05fcd446b3a9c8244caf13b645 : SUCCESS in 1h 33m 26s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98d9dfea51b3420da8b2d1becf160983 : FAILURE in 34m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1cc2e5a7a4fa4f3ab22acbdc0cd5a6b3 : FAILURE in 1h 16m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7db66ef4778947b4a055398e7e438059 : SUCCESS in 54m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5dfb409183b04f039dbc7e1d1174d45d : FAILURE in 1h 34m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/721cf7e643244a74a5d815b27df3b083 : SUCCESS in 1h 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bdd0dfbd97107ff329e036768f781a991b76afbe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-09 15:48:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ba7647dc198f13e58b2c82d64cc91e454bb16ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:07:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25841b3bc4243008a1d81cf91a033fc : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a2a3e6a1d1746de8ea7ed3dab1b9035 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/297d33c4ac034b5a9626681754dfe0a8 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0edb02c3767b47948b6e5f7eed4dbf13 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53a8b2b15da547c9a512d5be4ff0eebe : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39353fe61cd44acab794215ff8033ec8 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2d1051f6e424a4eb0fca51238946d03 : SUCCESS in 7m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c35bb831e85c4c0ab6cfbf9efbb0d49e : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e05568d420de4cdda8bd3439c3b2b814 : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4704e7afe18342358a2498c361c6c059 : SUCCESS in 1h 00m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a23a4039122e4acf84daaec7f1ddfc9b : SUCCESS in 1h 32m 56s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6350f4111cb343f9b0dfbffd24fd455f : FAILURE in 11m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c350191a4ceb4e7285e543c4d04317e9 : FAILURE in 1h 09m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0c8d9427fed479d9a0399778f23b8c1 : SUCCESS in 56m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a6edf220c5744f40a8ebe7dd7a211609 : FAILURE in 1h 54m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ef3cf46cf07496fabf1a104387ab96b : SUCCESS in 1h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f3649f497157ffe1f991dec817ba97b1129f862a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-09 18:12:35.000000000","message":"Patch Set 12: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"005c89af9a9e3186f3fc992e8872895545fa44dd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-09 18:20:14.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"1900d856eea6b3ff43ce6a8c54a6d3be692d88fd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-09 20:59:36.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nAlright, I buy it.","accounts_in_message":[],"_revision_number":12},{"id":"a1474f14aa6288a410b987435f142bb8caeff7ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:02:03.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"73b4a59b6970ef4cce4040a54e57e6b578524c94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:05:36.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7c9882564494a9892ecb3469dc02b9c : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567dd4d5b59c4d0f8c01489d921673c0 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5db1393b77243b09311c3ad632ec9d8 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e95bee5e4e8445cbfdc08cd75ebd6cf : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f31515f25734862a6d47f10626a5844 : SUCCESS in 10m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a33683a22fb5434c902f1aa5da1a3575 : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4df4da8ecbcc458fb4b628e0c1469b6f : SUCCESS in 53m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55a91933eca143ee8e204951a072cb01 : SUCCESS in 1h 59m 13s","accounts_in_message":[],"_revision_number":12},{"id":"383653f31f7d0a25f3b0c4e6526e536067f61f2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:05:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"d3a816502915486b3b8651bc06060eb52be6366d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:08:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff34c01f63d0419d815602d837766d98 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7d23a711dde9d0427c6d04b45125489673c158ef","revisions":{"888598f8381c77602eb0c43f30700440a44921ee":{"kind":"REWORK","_number":1,"created":"2019-10-30 18:48:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/1"}}},"commit":{"parents":[{"commit":"43577a6c047d9e76c690c001f81c45020d554c27","subject":"Merge \"Fix batch member update error on empty change list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43577a6c047d9e76c690c001f81c45020d554c27"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:48:48.000000000","tz":-420},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/888598f8381c77602eb0c43f30700440a44921ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/888598f8381c77602eb0c43f30700440a44921ee"}]},"branch":"refs/heads/master"},"ff05bb0257867ec36e9ec957b6f99e93d44246fb":{"kind":"REWORK","_number":2,"created":"2019-11-01 22:58:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/2"}}},"commit":{"parents":[{"commit":"36260b3df7eb547cc289473050bcd527274fb915","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-01 22:56:36.000000000","tz":-420},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff05bb0257867ec36e9ec957b6f99e93d44246fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff05bb0257867ec36e9ec957b6f99e93d44246fb"}]},"branch":"refs/heads/master"},"8aecc60c462b0f4bd3b0538ebc97a6c81a7eb5bd":{"kind":"REWORK","_number":3,"created":"2019-11-02 00:21:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/3"}}},"commit":{"parents":[{"commit":"36260b3df7eb547cc289473050bcd527274fb915","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-02 00:17:11.000000000","tz":-420},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8aecc60c462b0f4bd3b0538ebc97a6c81a7eb5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8aecc60c462b0f4bd3b0538ebc97a6c81a7eb5bd"}]},"branch":"refs/heads/master"},"30c9842ef43bbb944702121bde0a11287e5a3ba0":{"kind":"REWORK","_number":4,"created":"2019-11-06 21:11:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/4"}}},"commit":{"parents":[{"commit":"36260b3df7eb547cc289473050bcd527274fb915","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-06 20:54:06.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30c9842ef43bbb944702121bde0a11287e5a3ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30c9842ef43bbb944702121bde0a11287e5a3ba0"}]},"branch":"refs/heads/master"},"d726f159caa0d2b8c56f796cedbf4a5b7ffe41f6":{"kind":"REWORK","_number":5,"created":"2019-11-07 00:03:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/5"}}},"commit":{"parents":[{"commit":"36260b3df7eb547cc289473050bcd527274fb915","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-07 00:02:58.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d726f159caa0d2b8c56f796cedbf4a5b7ffe41f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d726f159caa0d2b8c56f796cedbf4a5b7ffe41f6"}]},"branch":"refs/heads/master"},"175a9055eda5a951c7b7c6619b08e94658c0c329":{"kind":"REWORK","_number":6,"created":"2019-11-07 15:44:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/6"}}},"commit":{"parents":[{"commit":"36260b3df7eb547cc289473050bcd527274fb915","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36260b3df7eb547cc289473050bcd527274fb915"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-07 15:44:35.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/175a9055eda5a951c7b7c6619b08e94658c0c329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/175a9055eda5a951c7b7c6619b08e94658c0c329"}]},"branch":"refs/heads/master"},"c9d11829e769fc4181cda09a8e5f7d3378a7675f":{"kind":"REWORK","_number":7,"created":"2019-11-08 19:52:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/7"}}},"commit":{"parents":[{"commit":"02f54bc562f157b02324682c604791fed889f22b","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02f54bc562f157b02324682c604791fed889f22b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:49:37.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9d11829e769fc4181cda09a8e5f7d3378a7675f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9d11829e769fc4181cda09a8e5f7d3378a7675f"}]},"branch":"refs/heads/master"},"3a6764910b67a264fd328857955fc77f89427aaf":{"kind":"REWORK","_number":8,"created":"2019-11-08 19:54:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/8"}}},"commit":{"parents":[{"commit":"02f54bc562f157b02324682c604791fed889f22b","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02f54bc562f157b02324682c604791fed889f22b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:54:51.000000000","tz":0},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a6764910b67a264fd328857955fc77f89427aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a6764910b67a264fd328857955fc77f89427aaf"}]},"branch":"refs/heads/master"},"e4c105feb3b2ae2f3375b564bda94d98f03856b5":{"kind":"REWORK","_number":9,"created":"2019-11-11 16:46:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/9"}}},"commit":{"parents":[{"commit":"4217859ad95fd2090a2488d7944a9be1cd057326","subject":"Fix update API when barbican secret is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4217859ad95fd2090a2488d7944a9be1cd057326"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-11 16:45:55.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4c105feb3b2ae2f3375b564bda94d98f03856b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4c105feb3b2ae2f3375b564bda94d98f03856b5"}]},"branch":"refs/heads/master"},"c8fee592525459c4b8953b0e1853a935a9555375":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-12-04 01:46:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/10"}}},"commit":{"parents":[{"commit":"a3081d850c37c8b9733b6e036a7d00eb1d31f6bc","subject":"Merge \"Availability Zone admin API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3081d850c37c8b9733b6e036a7d00eb1d31f6bc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-04 01:46:41.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8fee592525459c4b8953b0e1853a935a9555375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8fee592525459c4b8953b0e1853a935a9555375"}]},"branch":"refs/heads/master"},"a28176d5935494991cd0273f59fc4e4231e95237":{"kind":"REWORK","_number":11,"created":"2019-12-04 19:18:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/11"}}},"commit":{"parents":[{"commit":"d6dc2b9a3e673100ee42af79d7a4af1f65bbd9ce","subject":"Transition member flows to use dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6dc2b9a3e673100ee42af79d7a4af1f65bbd9ce"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-04 19:18:00.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28176d5935494991cd0273f59fc4e4231e95237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28176d5935494991cd0273f59fc4e4231e95237"}]},"branch":"refs/heads/master"},"7d23a711dde9d0427c6d04b45125489673c158ef":{"kind":"REWORK","_number":12,"created":"2019-12-09 15:48:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/692208/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/692208/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/692208/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/692208/12"}}},"commit":{"parents":[{"commit":"3740b67854bd9bbff791b215273e2e5fd7a75675","subject":"Add support for CentOS 8 amphora images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3740b67854bd9bbff791b215273e2e5fd7a75675"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-09 15:48:49.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d23a711dde9d0427c6d04b45125489673c158ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d23a711dde9d0427c6d04b45125489673c158ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
