)]}'
{"id":"openstack%2Foctavia~692268","triplet_id":"openstack%2Foctavia~master~I7900b0e282cdf521da1ee62d48b190aea872be95","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7900b0e282cdf521da1ee62d48b190aea872be95","subject":"Let flavors override amp AZ/boot_network","status":"ABANDONED","created":"2019-10-31 06:11:35.000000000","updated":"2020-04-08 23:32:49.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a9fd8cc7bae847c02bc0d1ac8948ebd8de94694","_number":692268,"virtual_id_number":692268,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-12 23:35:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-11-12 21:32:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-1,"date":"2019-11-12 21:33:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-11-12 21:32:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 18:56:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-12 17:05:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-12 23:35:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3294f16a62478a9f54f7398a69e61c7751fb3090","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-31 06:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"195123cd824d03a61bc407426ab3aef06cc999dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 09:16:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ad7a773c97e48009c9eb9680a60f801 : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/563bbbd3030f4c5a81a419210811a8f4 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f06f3a3b2c24376ba1c4687fdccb6ab : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ce77365dab84223965e689535f9cf71 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aafdf3ed04f48bbb476bea12cb87cee : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75f2dd560e434c059f23c4202961af21 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758a96319f4b4c7da772f7bc2ff864b6 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97f8858523b49f69ed037b23900b37d : SUCCESS in 6m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db152d5984e14f7e894a7799a15c6fa6 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0c6a99f3c2e94d5783d6e5a8ab1723d4 : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5255a2d95b6843e8954992e9bf8bc7e2 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2012f72b68d84e7496eed614f45815f4 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19fefd160bb0410abb34e9d417dd1701 : SUCCESS in 46m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/336d94e39ad242f7babbbd1290ccbafd : FAILURE in 44m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/437e732e8d3f46249d4fc8e35e297e61 : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/59f5a7bbfba04ee3848f484d906eb8d2 : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2c15391cf54748268ea531e93049c8bb : FAILURE in 2h 11m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c540bb8479e4a939d3a9eb75df99b35 : SUCCESS in 1h 50m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/455f13c6e80d4c888a9131e9838aec61 : SUCCESS in 1h 50m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14c1f2fbf8bf43a89ae0f5deffb49147 : SUCCESS in 1h 12m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a9bc461795014836919b45d2e9dce92e : SUCCESS in 46m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f4b93898c0e94f5ab2ab1e1a34be8149 : SUCCESS in 1h 22m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a041cfac219a42daa06b6341bf85f8d5 : SUCCESS in 58m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8885ee3070984ee7abf334b6f79ec0e7 : SUCCESS in 1h 01m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b2b938beefba40aebb044b6c0272b466 : FAILURE in 1h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0576f9f0126d40f5aad1c1ad92e01d99fb680fd1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-31 09:19:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6d35b4a90e5e2bca5b973993a914c77d51033a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:21:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8af6f39f81774be1ab2107f56c609f7f : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e604c0074794478bdadc009e35d3bc7 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e29df98a239474b8fe7708f533a35f6 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01996f4df303455c86436675fedb2562 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb5119e10e95487cac033dd540a51e93 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c1b4599af1f4e0093c67e2b49c82593 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1d954bc1324a74a3933a972c048200 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/997d3d8330af4740b9cc615ad7349c1f : SUCCESS in 5m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e4a5f81eafc64de883aad413eddb2265 : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c3a0cc4c80a249d1a276b1859b929061 : SUCCESS in 6m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1b5775ce4b584ee9b87d6f4c0179967d : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6203997038b34c7486ab1e8dcb3f1a89 : SUCCESS in 5m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cac912bed284416a9723aa3fcce621d4 : POST_FAILURE in 48m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b483b6f162dd4c48ba63e5d75616453b : FAILURE in 42m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb43835c4c1d41fe8f1c83fc43741cee : FAILURE in 1h 30m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/977d580d610d4e9192734baf30ab9fda : SUCCESS in 1h 45m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/429411328ca94ddfb425b3b11dee53de : SUCCESS in 1h 53m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7203c5cea1ea454ca718731026213d03 : SUCCESS in 1h 48m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c2da300818142649d00121bab473bf4 : SUCCESS in 1h 41m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9cae1059d0834b60954b20a0e7776cfe : SUCCESS in 1h 27m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64b54c733a0841dcad67965802334f81 : SUCCESS in 52m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f3d5e3a74cdb4b63a4f97be250bc4746 : SUCCESS in 1h 30m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f25769ba11094edfb6c2b89cefc6200a : SUCCESS in 1h 02m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dca77adbde52455fa5264013c4334f03 : SUCCESS in 1h 03m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d04cd76fed6a4d66a5042910f3434aae : SUCCESS in 1h 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6375e22ee6c64429603d333f8d74ca958debca4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-04 18:56:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nThe v2 mirrors of some of these files are missing. There are a few coverage issues and it looks like tempest may have a bug.\nOtherwise looks fine.","accounts_in_message":[],"_revision_number":2},{"id":"2d5e6b187d53b4ad58fc947671bd70fe4a13caa9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-12 21:32:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a3299fc5dc80aa1a65f57fdb7650fee3ff7eac47","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-12 21:33:05.000000000","message":"Patch Set 3: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":3},{"id":"ca7ee8c96fac40577f0cee1c163e88277998e7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 23:35:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5247e617e784ee09f552df27b44a9e9 : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23a06c0d06f347d6b4c9081b22612323 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76b7292e6d04d4aa526fc83889136db : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d4d15372dcb48b497fd8d241d676334 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39c67a8a6ab24d65b80951d8bd486aa4 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81780c1b53f14131aefe07d56277d9f9 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085a79adb1394b05bf1b822231f868c7 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bd04d65a47b452fa425c4a5d2c850d1 : SUCCESS in 7m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7e7ce3be11b4ae2afaa1a36279a9f9d : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b51e50760bee43ddb0525f5517fa909b : SUCCESS in 6m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bdf96e878b0f4953ab88e3f33fcefd19 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/965ed2baa69446feba197117dea21483 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1356b51a92f6494b9c1680e69c104e49 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5b0b9d59130d41ab99f84b06b4a27166 : SUCCESS in 42m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88b5527eda72457fbfa4d822d076f9d7 : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3c09158b45d74de98de662477c80cf56 : SUCCESS in 1h 48m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/664ce3c465c24c43a47808b24023a375 : SUCCESS in 1h 50m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53105b148faf4cf099a4d089f94e7843 : FAILURE in 39m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfa331e2026246d589044bd4670af25a : SUCCESS in 1h 44m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da005566e9fc4bcbb2ada4532bd8fe53 : SUCCESS in 1h 17m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac510fecbdca4eb29a70f68826e2f7f6 : SUCCESS in 54m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9b26870bf44a4a76bf799f8c124bbd0b : SUCCESS in 1h 30m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92c50bb76efa4a4188e9d0787cce6f11 : SUCCESS in 1h 02m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/115eba46b6774709a9abfeaf9a72b357 : SUCCESS in 1h 04m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2d483652d4d243dc8de5ccc399f6fff6 : SUCCESS in 1h 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7a9fd8cc7bae847c02bc0d1ac8948ebd8de94694","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-08 23:32:49.000000000","message":"Abandoned\n\nThis was done as part of the fully fledged AZ system.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e56f69aeb3a40d86e379592dec501b7cd01d178","revisions":{"b06d40c096846cfc5e8254df2e359f877dbaeac4":{"kind":"REWORK","_number":1,"created":"2019-10-31 06:11:35.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/692268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/692268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/692268/1"}}},"commit":{"parents":[{"commit":"43577a6c047d9e76c690c001f81c45020d554c27","subject":"Merge \"Fix batch member update error on empty change list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43577a6c047d9e76c690c001f81c45020d554c27"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-31 06:10:36.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-31 06:10:36.000000000","tz":-420},"subject":"Let flavors override nova availability_zone","message":"Let flavors override nova availability_zone\n\nChange-Id: I7900b0e282cdf521da1ee62d48b190aea872be95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b06d40c096846cfc5e8254df2e359f877dbaeac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b06d40c096846cfc5e8254df2e359f877dbaeac4"}]},"branch":"refs/heads/master"},"405ce9158c6c526c444b3f9b15db67a66d52d091":{"kind":"REWORK","_number":2,"created":"2019-10-31 09:19:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/692268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/692268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/692268/2"}}},"commit":{"parents":[{"commit":"43577a6c047d9e76c690c001f81c45020d554c27","subject":"Merge \"Fix batch member update error on empty change list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43577a6c047d9e76c690c001f81c45020d554c27"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-31 06:10:36.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-31 08:02:30.000000000","tz":-420},"subject":"Let flavors override amp AZ/boot_network","message":"Let flavors override amp AZ/boot_network\n\nPutting these two together because they most likely both need to be set\ntogether to function (since different AZs usually plug to different\nnetworks). Note that the control-plane must be able to route to any\nnetwork used as a boot (management) network.\n\nChange-Id: I7900b0e282cdf521da1ee62d48b190aea872be95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/405ce9158c6c526c444b3f9b15db67a66d52d091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/405ce9158c6c526c444b3f9b15db67a66d52d091"}]},"branch":"refs/heads/master"},"0e56f69aeb3a40d86e379592dec501b7cd01d178":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-12 21:32:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/692268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/692268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/692268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/692268/3"}}},"commit":{"parents":[{"commit":"c3fbb4c29a9635f8e602958e4e23ab1de87faf13","subject":"Fix filtering for provider capabilities list API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3fbb4c29a9635f8e602958e4e23ab1de87faf13"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-31 06:10:36.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-12 21:32:57.000000000","tz":0},"subject":"Let flavors override amp AZ/boot_network","message":"Let flavors override amp AZ/boot_network\n\nPutting these two together because they most likely both need to be set\ntogether to function (since different AZs usually plug to different\nnetworks). Note that the control-plane must be able to route to any\nnetwork used as a boot (management) network.\n\nChange-Id: I7900b0e282cdf521da1ee62d48b190aea872be95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e56f69aeb3a40d86e379592dec501b7cd01d178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e56f69aeb3a40d86e379592dec501b7cd01d178"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
