)]}'
{"id":"openstack%2Foctavia~692445","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ib9183eea30cfaa1b2a893b6eb4592ad3e9527b55","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Ib9183eea30cfaa1b2a893b6eb4592ad3e9527b55","subject":"Fix Octavia stable/queens pep8 job","status":"MERGED","created":"2019-10-31 20:06:47.000000000","updated":"2019-11-02 13:27:01.000000000","submitted":"2019-11-02 13:25:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692445-1572701143970-7361ecdf","meta_rev_id":"2e977730b64f50abd39b0cf27de1903f9bf1b73c","_number":692445,"virtual_id_number":692445,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-11-01 16:38:50.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-02 13:25:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-11-02 11:49:04.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-01 16:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-11-02 11:49:04.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 16:36:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-02 11:49:04.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-11-02 13:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3be641c6f0cae534da6fc5e70b685a2bf93c07ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-31 20:06:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce381f4e422d94147799ddcf14d95a365fc375d7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-31 20:07:15.000000000","message":"Patch Set 1: Workflow-1\n\nSeemed to work locally","accounts_in_message":[],"_revision_number":1},{"id":"9cb74e376e5011886bb9d49b5fdd0f4be21bc48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 21:49:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9ac7a6137ead42baab18cf9b588a70c7 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956d134bf35b494898ee8f49d882dca8 : FAILURE in 15m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b45c7a7b19394b34bcd5f144689d9f6d : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/99b5805ce7a345ac88d105ec4e6fd3d8 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4cf2e59dea4ddebba45258da507da1 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3a6a320e7d149bab2f039f9ba370671 : SUCCESS in 6m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/89749596e1614932b9e437bdb6ff22fd : SUCCESS in 1h 37m 13s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/38a8b4474df54da0bd21957a6e03fcfe : SUCCESS in 1h 32m 51s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/550fcdfa4f9643f2abb63a740087cf06 : FAILURE in 45m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8b7c5c7a9338444e83370d2440e160de : SUCCESS in 1h 12m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f4669094e704f83b575fc8811f69e56 : SUCCESS in 11m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d7519a1f3de41dda3d3bd5cf2dadffe : SUCCESS in 5m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e2efd812c58949b99368204a011df1e5 : SUCCESS in 6m 05s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/316c0e16813b40eb9451ff0f4f7b0547 : SUCCESS in 53m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/111f49182f924b2bb2a5b4c49bc1437f : SUCCESS in 59m 30s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/530f858c13ab4ca1a315669a30078e86 : SUCCESS in 1h 17m 01s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a29204cda6804b9ab23302f189f2e32e : SUCCESS in 1h 08m 43s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/52ad84b35bd44bfdac96d7c8ae060b45 : SUCCESS in 1h 00m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d37b83a913bf44dc93b6fe18b877bada : SUCCESS in 40m 34s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/26f4ab70ef6447f0957fa44225d383f1 : SUCCESS in 47m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25c4a5c24bef4248bf199c5422f049c2 : SUCCESS in 1h 27m 11s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a5c75492b248447697027005aea603c0 : SUCCESS in 1h 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d29350049021f1bc7a01a257cef0fd2a07e3a819","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 13:20:23.000000000","message":"Patch Set 1:\n\nThat just made things worse, i\u0027ll see if i can narrow the constraints from just pylint\u003e\u003d1.5.0 to include an upper bound","accounts_in_message":[],"_revision_number":1},{"id":"83c2820e3288ebe4d514afaa8591a374c63abd2a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 13:23:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1e361633f210faf754f7312dc58ea66438b849a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 15:13:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67c846a9dc594ba28ab3dca680518f5b : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace5713b501d43f08abc3545e23c1bde : FAILURE in 10m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96ab2ee45ef84abf9ac75c33ab91aa65 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cefc833e47e041ff99034aa5c8955b4d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2708f99d26ee4438b57798c5aa611737 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9f6a1c15eb34a20bba950a8f7831f8f : SUCCESS in 10m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/449b3dcefc9a49bfa6ba4e9e28ac41d1 : SUCCESS in 1h 25m 05s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a3518256a67a4da49c76f5d8efee06cc : SUCCESS in 1h 24m 30s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5e181731d9d4361aead3b0e252ecfdb : SUCCESS in 1h 28m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7ef4498899f843e98d9e8332af829a28 : SUCCESS in 1h 30m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5b8484c77014de586fb22bdddd0cd7e : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e8bd847f314a4e4f94be8487cbd12d8c : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e3a43f0167cc400ab21de140e48d37b5 : SUCCESS in 5m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4971b39b2c54cebbcb32f061abbc4a4 : SUCCESS in 55m 14s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/25a961712d4245fa913900a9ed551810 : SUCCESS in 1h 08m 11s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d19f249ef72b41ec81ed521f4261a0d6 : SUCCESS in 1h 12m 57s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/955bf5a7121543d3aeef029d43c04966 : SUCCESS in 54m 19s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/4c0144eaeed442349456da87a9a179a9 : SUCCESS in 59m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc4ec4d7dac042d0848468112a5a5cb1 : SUCCESS in 42m 25s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/db5de4d443284819b61952cabf2bf4bd : SUCCESS in 42m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8db1e9ec26b491d90f54f6872a03ab3 : FAILURE in 1h 35m 23s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/dceef87d3417406892549c88bee0b962 : SUCCESS in 1h 22m 08s","accounts_in_message":[],"_revision_number":2},{"id":"d8dd5b161beb5e5427c3a59eeece3a3f8a1a5b4d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 15:28:14.000000000","message":"Patch Set 2:\n\nSigh, didn\u0027t notice this during local testing:\n\n  No python changes in this commit, pylint check not required.\n\nBack to work...","accounts_in_message":[],"_revision_number":2},{"id":"daa6a1b105bd8a638504eaba860deb175a0b34ce","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 16:20:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87fb564d7859872df2cbad4c8211b12cd0bb1a4a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 16:20:39.000000000","message":"Patch Set 3:\n\nThis would have been a lot quicker if i just used \u0027python2.7\u0027 the first time...","accounts_in_message":[],"_revision_number":3},{"id":"885d2368784f61c05c0e461bc0fdff86874b45f3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-01 16:36:18.000000000","message":"Patch Set 3: Code-Review+2\n\nSeems to work. Not sure why stable/queens magically changed to use python3, but this is a good fix to me.","accounts_in_message":[],"_revision_number":3},{"id":"5e1f009f2d39f8359982515e4edfea62381d3b71","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 16:38:50.000000000","message":"Patch Set 3:\n\nYeah, I didn\u0027t dig too deep on the why, didn\u0027t want to pull that thread...","accounts_in_message":[],"_revision_number":3},{"id":"efa22c5c45c211c387fc60dbacc5cca3af31ac6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 17:57:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/878cc7987c3a4cd18465915989a561ea : SUCCESS in 7m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6634bccebe14f2a8045dc2a79fa9e9c : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/44190a046bfc45f4a3b11720876ffefc : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590b8d2d8a34455f9350404fa8cde694 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/305c1e72a5e44abbb90a0c4273b50bb0 : SUCCESS in 6m 10s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e0417f7abde3457ead615358912cc129 : SUCCESS in 1h 22m 19s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/19bb95fb912642e3aaaf5bb4ed67b0e6 : SUCCESS in 1h 21m 02s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9978f8180d44b1e9ab3e26777726dc4 : SUCCESS in 1h 28m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/36023975812a400e92d34f30f4a526d1 : SUCCESS in 1h 26m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1f006b2a314b949d8ac6ae2e7977ac : SUCCESS in 9m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/15f3921113144fb0864ce3cc953619e7 : SUCCESS in 7m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/332b9784a154471c939523680dc9d275 : SUCCESS in 7m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d6f5ec7ed1a4113b87f388240f1e3d0 : SUCCESS in 1h 01m 13s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a765ccdea934487ab09f4a877da7afcc : SUCCESS in 1h 03m 17s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/19eb6d5d0c2a4328b56781fca66dd351 : SUCCESS in 1h 05m 50s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9cb799c4a4c3464ea31684b0bc23fa20 : SUCCESS in 1h 02m 20s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d06cd6b5dfc94a46823e6650318ef6c1 : SUCCESS in 1h 13m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b133e6e95494422ea10887ad01fd554c : SUCCESS in 44m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/c63f114b851744eeb46784a54bf2aeda : SUCCESS in 49m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f8854cc263046f6bbd3f2778392c50a : SUCCESS in 1h 30m 59s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/d8f075fc389340e0820a15080867cf68 : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ad59339c06104e9008162d1af5b6fc4bc2c0f0e0","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-11-02 11:49:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ea1e47a9c9cd87cd98052b28cb4e1a385dfa50b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 11:49:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"91b93620ca197b23120188cf376295d7f2b88e9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 13:25:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8379a610d4efbd4e7db8d0f4aa8a9af9ca657e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 13:25:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9da6701f7e8f427dbbee33434ad17d56 : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1d25c4a92a14d3ea7ac66d2220e80a1 : SUCCESS in 5m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf485b717f9c4a5095bb28aed17f11b6 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52911b89c6e4560a9c3b12d3c41b436 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a57e966a777a434aa87d0ef9a63c69ad : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e7d499d0e746afa63868124c332f38 : SUCCESS in 8m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b08d045ea51c40eeaabbe031cc785b51 : SUCCESS in 5m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a92b8cd1c89f4e79b2911dc81e242656 : SUCCESS in 5m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/47b04752d3314ff69a359f64d0edf5e3 : SUCCESS in 1h 15m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/fcbdb021f81241c99f01bd15c8a1afb0 : SUCCESS in 1h 14m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5094353feb454fd9be70e2967656da8e : SUCCESS in 57m 58s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a121a19e360d42249db2b3e623d497f6 : SUCCESS in 59m 02s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d516af6707844ce297039d792546424a : SUCCESS in 1h 10m 38s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4b50e9ad41d14f86941d237e23872a75 : SUCCESS in 1h 09m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31d9a396942840178df495c55f645f51 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/b4680b33305b4844a2837a6ba6c04fc2 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb76e0e4957b4367b764b11667704946 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/40c921a54120422eb63e7b26c1445d0a : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2e977730b64f50abd39b0cf27de1903f9bf1b73c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 13:27:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd9dd6d82b84a49b05b38b0b5bdb694 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d468ca5a48c47f9b12a0948c4345dd3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae","revisions":{"a93b5a4217788eb7d72ab70812054751fb575843":{"kind":"REWORK","_number":1,"created":"2019-10-31 20:06:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/45/692445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/692445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/692445/1"}}},"commit":{"parents":[{"commit":"7148cf63ad91442afb853ea17398ffdf79e2399b","subject":"Merge \"Improve the error message for bad pkcs12 bundles\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7148cf63ad91442afb853ea17398ffdf79e2399b"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-31 20:05:08.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-31 20:05:08.000000000","tz":-240},"subject":"Fix Octavia stable/queens pep8 job","message":"Fix Octavia stable/queens pep8 job\n\nFor some reason the stable/queens job was running under\npython3, force it to use python2 instead. Required using\na newer pylint (1.4.5 is known to be broken), and pinning\nastroid as well.\n\nChange-Id: Ib9183eea30cfaa1b2a893b6eb4592ad3e9527b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a93b5a4217788eb7d72ab70812054751fb575843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a93b5a4217788eb7d72ab70812054751fb575843"}]},"branch":"refs/heads/stable/queens"},"2450e8a67cead02a5e570159f1ec14f387b08bdd":{"kind":"REWORK","_number":2,"created":"2019-11-01 13:23:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/45/692445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/692445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/692445/2"}}},"commit":{"parents":[{"commit":"7148cf63ad91442afb853ea17398ffdf79e2399b","subject":"Merge \"Improve the error message for bad pkcs12 bundles\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7148cf63ad91442afb853ea17398ffdf79e2399b"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-31 20:05:08.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-01 13:22:44.000000000","tz":-240},"subject":"Fix Octavia stable/queens pep8 job","message":"Fix Octavia stable/queens pep8 job\n\nFor some reason the stable/queens job was running under\npython3, force it to use python2 instead. Required using\na newer pylint (1.4.5 is known to be broken), and pinning\nastroid as well.\n\nChange-Id: Ib9183eea30cfaa1b2a893b6eb4592ad3e9527b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2450e8a67cead02a5e570159f1ec14f387b08bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2450e8a67cead02a5e570159f1ec14f387b08bdd"}]},"branch":"refs/heads/stable/queens"},"6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae":{"kind":"REWORK","_number":3,"created":"2019-11-01 16:20:07.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/45/692445/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/692445/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/692445/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/692445/3"}}},"commit":{"parents":[{"commit":"7148cf63ad91442afb853ea17398ffdf79e2399b","subject":"Merge \"Improve the error message for bad pkcs12 bundles\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7148cf63ad91442afb853ea17398ffdf79e2399b"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-31 20:05:08.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-01 16:15:50.000000000","tz":-240},"subject":"Fix Octavia stable/queens pep8 job","message":"Fix Octavia stable/queens pep8 job\n\nFor some reason the stable/queens job was running under\npython3, force it to use python2.7 instead.\n\nChange-Id: Ib9183eea30cfaa1b2a893b6eb4592ad3e9527b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
