)]}'
{"id":"openstack%2Foctavia~693224","triplet_id":"openstack%2Foctavia~master~I2e8c0be2221c0c23b752fdf2cdff065cddf830a5","project":"openstack/octavia","branch":"master","topic":"ipvsadm-exact-stats","hashtags":[],"change_id":"I2e8c0be2221c0c23b752fdf2cdff065cddf830a5","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","status":"MERGED","created":"2019-11-06 17:33:53.000000000","updated":"2019-11-07 20:37:34.000000000","submitted":"2019-11-07 20:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693224-1573158953943-c91a571c","meta_rev_id":"7a3d9bee6881073c7736244ede01e3f04444a707","_number":693224,"virtual_id_number":693224,"owner":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 20:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-11-07 02:11:15.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-11-07 09:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":2,"date":"2019-11-06 20:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-11-07 02:11:15.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":1,"date":"2019-11-07 16:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-11-07 02:11:15.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":2,"date":"2019-11-06 21:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 02:11:15.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-07 09:21:11.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-11-07 16:46:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-07 20:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44b2920705ee228f23acc8d05ec1267219c65f18","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-11-06 17:33:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e41332b7a3a33e868f12249c66ba845ca8fc3ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 19:57:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21ac89705ec847d3875c54088c95cc20 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ad74da638bd4c1fb6cb97d8f46a7200 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95c30319b8f4f9ea618e9162fda203a : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7879e63afbc0471f863635da6e35d4ec : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27449f036504213bf14b10f955a297b : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f363d0bd5bc428f98b9e355719a8270 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8454eb6b3b4459be8f533bb47e03a0 : SUCCESS in 9m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f251ea007c0540798fcf1e3d1314aaae : SUCCESS in 4m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/29e999b010424b34a664e2bb87684927 : SUCCESS in 6m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5e1de36199584da0b79e473f41f0e2da : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7390c5256d6a46afb5a8108d338006b4 : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ee0d117be904f4e95cbf6c512e7d242 : SUCCESS in 43m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e2ff20ae0d244ab6b9d5713a23114722 : SUCCESS in 41m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e3eba56d0ec4eb1be7380aa96decbc2 : SUCCESS in 1h 22m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4bd9c0b6b9924a799b30823ffd31dbf4 : SUCCESS in 1h 44m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/875dd6a9552d48a5876f866f948368a6 : SUCCESS in 2h 22m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d096337ec564a899d99f23352610f82 : FAILURE in 33m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/62e8f10e900345a2bfef276212e30006 : POST_FAILURE in 1h 42m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0260d178949844b1b427b02a3c5da6c7 : SUCCESS in 1h 22m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a93e623ea7184dfb9f906cfde95d9d95 : SUCCESS in 47m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/472a107a55914368af04cae25515bd51 : SUCCESS in 1h 38m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc9e4cf9078d4e9bbd78dd5e7accee4a : SUCCESS in 1h 00m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/70f4690ec6fa421dac24ae1469f7a3d5 : SUCCESS in 1h 01m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/edf59ddd31f0445abfd854cbdc8c9e2f : FAILURE in 1h 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb5ee7e2030a2d4ee69d223b3f85e55a0b8f83ec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-06 20:54:46.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"62f8ac22d59a9304199b11a51f7cc1b5ec3c1d44","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-06 21:03:22.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"63c91262b5706504251a40fbe3bfca2b5b763604","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-07 02:11:15.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d602d0fe27ee7a3dc5cc88b11f171c9630ed7dea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 02:11:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98a06829c315d6cfbe3018642942bf9859326c65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 03:36:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b344b14632443df8530800ab7366f68 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89299e5103f4dc0b34d9c99a6005738 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c6099c05bf24451bade7c8b5fb10d3b : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f3b2d0828b34da18421ba566ce31ba5 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1374a56984a40c58ac7d58e7926a883 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe18ed363e254a7b83a4689240e58768 : SUCCESS in 9m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e6fd14f727724636ab24101a64bc7b23 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f325c56ae5f64370a4d1d149c7dd57d3 : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c7f14ee584e4749abf374fb4eaf4e49 : SUCCESS in 40m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8d5d3d01623040d884e515c44adee7f4 : SUCCESS in 42m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a69a65cf105441558f1a3e6eb8595f33 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/93218f8775aa4d3cba7df90060ef7f8e : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66fe145d54414f63be48fba7ba17804a : FAILURE in 1h 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"452b9141904ec5d3395767267e4e2b4b76f684fe","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-07 09:21:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"99ba5a1582a5fc5d86bd0aac3d405657f7f84763","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 16:46:17.000000000","message":"Patch Set 1: Workflow+1\n\nbump\nNova breakage on the first try","accounts_in_message":[],"_revision_number":1},{"id":"c654f32eb89cd26b73123abf8b6522fb1c542064","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-11-07 16:48:59.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit d78a5ee4bd188dc07a93de3bd7b966c380b51a9e","accounts_in_message":[],"_revision_number":1},{"id":"fd3f494ff1e99f6b59a81ff0ff629a9c2f1ba1f0","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-11-07 16:49:19.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f1df4b201b8b0a37be20ea5e3c6241aed31c915a","accounts_in_message":[],"_revision_number":1},{"id":"404572517577bb48a01dee0a4d7271990749c9a7","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-11-07 16:49:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3d8d1c9dae6236d485b358aad4e417be0af6495b","accounts_in_message":[],"_revision_number":1},{"id":"648426a3b696105be7b05e9ae3ad0235491230ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 18:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67556ae1cf3d4971b850b80a74e6c217 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5ed49e3785140619379357dda0b9b01 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5db1d6aed4143bf826c210780d19a96 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcd8c6901b5349be9a1b081cbba05eaf : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/199a83d27c5345acb870bbd83b6a5c7a : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62e00b733fd54d939f46cca2e171c176 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d49a0871bb4fdb8f77f71b8df8a3bb : SUCCESS in 10m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8ed2d2cebdd74711bcac27731f33a9e9 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/70515ed55f70445fbce8fbbc13f9ed64 : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/13ca1049bfa94dc986ca9659e80978d6 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61b4edccd2924929bfc9799bf87d22d3 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3cd538b453c4f68910ea584fe188bda : SUCCESS in 41m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f11bfdea20ff4fe9aa4c534b9adaa7a0 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3e42f0287f7465fb0f31a27f0c11851 : SUCCESS in 1h 43m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/88e18ecba1724c42871c5bb29f25fb0d : SUCCESS in 1h 47m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/476f3a5689104b1d90504a44c30f4e3f : SUCCESS in 1h 53m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a383c1e829d469e87ea1a1781dcb7b4 : FAILURE in 35m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e752a611cd924cea98574decd9048c41 : SUCCESS in 1h 47m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e92ba1abe404fe5896a856eab83b6e5 : SUCCESS in 1h 19m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64151ca7fcce46318cea942253812f92 : SUCCESS in 54m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7dccbb55fe6c49149d562002afd0c771 : SUCCESS in 1h 28m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caf1e0e22a51489492a18fbbfab33968 : SUCCESS in 1h 07m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c8116fcd5b89438e80168999141b03ed : SUCCESS in 1h 02m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c565ce9656d3442987a17144f105d979 : SUCCESS in 1h 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f72e168f8fd2254099339215f2b86f6ba73cbfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 18:42:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"130a4e46232b40edf48264617cdd7749750d4f01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 20:35:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"daecdc9a15dc9bf1c7f61d0f0e10fb4f76606264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 20:35:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a86643432db40389eb72371f90660c6 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/732e0b69f7644aa4bac545763ae2af86 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/828953cf03f54eeda79d408690b1eae2 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec29a121dc54ac081886ac9bd225aa2 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb8a023c71c7417c9c47a6c0c5133980 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b40f77bde1479791ba6562014413c5 : SUCCESS in 10m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/35da609aaaba4034bada50d8e4131adb : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/661accd07e9f4561bd790456fcdeac4d : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48a139ac4f8b42f38be24191e46b5a17 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e694476fa392456296d1f71e20a555ac : SUCCESS in 47m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8f2c283b4a54012a0f742dff9a6cc3c : SUCCESS in 1h 42m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/862368943ace442da0a222d5c6458e44 : SUCCESS in 1h 47m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cedee8233ba241a098d668302f6bc69a : SUCCESS in 1h 18m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ab2d08be4c79d4225995bb605f298684ea905612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 20:37:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60aa14501229426eb3bf25081a3e23ba : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0682fb977a76d356ede20f516c849298d34eb7fc","revisions":{"0682fb977a76d356ede20f516c849298d34eb7fc":{"kind":"REWORK","_number":1,"created":"2019-11-06 17:33:53.000000000","uploader":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"ref":"refs/changes/24/693224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/693224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/693224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/693224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/693224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/693224/1"}}},"commit":{"parents":[{"commit":"43577a6c047d9e76c690c001f81c45020d554c27","subject":"Merge \"Fix batch member update error on empty change list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43577a6c047d9e76c690c001f81c45020d554c27"}]}],"author":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-06 17:24:30.000000000","tz":-480},"committer":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-06 17:24:30.000000000","tz":-480},"subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","message":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints\n\nCurrently the keepalivedlvs_query script calls ipvsadm -Ln --stats\nto query the local lvs for connection information. If any of these\nvalues grow large enough they will be abbreviated with human-\nfriendly suffixes (K, M, G) and cause the get_ipvsadm_info func\nto raise an exception when it receives a non-integer value from\nits command output. By using the --exact argument in addition to\nthe existing arguments, we can ensure the output is always expanded\nnumbers, per the ipvsadm man page, and will only ever offer integer\noutputs to the get_ipvsadm_info command.\n\nChange-Id: I2e8c0be2221c0c23b752fdf2cdff065cddf830a5\nStory: 2006791\nTask: 37331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
