)]}'
{"id":"openstack%2Foctavia~693392","triplet_id":"openstack%2Foctavia~stable%2Fstein~I2e8c0be2221c0c23b752fdf2cdff065cddf830a5","project":"openstack/octavia","branch":"stable/stein","topic":"ipvsadm-exact-stats-stable/stein","hashtags":[],"change_id":"I2e8c0be2221c0c23b752fdf2cdff065cddf830a5","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","status":"MERGED","created":"2019-11-07 16:49:19.000000000","updated":"2019-11-08 03:54:39.000000000","submitted":"2019-11-08 03:53:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693392-1573185194074-4445955c","meta_rev_id":"8616d9d214b8ed4501f142e43bded21a9c6ad5e8","_number":693392,"virtual_id_number":693392,"owner":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 03:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-11-08 02:06:08.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-11-07 20:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-11-08 02:06:08.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 20:39:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-08 02:06:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-08 03:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"682eb66c7d991c22ff0c696e7dfc36a4b5bd3bb2","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-11-07 16:49:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02c81e980e76cfdd1b836493de69d490ab740b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 19:23:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c16775180814900ac529180190a8af6 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61ed2f5dcffa429f9da08af1a59e2dcb : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd522354ba64d1387c4932c60ab5f84 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cbca0c6bc97f4924bffc99f521d47315 : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6709e555663a4e8ba342245092416343 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2dffb4d842d43bfa398e3f349eab961 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb384c8f8e3447c58409e509142899b6 : SUCCESS in 7m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/12e8820b86eb40019f745a6da4da821c : SUCCESS in 1h 20m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/098a25c24f03494e831060b0abfb6589 : SUCCESS in 1h 19m 17s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10ec1398f66145828fdf24b026308362 : FAILURE in 1h 23m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cd9961fe41a14048bbbfb16710a4cbcd : SUCCESS in 1h 41m 27s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/70862d1516e64343ae1600823056b6de : SUCCESS in 6m 29s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2fcc3b7de43b4c928dd38e8341a36cb9 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a155b72b586435b8505f7c9afba6358 : SUCCESS in 7m 02s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49ddfc2519964dd2bbb10a7a40887367 : SUCCESS in 1h 06m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bd0e7617d8a84a7e9ba572f05431d6b8 : SUCCESS in 1h 06m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c47927978b574b4c947491217f7977b2 : SUCCESS in 40m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5d2cb6fcff134a05bec06ddc9e8cb5f7 : SUCCESS in 41m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6314356f80846a4b0eb664d71c2ff01 : SUCCESS in 1h 27m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fe8bcbda273843fb857ddb3e06d5ce42 : SUCCESS in 1h 48m 49s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1abe94c99270472c99f2a5506319282c : SUCCESS in 1h 27m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e42a6bd770494325b6df25c4115da26d : SUCCESS in 2h 27m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15072cf503884b2d8bee4b01fdef6dc8 : FAILURE in 37m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8c8ef0b1e89439ea2ccf889a7d81ee2 : SUCCESS in 1h 46m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/43484f1594aa4046b06ae6fcbb8740bc : SUCCESS in 1h 40m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bef9370ce9b4f0d85a3e18335748462 : SUCCESS in 1h 26m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff69c0c1d21d42e3ab8917d221bd6cbb : SUCCESS in 48m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2634278000e6fb6b2043c838bd793ff143e5e530","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 20:39:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"42573c93aa58bc4432e4ba796cdf4bc0f01177df","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 20:39:29.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"6fa1593d74dd612f31e62fe29cb19650f0f764b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 23:04:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d42f212700c742fa9954b40b87bb27aa : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e8a214d1754527859e8fb7655a3a0c : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd038888061748c1887d5fa572ea7f59 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bbb14deca114424a05a9dfc706ca945 : SUCCESS in 3m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e9fd1e3986a434db1fee3d03f2580ab : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dc52435e3414ba9b18095e3a808bfe9 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fadc67f2c64b480a993887260cbcc0f1 : SUCCESS in 6m 10s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2e7669b3e14d45388e58e8ce0985a7fc : SUCCESS in 1h 37m 29s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b3e159b2a1fa4d7cadd07550143d9441 : SUCCESS in 1h 19m 19s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2663360146c54f0498c669f8a280e7b8 : SUCCESS in 1h 40m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f5cc435e485b49dcb143a6519deba8ea : SUCCESS in 1h 25m 25s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb03b9aacc6d44118eb0bed8ad544bf7 : SUCCESS in 6m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/83f7216ab0a34421937f4c5191abab19 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/741eaaa84a134d5e85acb6dc0964979e : SUCCESS in 7m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bf46d133f514abea78c516242a35240 : SUCCESS in 1h 25m 35s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/65d0df13d4634a39b37af23582ea965f : SUCCESS in 1h 04m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96f5f205a3704c43acf4886166a36334 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/98b60b4b2bae434c96f7e5aaf8eaa314 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22e2792017e04f64a9d8c27efa256d2c : SUCCESS in 1h 56m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f6cf4cbf1c08490b8071c2a2cad09412 : SUCCESS in 1h 37m 02s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/068efde967a3409ea90531bfac43083c : SUCCESS in 2h 07m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/937a8ae3dbe1474f95a003c336bb8291 : SUCCESS in 2h 19m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c73741e739d44db0950d54e3b53a2443 : FAILURE in 49m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/223d4520571e4e42bea01feb8c4bc9a6 : SUCCESS in 1h 35m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/acfabe23124e4f40a4540e9e932e7f50 : SUCCESS in 1h 15m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1234f849fca4cc9bbd383155cb3157b : SUCCESS in 1h 14m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18613f3797d6480eb29146380fe2f6fd : SUCCESS in 47m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73a94166719599b43d2421eb121c708e8e833242","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-08 02:06:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a835ae4c8a054e9dbeed56104e0d86620f444935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 02:06:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"561531564d58b57661905772401c51952437eecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:53:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/067b41a8c3464aae9d0d3f69b87d807f : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6333adc3bc84025933070bd35073cfd : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a73fbba1f3464481ad6254a12955d15b : SUCCESS in 4m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/09fa85b3c64f4edd95eb4158cbafc11b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d7e30f98060440daa934524facbb4e8 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b34b29bc11144338ac50083b992d4d9 : SUCCESS in 5m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da4b0806d08b435094022d76f2a0616b : SUCCESS in 5m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3ca4ff65c93c442b90193bb865ac66ca : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b0f19e858234c288ee6901589ca6688 : SUCCESS in 8m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d0e39d83737417d97d40a3eaf7e20a4 : SUCCESS in 1h 20m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fde447f48c4d43198a313b3d95bdb38e : SUCCESS in 1h 18m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/300264c734484907abe9f965deaff3d7 : SUCCESS in 1h 21m 33s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/040dfebe963d40c9a895b6405330a139 : SUCCESS in 1h 17m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e4e7542175d4dadb43a255cda7aed15 : SUCCESS in 40m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/57935b5543c9443aa7527018f5fdaac4 : SUCCESS in 48m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/adb273d208d54f09954c97bbd69fb7a8 : SUCCESS in 1h 26m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2f32d9f5e9614ab0a94aa6fd01572929 : SUCCESS in 1h 41m 46s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1cc9e205ef4146ffbd3832443669a414 : SUCCESS in 1h 38m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9bbba2f74b954b0f82aaf1d8751ebd60 : SUCCESS in 1h 34m 11s","accounts_in_message":[],"_revision_number":2},{"id":"da4ac9abe587bcc1a25588fd762861dfe298d9f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:53:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"70cced21816f96911d3ee8ec1a259f2f00e6c640","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:54:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e0847eed9f498bb524540c669c18b1 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f652a7d1b58456aece8f88881d95c2e4496c6e50","revisions":{"f1df4b201b8b0a37be20ea5e3c6241aed31c915a":{"kind":"REWORK","_number":1,"created":"2019-11-07 16:49:19.000000000","uploader":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"ref":"refs/changes/92/693392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/693392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/693392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/693392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/693392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/693392/1"}}},"commit":{"parents":[{"commit":"e51752e43305cc2d7446aa4af89378878708e73e","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e51752e43305cc2d7446aa4af89378878708e73e"}]}],"author":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-06 17:24:30.000000000","tz":-480},"committer":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-07 16:49:19.000000000","tz":0},"subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","message":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints\n\nCurrently the keepalivedlvs_query script calls ipvsadm -Ln --stats\nto query the local lvs for connection information. If any of these\nvalues grow large enough they will be abbreviated with human-\nfriendly suffixes (K, M, G) and cause the get_ipvsadm_info func\nto raise an exception when it receives a non-integer value from\nits command output. By using the --exact argument in addition to\nthe existing arguments, we can ensure the output is always expanded\nnumbers, per the ipvsadm man page, and will only ever offer integer\noutputs to the get_ipvsadm_info command.\n\nChange-Id: I2e8c0be2221c0c23b752fdf2cdff065cddf830a5\nStory: 2006791\nTask: 37331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1df4b201b8b0a37be20ea5e3c6241aed31c915a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1df4b201b8b0a37be20ea5e3c6241aed31c915a"}]},"branch":"refs/heads/stable/stein"},"f652a7d1b58456aece8f88881d95c2e4496c6e50":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-07 20:39:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/693392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/693392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/693392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/693392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/693392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/693392/2"}}},"commit":{"parents":[{"commit":"e51752e43305cc2d7446aa4af89378878708e73e","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e51752e43305cc2d7446aa4af89378878708e73e"}]}],"author":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-06 17:24:30.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-07 20:39:03.000000000","tz":0},"subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","message":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints\n\nCurrently the keepalivedlvs_query script calls ipvsadm -Ln --stats\nto query the local lvs for connection information. If any of these\nvalues grow large enough they will be abbreviated with human-\nfriendly suffixes (K, M, G) and cause the get_ipvsadm_info func\nto raise an exception when it receives a non-integer value from\nits command output. By using the --exact argument in addition to\nthe existing arguments, we can ensure the output is always expanded\nnumbers, per the ipvsadm man page, and will only ever offer integer\noutputs to the get_ipvsadm_info command.\n\nChange-Id: I2e8c0be2221c0c23b752fdf2cdff065cddf830a5\nStory: 2006791\nTask: 37331\n(cherry picked from commit 0682fb977a76d356ede20f516c849298d34eb7fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f652a7d1b58456aece8f88881d95c2e4496c6e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f652a7d1b58456aece8f88881d95c2e4496c6e50"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
