)]}'
{"id":"openstack%2Foctavia~693393","triplet_id":"openstack%2Foctavia~stable%2Frocky~I2e8c0be2221c0c23b752fdf2cdff065cddf830a5","project":"openstack/octavia","branch":"stable/rocky","topic":"ipvsadm-exact-stats-stable/rocky","hashtags":[],"change_id":"I2e8c0be2221c0c23b752fdf2cdff065cddf830a5","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","status":"MERGED","created":"2019-11-07 16:49:32.000000000","updated":"2019-11-08 03:54:36.000000000","submitted":"2019-11-08 03:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693393-1573185198195-d1912dc3","meta_rev_id":"dbed15631bbf8ea05885b467ad64d1ae4a6d6705","_number":693393,"virtual_id_number":693393,"owner":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 03:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-11-08 02:06:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-11-07 20:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-11-08 02:06:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 20:40:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-08 02:06:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-08 03:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05f73e26243f06592113a0234d6b831cfbe856dc","author":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"date":"2019-11-07 16:49:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5a0d2ee98b89c0d2c7d9759738cfe36449a3efd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 18:50:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9901ba66f5a045f39dc0efff03062656 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b97e89595f34b958dc63ba49cf9b6c6 : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb95935d5aeb4aeb8e2eab49dad958bd : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4ca9c7900f6749fd9dfaec52adff8299 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ddefbb1855493597619e7636acd4b7 : SUCCESS in 7m 15s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/09a7be30c12e4886a64c3ae1ab071208 : SUCCESS in 1h 14m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/02502bfb1e1141378e6bd08322bf2a0b : SUCCESS in 1h 29m 53s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77f63f06060c4bb2a7621db3b4e8f0de : SUCCESS in 1h 24m 43s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ce9c3d59cc274ca08e3ff815d28e51f2 : SUCCESS in 1h 44m 01s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8045fa988b3248bbb189c94987ffb014 : SUCCESS in 6m 04s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/76f84302e00a4eea901fbbf7ec3da5d2 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2889ca52507340119f23e6a8aecd1bba : SUCCESS in 5m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc84eba965594d668a5b05a1813bd69e : SUCCESS in 57m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9a510fc459f44d12a38f4ed59f0648bc : SUCCESS in 57m 04s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ccae8fe7c6d54d67a0d7f3e3bd9f76f6 : FAILURE in 1h 21m 53s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/dde963ea73954e4080152e45956ad6bb : FAILURE in 1h 19m 43s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c8828e2be8b248bfa6d97b5481de5c94 : SUCCESS in 1h 08m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e97eeafbc78644fdb984a7d07b6b6e8f : SUCCESS in 43m 40s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/0448e264894446629224ca5bcb39279b : SUCCESS in 44m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf8493e4e9d940e29e4e93c39948e48f : SUCCESS in 1h 34m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5747f6404ae449d19bbe7364053c9f98 : SUCCESS in 1h 52m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a7292fe2e35454bb876daf2d7a2d16f : SUCCESS in 1h 18m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33202d5f9f7a4166ab3a42fe6a2c8077 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"40c8afac4d136839c30e3aae0766d3657c35685b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 20:39:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a3bfb384fb349b4ef6d9045abfdf91f41fb8c53b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-07 20:40:06.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"6aaa7f722c5d62db9cf38396d8becc119def1e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 22:30:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c2acc7ea80d41a0bfa0be2f64f46e9f : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9ad0aeef46473d8f1b546e28dc3647 : SUCCESS in 7m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d937278185a405a8bc3284e0111680f : SUCCESS in 6m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ec2b2acfa98d4115924ec5fac47ed82e : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a4a1051a044c35b6064b0b300d4473 : SUCCESS in 6m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/15064c7181ac4bd8ae584662d3b0c2fa : SUCCESS in 1h 37m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3affa9a042d44d1c84ddc773bd4b8da1 : SUCCESS in 1h 19m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92766b868ed5482da4f203d57f223469 : SUCCESS in 1h 27m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4ee31b204ac642dbbc651538c1ca63bc : SUCCESS in 1h 08m 34s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dda80b47a3ff454bafe48d4ed1dccbf0 : SUCCESS in 6m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4683a102d0a84cbfab17e74f4cc1bdd0 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fc85ccb25624de6a300c6cd65a86d0e : SUCCESS in 5m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6eafa50f19e3414398a312b53ccec2f7 : SUCCESS in 1h 02m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4e41199c7eca4337988d7b805f8292e4 : SUCCESS in 1h 08m 25s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d8e79c011a7f4b6489bf8ffab3483a2b : FAILURE in 1h 19m 46s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5d6246c136cd45559f3530b87a93f87f : FAILURE in 1h 44m 33s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ef12968fd7164bcd818c297475932d1c : SUCCESS in 1h 04m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/730191c53a8942b48de7b796db874747 : SUCCESS in 43m 14s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ef5aac35cb7145adbb712616b78f45a9 : SUCCESS in 50m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df2dbe7bd761483a879cef147af742ff : SUCCESS in 1h 36m 11s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5942b09d572f4103997962c56c7b8596 : SUCCESS in 1h 27m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b892192aa9f24d6e9abf55f37e4ac2b8 : SUCCESS in 1h 30m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59a79d14e01442d7b75a192ea16665bb : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":2},{"id":"63db349a5e484f6e626b7ac2baff1a34394d5c45","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-08 02:06:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"961b5a4c6537e1d62e2146e1403ad62de75cada0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 02:06:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9663788225b10ef26a424f9714538daa5122abb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:53:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5559a59492b842a2b114e7fbf8a83339 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e0de69c9f0400184041a5e62ffe7bd : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/daf3b9b4835a40c9944aa99aea94213b : SUCCESS in 5m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/344e3b4576ca4c8b826d1b57b0e1cd71 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0e7f6db0ef46cbb500fb6d7bdc777f : SUCCESS in 7m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b328372e30d64200929cb696ab061313 : SUCCESS in 4m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1f134670e809427db341adef5649c1dc : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21e4e081d61d4be784441c1af8611389 : SUCCESS in 5m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4bc5c2f20265473b9aa79dfd77c63d4a : SUCCESS in 52m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b30823caf54e47f886dab1330068a486 : SUCCESS in 49m 49s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d4553fb704244246a4be6931dadd50c0 : SUCCESS in 1h 27m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a65b2f5f8ad149d8967d66894518ea65 : SUCCESS in 40m 06s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d4378e2f39544c37addceb0018e95d0d : SUCCESS in 53m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/143dbbf23888424f92a27b7b02a35498 : SUCCESS in 1h 27m 39s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/63bddc2ac17a4b989ff314c8170183e7 : SUCCESS in 1h 24m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e77cf7fde69405d964bba7c3abecb7b : SUCCESS in 1h 16m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f66b986083d243a4bd87e80b8d55f3c8 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5a7735a592b78e56cc119adcaa9031430ec18305","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:53:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c0ba6db5993cb628a6302f3bfbae2b17c4e70e66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:54:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445d511eebb040b78a7b96104a3255dd : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5aba906a1105018d222ccfb61716f15518cb072","revisions":{"3d8d1c9dae6236d485b358aad4e417be0af6495b":{"kind":"REWORK","_number":1,"created":"2019-11-07 16:49:32.000000000","uploader":{"_account_id":28502,"name":"Colin Gibbons","email":"cgibbons@blizzard.com","username":"colin"},"ref":"refs/changes/93/693393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/693393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/693393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/693393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/693393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/693393/1"}}},"commit":{"parents":[{"commit":"43d3e1bde3e1257b7e3622eb222d4ab6a47114cf","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d3e1bde3e1257b7e3622eb222d4ab6a47114cf"}]}],"author":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-06 17:24:30.000000000","tz":-480},"committer":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-07 16:49:32.000000000","tz":0},"subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","message":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints\n\nCurrently the keepalivedlvs_query script calls ipvsadm -Ln --stats\nto query the local lvs for connection information. If any of these\nvalues grow large enough they will be abbreviated with human-\nfriendly suffixes (K, M, G) and cause the get_ipvsadm_info func\nto raise an exception when it receives a non-integer value from\nits command output. By using the --exact argument in addition to\nthe existing arguments, we can ensure the output is always expanded\nnumbers, per the ipvsadm man page, and will only ever offer integer\noutputs to the get_ipvsadm_info command.\n\nChange-Id: I2e8c0be2221c0c23b752fdf2cdff065cddf830a5\nStory: 2006791\nTask: 37331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d8d1c9dae6236d485b358aad4e417be0af6495b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d8d1c9dae6236d485b358aad4e417be0af6495b"}]},"branch":"refs/heads/stable/rocky"},"d5aba906a1105018d222ccfb61716f15518cb072":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-07 20:39:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/693393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/693393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/693393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/693393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/693393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/693393/2"}}},"commit":{"parents":[{"commit":"43d3e1bde3e1257b7e3622eb222d4ab6a47114cf","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d3e1bde3e1257b7e3622eb222d4ab6a47114cf"}]}],"author":{"name":"Colin Gibbons","email":"cgibbons@blizzard.com","date":"2019-11-06 17:24:30.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-07 20:39:53.000000000","tz":0},"subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","message":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints\n\nCurrently the keepalivedlvs_query script calls ipvsadm -Ln --stats\nto query the local lvs for connection information. If any of these\nvalues grow large enough they will be abbreviated with human-\nfriendly suffixes (K, M, G) and cause the get_ipvsadm_info func\nto raise an exception when it receives a non-integer value from\nits command output. By using the --exact argument in addition to\nthe existing arguments, we can ensure the output is always expanded\nnumbers, per the ipvsadm man page, and will only ever offer integer\noutputs to the get_ipvsadm_info command.\n\nChange-Id: I2e8c0be2221c0c23b752fdf2cdff065cddf830a5\nStory: 2006791\nTask: 37331\n(cherry picked from commit 0682fb977a76d356ede20f516c849298d34eb7fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5aba906a1105018d222ccfb61716f15518cb072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5aba906a1105018d222ccfb61716f15518cb072"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
