)]}'
{"id":"openstack%2Foctavia~693486","triplet_id":"openstack%2Foctavia~master~Id31123cdbddf63315767312add9f4a0affb58901","project":"openstack/octavia","branch":"master","topic":"drop-py27-support-devstack-default-py3","hashtags":[],"change_id":"Id31123cdbddf63315767312add9f4a0affb58901","subject":"Make octavia-grenade job use python3","status":"ABANDONED","created":"2019-11-08 06:52:33.000000000","updated":"2020-08-19 21:13:13.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a90c3abbe4c24900742188026629306b2fe734cb","_number":693486,"virtual_id_number":693486,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-27 03:12:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-27 01:34:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 18:20:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-21 14:38:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-24 14:48:35.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-03-27 03:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4974e0f301bccdc7142277dd74f9c904babb1d9c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-08 06:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6efc04811096561eb5b45b97c5187d43d321a76a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-08 06:53:23.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027m pretty sure this will fail...","accounts_in_message":[],"_revision_number":1},{"id":"c6c34b21d592f87313595aa0b52d239b61f93c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 08:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f5f5bfd545c471a80fe864fcd569a09 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca9eb890a4704834b949cc67ee1474c8 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2df681e646d409098036af9a8bf665a : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa54064ca023419e8b43b180d3fd4337 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d4edc1c9d7405794bc98254427f5fd : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b24d7ee065474f9d9601e3997a411187 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb698a32323546c69131c09c31d548c4 : SUCCESS in 9m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/090f10a7c17c41c082c22b6920dc29c1 : SUCCESS in 7m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/db6af8c6fc5941f5b0026843b1c1a72d : SUCCESS in 7m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/490dc555b9c24272b49112b625f4e3eb : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48ec0b3b31f843a3a23d8d6f652f5589 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9cf4d4712ff40ea94c60c5f9f65e088 : SUCCESS in 40m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9b2234c2528e4f8a94a2a771066a8766 : SUCCESS in 43m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a56264d11572486c8c8dd9b3cef458ba : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a6ab011db2c24d828e2f939bf9ecd170 : SUCCESS in 1h 35m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eba33b4450e54b06a675335cbea03406 : SUCCESS in 1h 57m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b919c7afc87a455bb6817569a4ff7b53 : FAILURE in 31m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5b52ea148764e73a59562a9edc6ad2d : SUCCESS in 1h 45m 26s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8f1f50a7ced48d2b0421947a418451a : SUCCESS in 1h 00m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a893fc4b751642519971d2fbfc0137d7 : SUCCESS in 52m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e45563e4e8c24e78914d3f2f35f0a244 : SUCCESS in 1h 25m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96672e62c4c5401389c11c766c0c3a28 : SUCCESS in 1h 07m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/62d416ef44ba4774827e081acb5e6003 : SUCCESS in 1h 06m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d2e27aacbe26494ca3e969d260dc4402 : FAILURE in 1h 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e927098dd988df08dbdc9d59edf91f8d9e34f083","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-18 15:55:23.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"80d3e1304094da48d906d4737c65d6142ac284fd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-18 18:20:03.000000000","message":"Patch Set 1: Code-Review-1\n\nThe job is not running the octavia grenade plugin.","accounts_in_message":[],"_revision_number":1},{"id":"cd62443cf801bb3ef68fd0ff0b72afa64b1d6435","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 01:13:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb532be2dd2c50d87f49299eb7c55aee92f5cd20","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 01:14:26.000000000","message":"Patch Set 2: Workflow-1\n\nYeah, that didn\u0027t work as I thought.  I\u0027ll see if using python3 at least works, will have to look more at removing the legacy jobs completely.","accounts_in_message":[],"_revision_number":2},{"id":"ca267166796c28af45d475ff01812fa2b618d51b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 02:52:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d869b66b9efe425f884633d52de826bc : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a28f5577bc5b4e86853fb7a825670f95 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e1fbbb88ea04a7c9b664e2517696626 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/304d696b33144baba99698798d5235ad : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fec51c104000416ba6544c13729b9a22 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c5079bfe3c34f699508605589cbf46c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb228e7dbe14edab4eccb53eb09bf80 : SUCCESS in 10m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2644c72f94ee4ccdb9769c8674d08a66 : SUCCESS in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/055e49fc09954cd280d051ecd2835bb3 : SUCCESS in 7m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f34ee7523a1345a9884062a4ec179aa5 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1ba81d10e9b4658bf4d34915c681142 : SUCCESS in 5m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19c2667312b0467aa79e33137a8e6cdd : SUCCESS in 40m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8d5c99fec58042089908c1080d4a2614 : FAILURE in 14m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83bbac6d70cf4ed49420ad9bc39832cc : SUCCESS in 1h 30m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3bc8e030fb4c438aad641cb68256539e : FAILURE in 15m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c153267f2d2d496fa2ee8a6cb7fae9b9 : FAILURE in 22m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ef61c9bfd69485fa607df483a7102fb : FAILURE in 16m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8544f99a65cb4c429d8aad0f96265e41 : FAILURE in 18m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ff26d6505be4a798cfd8e15dde47d8c : FAILURE in 50m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3155c053be0b4c14810791b29d75d224 : SUCCESS in 46m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e563bc4452ff4539a3dd98ee07a36850 : SUCCESS in 1h 24m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8ab76bb8ca240ee8f28d5a955886418 : SUCCESS in 1h 16m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ed90639a7aa241afafe49469df9ce64f : FAILURE in 15m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/284d7e89deec43ba8aa90d7b9335f9a8 : FAILURE in 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1eca90055f4f53729508cbccc1f6498310a6db3b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 19:36:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b5f644f22b137a5e2a756365434418fe0ff44ac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 19:36:20.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"5fa7a6b5a2feed16e80f65ebc1fa51a08c26dcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 21:50:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9301355ce614c80b9dc7bb8f9e02271 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5031d94da4b9457eb4952d3770c2a5e2 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4cbb054dfcb4f78a2ed92e34a77d9b9 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4771ea09f52448968b7e5d2b74b9a88e : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09dfe0983c4847d283107e23971b080b : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02d0b4327e1d4edba8eeeaa08a3d56cf : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca711ae20374afab270c7558e5a315c : SUCCESS in 12m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5efd7954853d4d708b4615ed0e15d067 : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e6b049b28a8f4be9ba7a2e4c8afcde6c : SUCCESS in 7m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8fd7747396f14fe5b8dabbb3d931c116 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f728095ac6be40bf98392f01560af7f4 : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e7e1f32f5f5544eca7487f80e5d1a0f9 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2e731cacb4d843268df40b05e8449981 : SUCCESS in 43m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea78d1967225441b8da18c37a4268f0f : SUCCESS in 1h 44m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/862f8b991a8f43eeacbd96e8df10619f : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/12315c9746394fd197be0c61f36d197e : SUCCESS in 2h 09m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fdad4979c9f84cc8abf8549f59b15da3 : FAILURE in 32m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b894c93d8b9c4572ad64868797d6ef06 : SUCCESS in 1h 51m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7219999c9ea4f71b3202d1d11313502 : SUCCESS in 1h 33m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3e2f7f6b20145b6b220d5a1b4f7c665 : SUCCESS in 53m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dac754e867f9425bb6058d0e2b0fe0d4 : SUCCESS in 1h 45m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de8cd4d1c0374e8595ad2bc240f73b2f : SUCCESS in 1h 14m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ada115ba992e4517bf230bb4623f7484 : FAILURE in 1h 17m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2d1e07946cf749d5812f6a3204a90f18 : FAILURE in 1h 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8572da3681fab44574290d7ace8aee884e7ed803","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-19 23:04:53.000000000","message":"Topic changed from grenade-py3 to drop-py27-support-devstack-default-py3","accounts_in_message":[],"_revision_number":3},{"id":"d1a19648db8717a5e56df9208a8e00a4002efc9f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 07:59:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21530c9ac74464bf8ea17c4e02e338a2bcbf6b31","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 16:55:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"281c10028e68abdc79c937db45457159fb4c74d3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 17:00:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4957956b3bdac0c59881e0d2ea4f67f09eee0ffb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 17:09:32.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"58222b5a1c2f222027af6c69f6ade9fc418f42de","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-20 17:57:28.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\n+1, let\u0027s see the result and we can debug here. py3 should work for both either legacy or zuulv3 job so making it working first is good idea.","accounts_in_message":[],"_revision_number":4},{"id":"35ccfb9d99e5e64069977c490681fe6669a62443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:19:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/becc257bac6146b08bbceff5980365f1 : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90179c91c37a4696bc9e851214ec24c0 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49786edc91e4fac840ee782a9ae7c77 : SUCCESS in 9m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00b3fba01ec8478d9ed8d83dc8a482d1 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69608dc9a1c74fc897fee754cbb2b89b : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2706caacb6164f2bb4aa0d32cb0d44c9 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7318c2764807468ab3990c299eb700ae : SUCCESS in 10m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2269d76a636e4dc59a8cbcf8df318b03 : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b3193eb6c9914c5a88c21f3d04f5581d : SUCCESS in 7m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/436093042d4043bcbb910f24fbb55ecd : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/992d06abe72644b287e27f058d447bc1 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d845bd021e4a4cc68a41ecada96e116e : SUCCESS in 45m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f893867857f8434ba1b47d5b1ae26a79 : SUCCESS in 53m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9eb19f5a62944f75b1ede21f40d2b603 : SUCCESS in 1h 39m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d5b092752df34ec3b0355fc517d9406d : SUCCESS in 1h 52m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/aa11912780b243d4aaa37dc41d2955d9 : SUCCESS in 2h 10m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1587d000e2264213900f969756203d58 : FAILURE in 31m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/102262d5488f4332bc551e8c83ba94a9 : SUCCESS in 1h 57m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/323afb9d5fd94f62b0bba4bac6004442 : FAILURE in 57m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5cf93c7aecf747aabe0d4ee89f7ab9ad : SUCCESS in 1h 03m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9807de076aae48549f00f2dd8fb66e6d : SUCCESS in 1h 57m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61a5cda198124f08bde7a829e622b4de : SUCCESS in 59m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6ee455f487e946a49f3759d3e99aaf3d : SUCCESS in 1h 07m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/846c1cc0545c40e8a62653e186e2db18 : SUCCESS in 1h 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4bdadfafd43a0748cefe57667ba3ba7feacc8380","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 20:10:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fcb352c5ac58cf4f1957d6f32b1e951280884b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 22:47:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56614ee761cb48f3853c03f5deec81ee : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/374e3bd8153e448c9c5c290b1878d2a4 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebee9d0dc9d3427fb14e46c5534581ef : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/473895343ceb4a28b9f4a2dd9a8d13d2 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/160d9e4b435b4c6e9ee9680cb16df4b6 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d58d71edc295434e915a8d8f51e495f0 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3533e0c3b14d3994294dedaa23bdf5 : SUCCESS in 12m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2ce19a177f48431e9aab76f238cc73c8 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51126a31a3114c1da24833076375a7be : POST_FAILURE in 33m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e90e59648b8e4a30a4a0dcb3e87a860b : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ae88776d58f42cf9cc2791365e64e71 : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84b4c9d5fb2a4131ac4569f41f72ddd7 : SUCCESS in 49m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/329b5eb5a2b3482b9a4d268aac1565f0 : SUCCESS in 51m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f778cfc342764a0da16ed2b598320a86 : SUCCESS in 2h 11m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5af2f905e4554570b23dbddb94e0946e : SUCCESS in 1h 53m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b342b17acdf140e28f971fd8ccb1b4cb : SUCCESS in 2h 33m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9cbeec31695483985652824d75e5c4b : FAILURE in 32m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/78995addde00463fb72534b86d3871ca : SUCCESS in 1h 44m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b26bd7fc82d94d4cac97b5481623b629 : FAILURE in 1h 05m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0dd41599c18445b283cf69b3746fdadc : SUCCESS in 51m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5b3499fd0554ce88a9c134dba0249f5 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75a38d43dea446dab1e3e379f0de13a5 : SUCCESS in 1h 06m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/300aa7f752554b04a020dca6684e4d52 : SUCCESS in 1h 05m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d22031a2dab54f0c8d54fc4351e11357 : FAILURE in 1h 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5632ea26756eba81565e695cc11160333250f938","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 01:07:22.000000000","message":"Patch Set 4:\n\nseems like host name is \u0027openstack\u0027 here which then raise error. \n\nhttps://zuul.opendev.org/t/openstack/build/b26bd7fc82d94d4cac97b5481623b629/log/logs/grenade.sh.txt.gz#30398","accounts_in_message":[],"_revision_number":4},{"id":"690f644e0b8642334755ee126aec2b084e823430","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 01:15:38.000000000","message":"Patch Set 4:\n\nprovider name seems incorrect \u0027openstack:\u0027\n\n- https://zuul.opendev.org/t/openstack/build/b26bd7fc82d94d4cac97b5481623b629/log/logs/grenade.sh.txt.gz#30397","accounts_in_message":[],"_revision_number":4},{"id":"40ccf811a91e3a977552ccf188f8fe967a3187e8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 14:38:24.000000000","message":"Patch Set 4:\n\nrecheck as devstack default to py3","accounts_in_message":[],"_revision_number":4},{"id":"48b6d20b18e587fc3948754d1d5f57c30e7598ef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-21 16:24:11.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"adbadd6c712515d532a533c0c3a27688193608e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 18:22:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3602bd0ced7641908b0121340ceb7e5e : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb1058bffb1c47be992bba325e89210e : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b27263398d458590d9909e71131ca5 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df104256e7db4e5ea26af49035d265ae : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9c12edf955b42b49140fe225466c072 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07ff119c03f64fcc9205a03b6a489a46 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385264a836ff45f0a236f7b5cd88a01a : SUCCESS in 10m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8fccc5bf855b447da6071151532bd9ea : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/88b214357b16407a91499be27d90930e : SUCCESS in 6m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40acaf39735f425f9fea21f1fa527c9e : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd5460ef6c774da5956542c821e8b2f4 : SUCCESS in 5m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/755e22d9412c4c0faf1c631965bb6ab3 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b0994253a549458990cace3493050bed : SUCCESS in 51m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59608e2bcfc04216a521fed9c6033c72 : SUCCESS in 1h 28m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8e66fa031b1d412f94885f0238521d78 : SUCCESS in 1h 30m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4f1bf99585c34e8d82246589b81a1082 : SUCCESS in 1h 54m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0258c66bee63454a996f2d37ed6941c0 : FAILURE in 35m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa4e1671ed8547fda7cc5ab9ceac3bc5 : FAILURE in 1h 37m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea7cb56ea044446fb326e8d5bbb90239 : FAILURE in 54m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a4f216476fe24675824620e805929417 : SUCCESS in 55m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/21b635dc9c7e4993a7075bc3484c1a41 : SUCCESS in 1h 45m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/862fa70df7e04df8b9193bc16ec6eeb0 : FAILURE in 51m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2f029a2e12e446769638e347ab9bad94 : SUCCESS in 1h 08m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4942cbc95b1e48929f2a159ae55c2380 : FAILURE in 1h 29m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c359824264c7f13e0a98f725e52cf8a252ad0056","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-26 14:34:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"dfff078bb6e39a8fb046f99c4de5b1fbabecadea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 17:17:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d271c69c19646ec863b4a47f0ca1b3f : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2b9bcbb1ab348e9b02e6bc2582237ab : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/586bf0a8a0184b10b3bcd8c8384d21fa : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/854eeabf1948438cad36c7927fd66a6f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7532d85b85f41619afa5e357431860f : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba32931e546740aab0de2a70dd9ec4f8 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf049a64269494b8c7ff9c2be6b6699 : SUCCESS in 9m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f7779983a6a242d1a23fa665a51830d1 : SUCCESS in 4m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c33a85d7f26845b79b52a45e743323e7 : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ac461019a9346f7984dd2f25c02fe23 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3e17882b1db4ff7b2a5fe8c91cd7c28 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e04134945fee4e36832c2dabae9e888c : SUCCESS in 43m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7f925ae373e24831ac78b842aa3f2a2f : SUCCESS in 48m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1110701e6dd3439ab48c0c346315840b : SUCCESS in 1h 48m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4b33ab00dde544dba89b0d5c2dbd49ba : FAILURE in 1h 53m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f8768aeecc854f2b9268417be913f5ed : TIMED_OUT in 2h 36m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/26671f8cfa7743259d6b232e53fe5ff6 : FAILURE in 38m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b6c54a2f3e047cb8419ce29dbd7310c : SUCCESS in 2h 01m 01s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/028b257155af401596aded4895e92ed7 : FAILURE in 1h 12m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33089bd91783481988a4c3263372b663 : SUCCESS in 1h 14m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2b4333fdca2e4016aa5aebe6f27485bb : SUCCESS in 1h 55m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30f04b66894d4c059997c2317630b805 : SUCCESS in 1h 31m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57c6c557e4aa4371a8eb0f9381849dcb : SUCCESS in 1h 34m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46591549f3ef4a9c917b35c2b67c2121 : SUCCESS in 1h 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cbe29708846f8376387cd5d741ed02b43b974026","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:04:58.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"73118b1ab42219e5515a7320572ee8b2cf121a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 21:40:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daecdc13519e48128c3fb573a29c7fb7 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9306b864da6a4dfda280c6438730fbbc : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4a074aebe84baea9e1d0868f605801 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ef15c1a98344025b0720381f61e7c0a : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f2f0fe9999748a68ec67787a80f0a5d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0269af33e8684954b5ceca325ac994f8 : SUCCESS in 10m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ac6df1f1cb4845749e378cc98486545e : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/589964e8d266499aa82717e023d8b620 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ecb7723be84411f8d1d83e88715733f : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4aac63954e0b474abffed9d961224e86 : SUCCESS in 43m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbc3ca7ce37f40348eede6da5f4224f9 : SUCCESS in 1h 24m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1441a96dc4aa49a5924c00bf8c7c0e2b : SUCCESS in 1h 00m 07s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02ba1f1652e446e88e8513d63fcc77f1 : FAILURE in 9m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3410c21976f04670814f8c509d73c04a : FAILURE in 52m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64e277da3c734227a39a267832dce9da : SUCCESS in 1h 30m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81d7b7abaf9a414dbe9fee0e6f43e593 : SUCCESS in 1h 03m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ef16150315a49fb55bf8eebc080541015115d8e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-21 23:00:00.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"88723dc34f39ea016892ebb8bdeda86a58ff4856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:02:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd074071cda84a769a3634204d7b9e01 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d005e1347f7f4e64b4c754aefd46df51 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c01afd8117468f8215eea0d9748d8d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76aee85e8e1b487cadd750ea1a5523d7 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/027a7c8c24a7403489f8d6b082283a67 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a41479667e43c8a427b8c3d90c05ae : SUCCESS in 10m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4409889f6c644c92abb0bc6ee3ea6655 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/30e69cb6bb0a4bfdad707ef622f427ea : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0cdb8ed320d046d1af4c4507efc4596a : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9843900d5a34b9b891ca1970c8cb167 : SUCCESS in 47m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2963d799108e4d5c93af79ce81c7224f : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9fde382bc59841f785a7b8616f574f95 : SUCCESS in 1h 27m 32s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b73e62082a744764a6a1f74d00e835e2 : FAILURE in 17m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5d4afb30699443e9e4cb3637ef04c29 : FAILURE in 51m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/17c7ee93d44848129f90a9f3760d831a : SUCCESS in 2h 00m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd476a36a7dd464fb3b6b9636f3cbe7f : SUCCESS in 58m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5e2ba4ecf0123b7f240c3ab5584d822913758ace","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 22:28:41.000000000","message":"Patch Set 8:\n\nrecheck\n\nI think this is fine now, just re-checking to see if the octavia-tempest-plugin change to use bionic nodes will change the outcome of the job at all.","accounts_in_message":[],"_revision_number":8},{"id":"44787f63a5ba9086fea99a0ade9a04cad5a06a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 00:05:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d61aeecd0d94049946fb373ab9add65 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/401b6ebb5e504a21a9e9fa1fcca4a7ee : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372ea2c1adff402c81cf36f58cc39464 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9d253a81039484a8c8022b24d42c5bc : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3b37993146d4ad8b3ac3356f0baff46 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e45279d7d949b6861d234140fa2910 : SUCCESS in 12m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d5cebd9de3d6456baf1cdcf11aebec7b : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3c4cf4544a164fe286bec98ec2cf0a7e : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9fd3cce8256474e8a6bf85846fe6a35 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1fa62a3ba8b4842a4a8b7f1808b8650 : SUCCESS in 43m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03fbe36b53c744fc8bddcf547d65ae44 : SUCCESS in 1h 34m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32378bbb24444391af403a394cc8d44a : SUCCESS in 54m 44s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0a9c97000d7447a5a007ffad4aa7c21c : FAILURE in 36m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d31d62f9e3f7470a8ecf90f78267cce2 : FAILURE in 54m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6fd7f29776c54935b7b621cecd0169aa : SUCCESS in 1h 29m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/958629cbdfeb42efb7765321cd370fb7 : SUCCESS in 1h 08m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e5346496adefcfe63c3d2e2d16fcb66b8a7f4f22","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-04 22:34:29.000000000","message":"Patch Set 8:\n\nrecheck\n\nlogs show an issue starting nova","accounts_in_message":[],"_revision_number":8},{"id":"032267c8d4055df1b246ffc3b9dff6c9b0ddef30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 00:28:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/560108b7bbcf4642abd213365f4d40d1 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afafd9eed29640ab873116774267202d : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c724d3df984a349f23db51015ad0e7 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ac9c306f0941e48548355f820b7c31 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77d1332fbf734bf3bd5156b25d506822 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01299a1271084025b133d9f27bff9e80 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94e86d7f55f412e852459c844608f0b : SUCCESS in 12m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de93dc207f844ce1a310f872287dcb07 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/80f4aeb82f95446b85f7bed3b3f365bf : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fac5e4750fc04f8fa01fa90d6a0227b1 : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e27f2e30461b42c686d768082e511f96 : SUCCESS in 42m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40e8a3b3ad654d329d4f03b7cac1d1e8 : SUCCESS in 1h 30m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c760d95e548747b498902dffae0d8228 : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5359b2d227a64abb8df4cc4af76839d5 : SUCCESS in 1h 04m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afd10062ae1045a5aa9d6ae604ad8803 : SUCCESS in 1h 02m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fb18746b3ef5464ab68b766c422c2f6b : SUCCESS in 1h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f258acd9d6774031bfad082b33969499 : SUCCESS in 1h 48m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e55395ae71bf4950ad7d255b457fc9ce : FAILURE in 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/78c294f965eb4a999793b9292ae4ec0c : SUCCESS in 1h 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4f392d4dd10f6faece4f006eddfe0a6c266f2639","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-11 15:03:39.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"fac3cfbbdfeb7663a8850212d585f285cb949748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 21:44:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a17d3f939bfd4849a356fdb82819ffcf : SUCCESS in 11m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83a3359048ff416f9573402a4189cfe4 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e118a70f0849eeb14ba6ccc4957d8f : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5b259864a0d4b899e65d52c549a4f0c : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18fa8f57efa9421b94e4dcec3030c007 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc8edd6bd965483691f5ff0a5084ed5b : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d2931270de4f8a873c42959560c421 : SUCCESS in 20m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/66f1c897e32c493489cf14779b1f0cfc : SUCCESS in 10m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dc237fff3cf641a7adf91913d53dd328 : SUCCESS in 13m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3443446b36314330bb3a5cfdf61f2ef0 : SUCCESS in 9m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48b000eed2334a889540793f78e5e3e4 : SUCCESS in 54m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/234cc28bd28d415e94cc012133d08fdb : SUCCESS in 1h 54m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d65bbe7b20e4c848ab539078a34cc27 : SUCCESS in 1h 04m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1daf052b038c471f9a5e0907cfa26431 : SUCCESS in 1h 00m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc1f60fcc90f478fa04d2c473717b173 : SUCCESS in 1h 11m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f23645de56a4619a256bc5a80ba9780 : SUCCESS in 1h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ca78bf4a9fb4f25ae86533b45a27e9e : SUCCESS in 1h 20m 57s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bddb40ea1b0845bfa42c59d01ee12491 : SUCCESS in 1h 17m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d5e3978294c4484bbf290dfa01f1450 : FAILURE in 1h 52m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a90a2be9f4274aa06a3ca847900a2c1d7847ffff","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 01:34:53.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"c39d9dd731d75ce8cbce02e9cd1fd8c2a4dacb37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 03:12:22.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b56966db3643999bf289e3f26bc9fb : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/540bce6a954c48edaae8578d1fcae751 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded828fe1de640488609f5fa22a7b86b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c2eb10a462c44d3a9c357b6a69caca4 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b28e4a08361747c8bb949b8b9dfaa400 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e21781fad2354652bcc21b4e7923725e : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc1ec571bbc473e90e6d0bd697a6dbb : SUCCESS in 12m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f85b1f7bc6db4ebabc9e5ef7b50f5ab6 : SUCCESS in 7m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2691efb9f25b4235950b44ca252270b6 : SUCCESS in 11m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd5cb1ada5c94ef192aa4ddba62e1c0e : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3226203c4a044548e2f9a055910c510 : SUCCESS in 1h 00m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/249c9a40578940dba7fa6f7b9fabf4e3 : FAILURE in 1h 29m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a9f2f78d36d4e06a199d2d34d4c06ab : FAILURE in 59m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e60c85a1a20d4d34952602f4ad83678c : FAILURE in 58m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78f3b12eb1424593943ca7db75d79fb0 : FAILURE in 57m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6a05d6303b744d69ebdac30b59f61b1 : POST_FAILURE in 1h 31m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbaf61c5b1dc4929a9fa8aeda6fb51f0 : FAILURE in 1h 16m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7c50d64c6e6423e896acd8835f3ad70 : SUCCESS in 1h 17m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9cb9116647064a7d95661aafb3d1f3e6 : FAILURE in 1h 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a90c3abbe4c24900742188026629306b2fe734cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-19 21:13:13.000000000","message":"Abandoned\n\nnot necessary after move to zuul v3","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fa6d668b87a484123f8c059999f0c05a02575bc6","revisions":{"42700f3accd7142118f81565f68bd3cea314e6df":{"kind":"REWORK","_number":1,"created":"2019-11-08 06:52:33.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/1"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-08 06:50:42.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-08 06:50:42.000000000","tz":480},"subject":"Add grenade-py3 job","message":"Add grenade-py3 job\n\nAdd the grenade-py3 job and remove the legacy\noctavia-grenade job.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42700f3accd7142118f81565f68bd3cea314e6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42700f3accd7142118f81565f68bd3cea314e6df"}]},"branch":"refs/heads/master"},"833c532bbe70c9b084669f544e11dae9cf2b8dac":{"kind":"REWORK","_number":2,"created":"2019-11-19 01:13:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/2"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/833c532bbe70c9b084669f544e11dae9cf2b8dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/833c532bbe70c9b084669f544e11dae9cf2b8dac"}]},"branch":"refs/heads/master"},"291267aab2f1bb72e93bdfb92354b78c980b025c":{"kind":"REWORK","_number":3,"created":"2019-11-19 19:36:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/3"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 19:35:35.000000000","tz":-300},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nMake parent of octavia-grenade job to by grenade-py3.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/291267aab2f1bb72e93bdfb92354b78c980b025c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/291267aab2f1bb72e93bdfb92354b78c980b025c"}]},"branch":"refs/heads/master"},"74f253d12021fe630d6628b817f58179075a1013":{"kind":"REWORK","_number":4,"created":"2019-11-20 17:00:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/4"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-20 16:59:31.000000000","tz":-300},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74f253d12021fe630d6628b817f58179075a1013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74f253d12021fe630d6628b817f58179075a1013"}]},"branch":"refs/heads/master"},"8afbedaa11ab6f376547f58c727b1435a1a8b7cf":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-11-21 16:24:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/5"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-21 16:23:30.000000000","tz":-300},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nDepends-on: https://review.opendev.org/#/c/695466/\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8afbedaa11ab6f376547f58c727b1435a1a8b7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8afbedaa11ab6f376547f58c727b1435a1a8b7cf"}]},"branch":"refs/heads/master"},"632aeefd9f3bb555afd3bb84dd44f17d5fd6a50a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-26 14:34:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/6"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2019-11-26 14:34:04.000000000","tz":0},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nDepends-on: https://review.opendev.org/#/c/695466/\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/632aeefd9f3bb555afd3bb84dd44f17d5fd6a50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/632aeefd9f3bb555afd3bb84dd44f17d5fd6a50a"}]},"branch":"refs/heads/master"},"1313094681bb4f02d63e4683141bf612574699e6":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-01-02 20:04:58.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/7"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:04:38.000000000","tz":-300},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1313094681bb4f02d63e4683141bf612574699e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1313094681bb4f02d63e4683141bf612574699e6"}]},"branch":"refs/heads/master"},"4b732c1dbe8ea0fe0cad224659b36eb2d2867aae":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-01-21 23:00:00.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/8"}}},"commit":{"parents":[{"commit":"b6deb12b5dcac2b2c714d70c195d4ab4420336d4","subject":"Merge \"Allow the Octavia wsgi to accept argv parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6deb12b5dcac2b2c714d70c195d4ab4420336d4"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-01-21 23:00:00.000000000","tz":0},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b732c1dbe8ea0fe0cad224659b36eb2d2867aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b732c1dbe8ea0fe0cad224659b36eb2d2867aae"}]},"branch":"refs/heads/master"},"fa6d668b87a484123f8c059999f0c05a02575bc6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-03-27 01:34:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/693486/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693486/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693486/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693486/9"}}},"commit":{"parents":[{"commit":"73fca169abbc04ed51345e4df6f291c83e745cb1","subject":"Merge \"Exclude acpid on Red Hat family DIB builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73fca169abbc04ed51345e4df6f291c83e745cb1"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-11-19 01:12:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-03-27 01:34:53.000000000","tz":0},"subject":"Make octavia-grenade job use python3","message":"Make octavia-grenade job use python3\n\nAdd override to octavia-grenade job to use python3.\n\nChange-Id: Id31123cdbddf63315767312add9f4a0affb58901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa6d668b87a484123f8c059999f0c05a02575bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa6d668b87a484123f8c059999f0c05a02575bc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
