)]}'
{"id":"openstack%2Foctavia~693586","triplet_id":"openstack%2Foctavia~master~Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253","project":"openstack/octavia","branch":"master","topic":"fix-client-auth","hashtags":[],"change_id":"Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253","subject":"Fix multi-listener LB client auth/re-encryption","status":"MERGED","created":"2019-11-08 19:52:22.000000000","updated":"2019-12-09 23:11:32.000000000","submitted":"2019-12-09 23:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693586-1575932741641-349ca902","meta_rev_id":"8d2113803d6257f6313f54a52f31e8e88f320427","_number":693586,"virtual_id_number":693586,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 23:05:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-12-09 21:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-09 18:12:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-12-09 18:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-12-09 21:00:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-09 18:12:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 06:18:40.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-26 10:09:43.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-12-09 18:12:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-09 18:25:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-09 21:00:43.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-09 23:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fccdc330100f571dee07a6552ec9073031e64dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 19:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98ab1428965d286d449d61590d27e009cc73f200","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 19:55:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"390ada78ca9d00b62956cb8e5d2838629e67077f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 20:07:56.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"3adcd17b4b823cca839014b8312b9a8515cfca84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 22:08:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11af4f3f6c834fd7ac6097a756274798 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19454f8e7cad4aae8db70a555e19671e : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcfe88bc4c2f4b92b596080488a91ee5 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54b43662fc5c45228e905cc53b2691b2 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a54fbe4dc6734c70b3dc0ef171593d94 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a18d0da271044ac9d5cfd12e51da3aa : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d74165e7de34d028caa7d0881e15da5 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7853f006e5fa43aa8a13abaee4f03bf9 : SUCCESS in 6m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a2f74d1f36894f9c923b63c1338a2320 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2b6d72bafe4d4203a51a480687511397 : SUCCESS in 6m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/11722785f0fd42dd916346e76747fec8 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab2a4e873a9e437eb50f5b4f859fc3c6 : SUCCESS in 5m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad8aa8b1b3e44c5992f60a8e41a22e2a : SUCCESS in 41m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b0b055c53fac429bb7455b6940719d30 : SUCCESS in 46m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59b32f128e5146cdb18db0c750dfbd87 : SUCCESS in 1h 30m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b77c0dde48144a36b62ec03f65103518 : SUCCESS in 1h 35m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/67368916ea4f4332af83cc6db90a7242 : SUCCESS in 2h 07m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07308fb16eec4729b0881e91744d344f : FAILURE in 35m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d9c3d45b50840c6881ebb30000b176e : SUCCESS in 1h 56m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e011fcfa5a504718bf3fa44bd8343fda : SUCCESS in 1h 20m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6993588ea2e49a980130d7b56dacd8f : SUCCESS in 52m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef68f7f756da4a67bf6ef004535766fc : SUCCESS in 1h 31m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfe12c569df84c81ac0569bd1352bd11 : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1669dce1049648b58dc0ec935c220cea : SUCCESS in 1h 08m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3cb85ec321bd40ec8bf707c5b4782d67 : FAILURE in 1h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0bf5c80acc07194f1464d0df00fed40e7148efbc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-11 21:28:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"021d40d5b2b9c80287ff425c2a5814e0f3feb123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 23:36:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01e2d61050ec4a4e9cab8833e93efcc6 : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee7caeb57db24eb3a199d5650bc9617c : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2065da7748a4fb487c092c97e9aa742 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7736d002935a4588b07e3d6041f60d0d : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e373487fb204765b69282b574f65426 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b90c6c5d465944688437df7b80de38d8 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ca1cf669664e1abdda5e35240f9fec : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6b59f832b9440c9fc4d5db632b3a4e : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a756aeaac51b4cb58577e865bdc1c204 : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/888286ec4e9848b49f24f06e1c28d3f9 : SUCCESS in 7m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/96b860f7d71944a892e0d1d35955e7ef : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa6085787bc6451c9b299c086759e6e5 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72452ac229cd4ff6b2b41b06f413362e : SUCCESS in 55m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/49755dff97f94437b7e364a2f46dc114 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f7c13b86bb0496daa11bacb6727e4bf : SUCCESS in 1h 29m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0e4993d4c6bf4efabc492677419a3be1 : SUCCESS in 2h 05m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9d50359197604a498d2519b1e9c86db2 : SUCCESS in 2h 02m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eec691f801474df79d4b691fc1fdb123 : FAILURE in 32m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/79eb47f45fde48359056aef95896afdc : SUCCESS in 1h 35m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24adde7c51814175b068fbaed594930f : SUCCESS in 1h 24m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6bd3e45a2e7347648871606181fc1639 : SUCCESS in 1h 13m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be32bb48951245479eaad0e7381a99be : SUCCESS in 1h 30m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee2e3c98ee6b4b0caaf6b759a1c09775 : SUCCESS in 56m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/52448920ce80406f8c1b7918a190e7e3 : SUCCESS in 1h 34m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/88a9b043ae1f4842953611cf856e1968 : SUCCESS in 1h 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b275718f808d0c257dc8b655813b222a0d0e075","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 06:18:40.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"95adabd7cb64030db739d2f31f2a0bb0534f2f02","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 10:09:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c8a4b41193ebac863405f9ff8e25223deaa8811a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 01:47:06.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c228753b6a5387113c3448bbd108aa72ba8b4cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:37:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5d6e95cf16745ddb9d94bac869fbbd7 : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c47e96c9d474b19b089322ad3447858 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6075b7f24d954395afd3ad77b985d1f6 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0907be1dff5f4d66b01cddcdf1c2e8d7 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b3e01403a834812ac3f62725da09347 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28adec079fdb4e79b49fb65c4bea7914 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93520b0f2acb4cd6adee6e03550df2d7 : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d90c9c70f97a4d2483e48c6d680cbb08 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ce32ea9fd554dce8097210fa74fd473 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78723e9238c2409fb6003bb90caa35f1 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/296647ef2b524e889cb92cea1be432ae : SUCCESS in 45m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de38f198165e4920b58da9fcbf10714d : SUCCESS in 1h 48m 50s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78f4c22a9a12484185ad39c06b52d588 : FAILURE in 33m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de0e8c5b34644998a6da77401a2ececc : FAILURE in 59m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45d2057c56c34f668281d700baaf369d : SUCCESS in 58m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df3b26b8d22b4bf280131ad763bfc05c : SUCCESS in 1h 39m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36333d5e63dd4b23b67c348838181607 : SUCCESS in 1h 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1a88c50ebf58f10962e4b31c889c81ab9d95243f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 19:26:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a6f37ddb899159523fd93534c0904fb57e6d188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:30:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9609a383a434661a112ea106b80eef3 : FAILURE in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/495f2314aefc423a839a90227738344c : FAILURE in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41bb443e3fe44c7eba417f65f3e6a2d2 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51a5b10c7448436bb64ec01e7576ce26 : FAILURE in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ecc1e591feb48d4a6427ca3789aaee0 : FAILURE in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c819e1955641e5a60935eeef70f572 : SUCCESS in 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e1545b5a1684b2d81242438c64b698e : SUCCESS in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5a18bb6af53647d5a603f3ea382911aa : FAILURE in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9a6a626859d94e9d9f98ad46b6c31ac4 : FAILURE in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2af08d177ef94052b3861fde5a2bbd75 : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3baafd267c28470aaa523b6eee784b95 : SUCCESS in 43m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4bb087725584cc4ab5db2a9cc815e86 : SUCCESS in 1h 34m 32s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cb9c17c2ff3d40309c13172caff60c27 : FAILURE in 37m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b02a010d7f1e448fabaa415613cd4900 : FAILURE in 1h 09m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b7685eb3e874dc388de9ccdce6a0f68 : SUCCESS in 59m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/370c934375ec4b71bfe7920ed81414f8 : SUCCESS in 1h 44m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64036e7498de44639ff0bddf64b3bef8 : SUCCESS in 1h 07m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"66daa6015f3d3365785b7a27da79dab9ac981841","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-09 15:49:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"39d0e31c0d0171d95c971ecf97520a96554a086c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-09 18:12:59.000000000","message":"Patch Set 6: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"25f2e76289871bae6adabae381513e4164a01e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:14:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/603c60ab08f644a9bbc6a04473ac37ff : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a483ed1080374dd39a97ed6765bc37f3 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d4cc48ae624ac5ace707c633f60ca9 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75dac23066834238b30637b74442b657 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43b261b9b75e4f688ffab3d20db7ccbb : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7350f41a3a4087b16544d7c2a4f31a : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d60f5db87d3b4863b2aceab5c1191a0a : SUCCESS in 6m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0193030078e84dffa16e537d64d46bf8 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/67ceba5a98754604b7968b3788db52e1 : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23a0fd44ea484c6b9d3064b33f6f0ed7 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/32fe8ffca119407c83ce7533aab028a2 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1005344fb4c84850b10f28a5cdce5e87 : SUCCESS in 1h 55m 40s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17dde938a2064691867c64a70ff5ebd2 : FAILURE in 17m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/202ba513f7d7446fbe3500aa7caab43a : FAILURE in 1h 10m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09726ed8aaeb417c896ffc5ba8cc5552 : SUCCESS in 57m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9f73c3f372848febae2028dea544725 : FAILURE in 1h 59m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3fd405cf2d9463989c5fa1aed280d66 : SUCCESS in 1h 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5659366ed4d8370851fbcca5eebf158d1264a493","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-09 18:25:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"ac02715b369849499de7ee185fcc6f900f20dd02","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-09 21:00:43.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6bac4f0975535b5bf653f7c71a59c5148888487d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:02:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c9d09087ea44a255d925ef170ee00fffb27a382c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:05:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a21212889b0f6984893b9c51337a7511b4281984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:05:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14e929dd374641bf8130b2ea03a51aa2 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b27e1de373a4e5d85d3948d0719f3d4 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edefcf1e60aa4d42b86fe8dcfb867eec : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1c6293be24e465eb128239e84e25c99 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf9f532d1c5426e8af3292b881cff60 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a260234b517646f99c5a7e2d1a6529e1 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8bdbfa3af267496d9256edfc41670de5 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1082717a3a3451daf5648a2d378e6bf : SUCCESS in 46m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6dea4d9e343446a3a1d721f455968ff3 : SUCCESS in 1h 46m 18s","accounts_in_message":[],"_revision_number":6},{"id":"f89f8178567013a1f1208c2c4c1264035b2261f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:11:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb6a485a46948deac64cff8c2c02034 : SUCCESS in 2m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa27938ee9dd45e88a2e6941fa03f360 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cccd47e05abf98bfee2aa07a8528d93a327078c6","revisions":{"8d7714f8874028710896e1149a82b7ddff6e5b69":{"kind":"REWORK","_number":1,"created":"2019-11-08 19:52:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/693586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693586/1"}}},"commit":{"parents":[{"commit":"c9d11829e769fc4181cda09a8e5f7d3378a7675f","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9d11829e769fc4181cda09a8e5f7d3378a7675f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:52:15.000000000","tz":-480},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d7714f8874028710896e1149a82b7ddff6e5b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d7714f8874028710896e1149a82b7ddff6e5b69"}]},"branch":"refs/heads/master"},"83a4612c91e431ccf8d0e3055aed6663b2949801":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-08 19:55:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/693586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693586/2"}}},"commit":{"parents":[{"commit":"3a6764910b67a264fd328857955fc77f89427aaf","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a6764910b67a264fd328857955fc77f89427aaf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:55:07.000000000","tz":0},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83a4612c91e431ccf8d0e3055aed6663b2949801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83a4612c91e431ccf8d0e3055aed6663b2949801"}]},"branch":"refs/heads/master"},"d97638842798d4a7d564f4d61897112bacd6111a":{"kind":"REWORK","_number":3,"created":"2019-11-11 21:28:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/693586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693586/3"}}},"commit":{"parents":[{"commit":"e4c105feb3b2ae2f3375b564bda94d98f03856b5","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4c105feb3b2ae2f3375b564bda94d98f03856b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-11 21:27:51.000000000","tz":-480},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d97638842798d4a7d564f4d61897112bacd6111a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d97638842798d4a7d564f4d61897112bacd6111a"}]},"branch":"refs/heads/master"},"2aa2e670f81fe204618d5e3ead43cd37d96aa2d4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-04 01:47:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/693586/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693586/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693586/4"}}},"commit":{"parents":[{"commit":"c8fee592525459c4b8953b0e1853a935a9555375","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8fee592525459c4b8953b0e1853a935a9555375"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-04 01:47:06.000000000","tz":0},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa2e670f81fe204618d5e3ead43cd37d96aa2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa2e670f81fe204618d5e3ead43cd37d96aa2d4"}]},"branch":"refs/heads/master"},"0706492ae19204f38e5026095f765d4d9a718df1":{"kind":"REWORK","_number":5,"created":"2019-12-04 19:26:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/693586/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693586/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693586/5"}}},"commit":{"parents":[{"commit":"a28176d5935494991cd0273f59fc4e4231e95237","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28176d5935494991cd0273f59fc4e4231e95237"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-04 19:25:50.000000000","tz":-480},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0706492ae19204f38e5026095f765d4d9a718df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0706492ae19204f38e5026095f765d4d9a718df1"}]},"branch":"refs/heads/master"},"cccd47e05abf98bfee2aa07a8528d93a327078c6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-09 15:49:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/693586/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/693586/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/693586/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/693586/6"}}},"commit":{"parents":[{"commit":"7d23a711dde9d0427c6d04b45125489673c158ef","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d23a711dde9d0427c6d04b45125489673c158ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-09 15:49:03.000000000","tz":0},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cccd47e05abf98bfee2aa07a8528d93a327078c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cccd47e05abf98bfee2aa07a8528d93a327078c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
