)]}'
{"id":"openstack%2Foctavia~693591","triplet_id":"openstack%2Foctavia~master~I8e449b19af72d72f6effd52e0a2debb5754a19b3","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8e449b19af72d72f6effd52e0a2debb5754a19b3","subject":"Fix a potential race condition with certs-ramfs","status":"MERGED","created":"2019-11-08 21:42:54.000000000","updated":"2019-11-18 18:17:29.000000000","submitted":"2019-11-15 13:34:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693591-1573824875430-ded4c5e8","meta_rev_id":"30722379a04b458ce4a9a790d306c81da430e854","_number":693591,"virtual_id_number":693591,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"date":"2019-11-13 01:23:36.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 13:34:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-13 17:19:08.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-11-15 11:56:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-11-14 13:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2019-11-15 11:56:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-11-15 11:56:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 14:50:46.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-13 17:19:08.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-11-14 13:16:25.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-11-15 11:56:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-15 13:34:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8975679a7f4ddd86d08f5caf8ebfa049bc867796","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 21:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0979aec2c209202ce5cd84b6fa764e916b27461","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 21:52:12.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"1ca68ccb5b2a1ae669923ac7c1d46ec1460b4888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 00:18:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8d72f96acb34e818699100b15b385d3 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40284bd40c5f4f06a6f2091aa52e89f7 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3e7ea3f0ce4f8ebc2cc74c59861352 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cc0070627e84b65a599d7668c5f3cf7 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a265cb8f3964c4980c28147e9a529e8 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/513a09ce79944cd98a880560bca0e12e : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440665d459ba4bf69f25d97e08af239e : SUCCESS in 9m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1c11931a9c754ec49496d3dec6d959f1 : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3d35ad06180d4200a267f8d4c61947d8 : SUCCESS in 7m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1674b53cd24c4765aac28fa07d03a1e0 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fab7fc36badf4f8c849733665d57f3a9 : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a82a91bebce645d7bfece975ef650dcc : SUCCESS in 43m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4f34790cb104462d8f6b693083019b97 : SUCCESS in 47m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd7d4e807924456893d8502fe61388b9 : SUCCESS in 1h 36m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ef40b44442f1429c9c4bc64bc2a9190e : SUCCESS in 1h 48m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fb4ffb1644b245418e7f8edc76932838 : SUCCESS in 2h 12m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e669e671f37b4a31bc512841fb067a1c : FAILURE in 35m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8bcd260a6a2c4ffd906f4c61d175b1f9 : TIMED_OUT in 2h 33m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/146c6b0caec84b4993f6d4722074060f : SUCCESS in 1h 26m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/431d403b04014aafa0c6a3f59d62b0a5 : SUCCESS in 56m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3fece85f7fa4a85bdcecd1d10fe151a : SUCCESS in 1h 40m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73346c2832154ef3a8fb72062cf003cb : SUCCESS in 1h 04m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ce2541010d054bb795c1022f57add549 : SUCCESS in 1h 12m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/91c3d543ed14485289a3bdbb057961aa : FAILURE in 1h 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0c822f61d38a7f9e3d3ea3e7f1f9612401e2cd4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-10 14:36:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nA release note may be worth including.","accounts_in_message":[],"_revision_number":1},{"id":"56410a7d9557f59cc9e065bcb64e50e3a17f53c1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-11 14:50:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1514d249444476dc566b3832406391eed34b6640","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-11 21:45:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"702c7d1899513981fecb695cd1026db96104ab25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 23:57:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2556cdf845e468098e239f0f35e75e6 : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23ba72f1434a440d82555e055c31c862 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9a0978769d462384f28f898affeea8 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54bf79e13244452b98ee77eec320fe4a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac04312bcfe64d96b6f5438b84729d74 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64896a73099e445e88c0fa9b4e8335bd : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01755731f7764e16b5326935182325bc : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/749c2a542dfd431fb799ec1d24715881 : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f8003dcbef7f42938389f470fa87879f : SUCCESS in 4m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a79fb5665fc3423da6ea2d928b05a12a : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cf32324626c14ffe83f83de8931d4ae9 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2509c18bdea34d7ba62fce4c20706667 : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/05bd739c534d42babeae6d00a8b2502f : SUCCESS in 44m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/76c20e993ab24cf4aad8633186dae255 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2527e8092353499597007d400a998cc8 : SUCCESS in 1h 26m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/12cc1817397646d29fb09c573e318e83 : FAILURE in 1h 35m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/74fadd5dc89141689944100cc0491fa1 : SUCCESS in 1h 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/839bcb64f9484669b0b898468381b652 : FAILURE in 30m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a822f3ec54de4343a4a01751e47849a8 : SUCCESS in 1h 07m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c283679d1374eeb85ece80b605ceee4 : FAILURE in 1h 19m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9122222b6024c068b0f4024937f43e5 : SUCCESS in 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9e8d45c04364986be18069885b77dba : SUCCESS in 1h 10m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c5c65ada98b4130807d8722111ceaaf : SUCCESS in 56m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f7cfc99811b447c2934c466a984a67f7 : SUCCESS in 1h 06m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/20e0f708dc914a7184c4f7edb741cfd2 : SUCCESS in 1h 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2509f036b7bd4b2ea5730888e827b18f4dc464a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 18:39:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"456136186ab9773f51b48ae5ac12da433ec499c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 23:04:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c91c139cdc8744938342a2114b27ec82 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1cec304227e487cb6cf23712e8aacbc : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b52ce47d544790a57cc566e554a639 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9d1f0f938ad4b2691876902dade9098 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f03694b7a5543f2adfdd13e5166846e : POST_FAILURE in 30m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78192e87eee946cd939c3c70f944e848 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae73e610ace4a0586b955f254ddf8bb : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2125e8ad06a4e1a9dc1c1eda234a9eb : SUCCESS in 5m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5c0a24657da341a999613dc0e5a975ea : SUCCESS in 8m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3576dbd7a49e4d149413128963cfc03c : SUCCESS in 7m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fe1d724974fc4885b69bbfb50899b2bd : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27015dfc453f4a0e950cfa76ec98386a : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a97b5aba80864f84bbc6a45c1dafa90f : SUCCESS in 39m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d38b607eaf054c7d9a917b3ea2ad8713 : SUCCESS in 42m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de417c6e7a3f4152b7988823742c833c : SUCCESS in 1h 29m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/71eb68c21cbf4ee1885432ee88321cdf : SUCCESS in 1h 30m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e9acc746a2324e1081e7328240a9926e : POST_FAILURE in 2h 12m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa0371ff3fdf443483c4dd06c3997254 : FAILURE in 33m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/064fd252b36147ec80e4249087c1d7da : SUCCESS in 1h 45m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c99f30e4a9cc4b06b13e0cb5653e5676 : SUCCESS in 1h 15m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3835ddb65bd3495ebd537e9e3fa19b33 : SUCCESS in 47m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/afc6e0817e304907b0cbc1e0c247a9b6 : SUCCESS in 1h 45m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2738eaf214ea4d8ea323b9d093abbcab : SUCCESS in 1h 12m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c5eeb16425de45cd94982fd9e5f51a41 : SUCCESS in 1h 03m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/43e1514676154b6ea4d09038e1a73840 : SUCCESS in 1h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c941a7d0e6e571b421b9e189d6f4f07cd5443d7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-13 01:23:36.000000000","message":"Patch Set 2:\n\nrecheck\nPOST_FAILURE...","accounts_in_message":[],"_revision_number":2},{"id":"9dac6f4cf0dfb69aabcea49fe1fb4acfbb350098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 03:27:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad226ebd5d243d28b6a4453e30628b3 : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fbf26026ed946a083aa0b82d8b3ed93 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fadf79cf6b04107a426654852361443 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa32d38071204d8c83df6fb94741b8a3 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d84e72db22294f538628c0a1e5395e90 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7024020f3f3c435e8a940da8891042c0 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0e242fd88f4c44b868e89130d65855 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/432974c45a0747b08f5d1581df517b83 : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca66f5b08b794c47acbd21442e66c814 : SUCCESS in 4m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b7a0388766e048ab91083b5223e7e8ca : SUCCESS in 5m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5487342a7a044a5a815f409527f22558 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/324d4a00e06648a5ac71aeddd6717395 : SUCCESS in 5m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/acf978a323dc41a58cef05e1c7d79f32 : SUCCESS in 39m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2d2727c6aecc4aef918d83e44aa457b3 : SUCCESS in 43m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61517888a59349eaba7107f80e222bc7 : SUCCESS in 1h 19m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/de07d3abeb5b4f30b973f82dae5bc6a9 : SUCCESS in 1h 57m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/34bbc7ec43154d7da350b56294b7aa41 : SUCCESS in 1h 51m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c71cb336704420cb6afc01a1ddd5701 : FAILURE in 34m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce5570c3382f4af3be10dadd186b2ef6 : SUCCESS in 1h 13m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2bb61d158a5441508b12f3636a96fc0a : SUCCESS in 1h 23m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89df16dba6884ed48b5fcf33da467b1e : SUCCESS in 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dacdab9c8c1f4d359eb626e625b80d68 : SUCCESS in 1h 29m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c55672732164a88b8a1d4dbed363175 : SUCCESS in 59m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0a258669c5b44d8bbeb801182339cfaf : SUCCESS in 1h 04m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/499959762cbd4c0ab6bd21af0421a1f0 : FAILURE in 1h 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9640d4d98bd2b60ccc29dfb213d4153f14f654e","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-11-13 17:19:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3dbad7bf0f860bce0dc40f1b10b74195d263e495","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-14 13:16:25.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"88e3fee9e06d942a1b26da1180c542017561023d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-15 11:56:36.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a2bf6504219e2464dc0ef99acfb36b60e571c6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:56:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2511ff42150ba58793029c7d3f3ec034794a2a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 13:34:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51b02d1acc99466f949d6c6496ac9b70 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a1c870b1e74a4fa1ae28b13aa7f1ee : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19aa1e0ccd524cc49030418c008814dc : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f03e197dea93413b8072ab46a0c51df2 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f68e6b38469146cf81fa70ddbf792d81 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e9667591934c6bb836dbdce3047604 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19c43013aab3412fad15d1d9fcfc34ed : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/66f3d646202e4f9b873b6e0ed1428d6d : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b26db1d43d245e58e6957b50aae9e23 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/084d03d7f1dd4d7a8bd5b61dffcf3099 : SUCCESS in 49m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a11d9b428327483dbf1a0c90729d7b93 : SUCCESS in 53m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b8f574d535a43bc9a82152d12a48771 : SUCCESS in 1h 30m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f2721258a7be48098e6ed7536a4e37c7 : SUCCESS in 1h 33m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de3e05231fac42b19fd5e482386e40cd : SUCCESS in 1h 32m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d944f42023d810a93993221f0977b02198705078","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 13:34:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"65f93a21762cba3323cf55127ac33fbfc0bb96b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 13:36:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9f5a58ba4b4b969920d230d8f9a951 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cce169ec53947479df293504131d374 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2},{"id":"30722379a04b458ce4a9a790d306c81da430e854","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:29.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 7c1ab033326559f27a489bade083a14164908e2e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1900ee71a51c25d4c0fa1ee79447219cef05b5ef","revisions":{"75f9a7777980da514eef8f71d08e20120f3971d9":{"kind":"REWORK","_number":1,"created":"2019-11-08 21:42:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/693591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/693591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/693591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/693591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/693591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/693591/1"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:42:53.000000000","tz":-480},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75f9a7777980da514eef8f71d08e20120f3971d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75f9a7777980da514eef8f71d08e20120f3971d9"}]},"branch":"refs/heads/master"},"1900ee71a51c25d4c0fa1ee79447219cef05b5ef":{"kind":"REWORK","_number":2,"created":"2019-11-11 21:45:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/693591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/693591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/693591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/693591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/693591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/693591/2"}}},"commit":{"parents":[{"commit":"0682fb977a76d356ede20f516c849298d34eb7fc","subject":"ipvsadm \u0027--exact\u0027 arg to ensure outputs are ints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0682fb977a76d356ede20f516c849298d34eb7fc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-11 21:45:35.000000000","tz":-480},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1900ee71a51c25d4c0fa1ee79447219cef05b5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1900ee71a51c25d4c0fa1ee79447219cef05b5ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
